)]}'
{"id":"openstack%2Frequirements~967490","triplet_id":"openstack%2Frequirements~master~I674f38a33b9e3e177c192a5677d0dd2c30bbb62c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I674f38a33b9e3e177c192a5677d0dd2c30bbb62c","subject":"update constraint for python-ironicclient to new release 5.14.0","status":"MERGED","created":"2025-11-18 09:45:26.000000000","updated":"2025-11-18 22:00:11.000000000","submitted":"2025-11-18 21:58:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967490-new-release","meta_rev_id":"fb596895ebab6cac1f2f07b2a32331399c14479e","_number":967490,"virtual_id_number":967490,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 21:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:05:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 10:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:47:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:05:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d777a72ff6126b640f8d1d469f60521c9a6fc4ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-18 09:45:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c61fdb2fcfdd3c310a0667be69bb377dc6f340b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 10:45:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baece77de14e41978916b07de16a9a56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e1c10cf1b14f5caad8f8de212992bf : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3562b9cadb9e4dc38074be23b3f357ff : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8354f8c9f3a64c65bafb30d42383f7a8 : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83b28bff1d924f1b96b3f726de7bdffe : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbf8f6c3cc6b4bc09f516aa5f1088e40 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6242ec590ee4799a6142f01b11c1cfc : SUCCESS in 3m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a617e99d48f1496789acccdd2bd0321d : SUCCESS in 5m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/445258f34f8643a28ed494464624ba7a : SUCCESS in 6m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/06382365964c43b1ad85dec584581707 : SUCCESS in 12m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/601e5234c1c4402f99d54a12718d83f1 : SUCCESS in 10m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2bc1c9d77dbd4cc2972706dbfca7eda0 : SUCCESS in 3m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f44d508b45be407ebe9ee9ecdf8d86f6 : SUCCESS in 6m 43s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dfc8b397f1574b0dbec80fb5c8b2ef5b : SUCCESS in 3m 17s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/552ddfa610334a12a97c3d1ef82669cc : SUCCESS in 12m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1e4477eebbdd4901bdc71f99762bd313 : SUCCESS in 3m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f303076e8584420ca00e22acc4e7bfba : SUCCESS in 6m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/495c1c38be014373a46389a69004abc2 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9930fada28c445eebf5c6a352d4271f0 : SUCCESS in 17m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d65af06a8f64f8c9021f5fa15aa62c2 : SUCCESS in 9m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b5ea37d173f14bef910d19849e0a493d : SUCCESS in 8m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ebc4412b990941dab64bfb776625a1f2 : SUCCESS in 12m 41s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/357389dc2f3d439f927b3276abec97e1 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/29a5e5097f1444eeb75ff26226fc8dd9 : SUCCESS in 9m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9bfa97ca1860426dbbf097a576bb1f8c : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8c52456dff1542678b908ceea3a08d1d : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/66587a89a8874b8fa3f529f857f2df05 : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c18d327329ca421b8fdb61598cd83551 : SUCCESS in 23m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cffd7bf93d6448b791634e3e63ce42a6 : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/81728603209447e0a42a9d5a64901c8f : SUCCESS in 14m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9187b9215bb246f48fc1ef583d6a6626 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9b0beba8d23447a9e0bc3ee56fb0865 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fa437683d344060b3038bfb937f7552 : SUCCESS in 26m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/89b1f75d783e4f4c8a4697f0b1c30d63 : SUCCESS in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4578c093e5574c11b6b77933f3200b8b : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13f02ef896764a2d9fb0444d688afe21 : SUCCESS in 6m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f3396a75061e49ab85e88cb54203a931 : SUCCESS in 6m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9c6e2044ca9942629749af38bce1ef83 : SUCCESS in 5m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51be704bf51d440f8dfe72ed806c03d5 : SUCCESS in 32m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dde16a1c14044502a30fac85e4dbf618 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e965c690bbde871a8a2c407452f20c5efc833c1e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:47:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6d5ea58e1c89452f5510cffbb420e2b3f9447f6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:05:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0dd2259798889a237fff46c845f8b91d81350a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:06:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adbc4045e90003154d8ccf8b331a7b12003a70ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7acc647ad3d84fd3a8672f553643ac0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0d0978992248308fffc3b808c8f88c : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb1870245a24565a75d79db6d99f141 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cec1e1711d3143b2ad88907465a1b67c : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5efb56a1eb974beebb5bd954ec85bcf1 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b365286646d4eb193b37d8f3be9efc0 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1444af85e344e7286f814decdae1b0f : SUCCESS in 2m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/38a218e970c243c49cc22e88ed9839ef : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7961e329189043e89b5fada3d8ac34ef : SUCCESS in 9m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a85266a46f8240a7998e9d464cc06da2 : SUCCESS in 6m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0645e7719ce04b12878ddf04c1aed58f : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1901c5c8764345a282e9483d5f6bbc05 : SUCCESS in 4m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fb5fb42fd4e547288523ae5186b15359 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c3e34737583640ea9b897b0fb4c6f2aa : SUCCESS in 4m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d0feb98a377241209962f8fce75342ad : SUCCESS in 21m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dcb366a959a483f9ec25023c3a87f92 : SUCCESS in 8m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fb3dc369f60b4306b1c9a4249bbb2830 : SUCCESS in 10m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a01f59f5f70a4225ad5cb17fe7dff109 : SUCCESS in 14m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/95c5442ffc5146aab1a909945e6e28dd : SUCCESS in 3m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/799346265cb04dcdb63dd8865b083418 : SUCCESS in 6m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a6ea4dcab56a4395984a74eec44ce92f : SUCCESS in 29m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/919c3adb844b491b9a1461a3c3d1cfe6 : SUCCESS in 5m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/25cdb4270797415ea5ecb27d18e057c1 : SUCCESS in 11m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c24731c9eacc4e039e3d5ecac0f95eea : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29f1926e677b4aa88a27007838a67041 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/405c5bf2a6b94f70a6e4c105bb606fd7 : SUCCESS in 27m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/248ee522d0074a6384da0bf6556970a2 : SUCCESS in 7m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/edc5f4f49ef14c88a348f74f4b6a1aa3 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0572aa90d5bb4041b6b940bb1be08cdd : SUCCESS in 4m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3c19bd0d488f48a0becb148937623f4a : SUCCESS in 21m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b4ef0f1f7ffb4d45a6f6b638a112fae2 : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fac86c1ac0cf4fc9b6e15503177bb106 : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e78bd010ddc4b0bba0cd04eb58b3f60 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c1a2be683a58f549da277c08c31cf4f89e93ac46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb596895ebab6cac1f2f07b2a32331399c14479e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:00:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2937443e20e4df7b605308f65b6bc46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed6b27a7dd54b93bb695651a1e60841 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e90b06a6c1d026217955e3b1c37f6147e747881a","revisions":{"e90b06a6c1d026217955e3b1c37f6147e747881a":{"kind":"REWORK","_number":1,"created":"2025-11-18 09:45:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/967490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/967490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/967490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/967490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/967490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/967490/1"}}},"commit":{"parents":[{"commit":"8b17b6b0f89e941a1b065aa608f44491bd3b832c","subject":"update constraint for os-ken to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b17b6b0f89e941a1b065aa608f44491bd3b832c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:35:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:35:24.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.14.0","message":"update constraint for python-ironicclient to new release 5.14.0\n\nmeta: version: 5.14.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ic7f2088392e0d11d2b0eb1c5f2d767d987e58610\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I674f38a33b9e3e177c192a5677d0dd2c30bbb62c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e90b06a6c1d026217955e3b1c37f6147e747881a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e90b06a6c1d026217955e3b1c37f6147e747881a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
