)]}'
{"id":"openstack%2Frequirements~967500","triplet_id":"openstack%2Frequirements~master~I28b45a3b4f092d344a65eb87c3146afdc16373e5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I28b45a3b4f092d344a65eb87c3146afdc16373e5","subject":"update constraint for oslo.metrics to new release 0.14.0","status":"MERGED","created":"2025-11-18 10:01:08.000000000","updated":"2025-11-18 21:59:24.000000000","submitted":"2025-11-18 21:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967500-new-release","meta_rev_id":"8a4d532a90ef7e3484003da57a3e9ae15dad5f4c","_number":967500,"virtual_id_number":967500,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 21:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 11:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:46:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:04:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"6f2c3b36a4b47efe376facdd44e3d2698e89ace1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-18 10:01:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91a6e56a94c5b7d981a8c1122c94f7f60199e717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:42:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6bdc26a3ee4a1a9fee0e111936a99b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d14566410d428ba0191cf4e1917975 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6bd14d9b764005a211d35df7c98f49 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56855bf5970b404c977de1d4375f1c21 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df2a0df78e5a4b168a370308b8326a64 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83c14cebb08e4b849f5f63ba95e160fa : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7ac689466144c579fc5c476a1651484 : SUCCESS in 3m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/902169b70cd04fc7be23716341b29168 : SUCCESS in 4m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c2f898da05b64c3c8b94c2e99774ab7a : SUCCESS in 9m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/121dc7de7508491eb68631c7a8c4db47 : SUCCESS in 3m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3f990c58a186486a998211b626f61546 : SUCCESS in 15m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/39f3913835584e53b218d20a1d1f07ee : SUCCESS in 5m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ee8144d7916349d69e5431f08205babe : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fe88067e8dc14b4c916502d94fe79718 : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f4f2c2fe91624d0099e2b88aeaced6d4 : SUCCESS in 8m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4a38a20cb76b4264a7e6493d89396bb3 : SUCCESS in 4m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/045fc5ad0e554c7bb6f59555ce8730a8 : SUCCESS in 3m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b69d66cb8fdb44ba9307e3477625973b : SUCCESS in 4m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/888ec8b338ea4a59bf7ac5e2f35197cc : SUCCESS in 17m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e3780bbf7b748ef83c59877503cd98f : SUCCESS in 9m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/80f27c3aec494429b92ec3452558f173 : SUCCESS in 8m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f8648c2c5af244e0b4417d3286497933 : SUCCESS in 8m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/143a3b5376bb4e749436a8e5421b909b : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5e5b684b6f8840a2b0f0433cbb01bfa7 : SUCCESS in 11m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fba8c203b2174e939342cded33bf7492 : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f5bdd7f08d6449beb345f4c05581d453 : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fe4cf7d25a104ac3b22ba29428ac7725 : SUCCESS in 10m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a4ebc05d948f4b9fa33beb56e208132d : SUCCESS in 26m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d510ba777dd34479bc639af2a838a21e : SUCCESS in 6m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d32c1211c8aa4c04b4392a039127b198 : SUCCESS in 9m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d55ef4e146724bd8909b4efb0dd00f4e : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4a5caa13a63548dbb08c165328b19df5 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45d0d3f0a3624c198df8c0315064fc16 : SUCCESS in 27m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8bec1673116249a08a0e3f9dc5a3de2a : SUCCESS in 8m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c7b10a84d5164a2eaf60bed2e2b4a089 : SUCCESS in 3m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/538408576f0247a8b6091fe2cca4852a : SUCCESS in 7m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b961e49db5fd4a5f83d582f2ee665a68 : SUCCESS in 13m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/381d181067e24498bcba7bc18613f0fe : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28b2842652854732bc6eb33c1f383d8c : SUCCESS in 1h 04m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9e76511c5fc4c5c947bcaf6df04f193 : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0afa3baa10268cf13e631cd9ffb2bc769facac6c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:46:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ba089402bbafdcc0efd3ae27329569b521e4bf3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:04:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bafae557f7acc6ee3dcb765821627cc79217610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:04:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2205fc689b3dfe098565a91a857800657e6535e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161df13034db45aea63bd42c8ce319e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1260f142592d449f8d70c87f9ac286db : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb6a09d63ba4f3690c2f646033d145d : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00a0efc6330f4d73848d9d8e51286aac : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53298b9c1ce04f7d8faa8bf3f72f94e8 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8d87156fe6f45949844de857acbb88e : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/724424da0bee4cf5b808ddffe5cfd05c : SUCCESS in 2m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5147fd1c66504e4b9f88da8733257132 : SUCCESS in 5m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45890fa78aca4de290d4fe6cb366a660 : SUCCESS in 7m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c5eb22f7bc4b48028b7f2cca26ab94c9 : SUCCESS in 7m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d269ab8ffa57448288acd6af6f2b421d : SUCCESS in 11m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1770e9f171934b20b7375cfe4cce27c9 : SUCCESS in 3m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5100766f2c7e415fa7ca44d6e9f45841 : SUCCESS in 6m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bc8dbcdad09948b9874919df6aebc6e0 : SUCCESS in 8m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dff2c2f046c74f779877d7e5582f9ebb : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e8844ef439f42109a230efdb2f7f954 : SUCCESS in 7m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/76d65a3cb4cb4f8c9befa2d2ee05a6f9 : SUCCESS in 6m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3d2deba561444e2ab53dfaa255ce91fd : SUCCESS in 12m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a585bcf99ef84fc692113932c2cd644e : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/62bec986cb1b4c9688a3868485dcbba0 : SUCCESS in 10m 38s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/77430c7e0e774e7dafa72acb80fec691 : SUCCESS in 26m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3cf2302dcf1542b9b9c1e96a27c66fe3 : SUCCESS in 8m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b3c37354a28b4234a48cfee36ae69c69 : SUCCESS in 13m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7d3079b63c424554b6601381be694364 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6876e00f7fe2413780f05bccb68676a3 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1dcbfb7e17504c61b74ef2022e4a19a4 : SUCCESS in 26m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8a99fd64112a4a699ee9636a79347228 : SUCCESS in 7m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/835d6f70187e431eabf8a9396752aa07 : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4aedaa62f27542fbb7043cf128ae8a2a : SUCCESS in 4m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ce02ae07cf8a44e4860d0deb78524ee4 : SUCCESS in 19m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/be6fd93e69d04e07bfd3c63495604426 : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b321154ca34848b4965c08054f9d6d3e : SUCCESS in 1h 12m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9141ba9ce028464e9cf50ba4a82d534c : SUCCESS in 1h 49m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cb65569178176af5a1d20ae6d92b157802b9b8e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a4d532a90ef7e3484003da57a3e9ae15dad5f4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:59:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23bc1fbc47fb4b189d2b7520bf6f49eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd4f0a978194326acc07a40a7f2ffc7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6175fb3b6574a6e14dd81dcb9378869dc2984d32","revisions":{"6175fb3b6574a6e14dd81dcb9378869dc2984d32":{"kind":"REWORK","_number":1,"created":"2025-11-18 10:01:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/967500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/967500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/967500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/967500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/967500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/967500/1"}}},"commit":{"parents":[{"commit":"8b17b6b0f89e941a1b065aa608f44491bd3b832c","subject":"update constraint for os-ken to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b17b6b0f89e941a1b065aa608f44491bd3b832c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:51:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:51:06.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.14.0","message":"update constraint for oslo.metrics to new release 0.14.0\n\nmeta: version: 0.14.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5a9e7e5744f4964c73fd0291251ff205bd7d1245\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I28b45a3b4f092d344a65eb87c3146afdc16373e5\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6175fb3b6574a6e14dd81dcb9378869dc2984d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6175fb3b6574a6e14dd81dcb9378869dc2984d32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
