)]}'
{"id":"openstack%2Frequirements~967739","triplet_id":"openstack%2Frequirements~master~I6e8d31af54e0b703a572aea687f09b0b6a95eca1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6e8d31af54e0b703a572aea687f09b0b6a95eca1","subject":"update constraint for oslo.middleware to new release 7.0.0","status":"MERGED","created":"2025-11-19 19:36:24.000000000","updated":"2025-11-20 04:32:20.000000000","submitted":"2025-11-20 04:32:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967739-new-release","meta_rev_id":"2676f766f65e3dac86a4d28d5870b9d4476e5526","_number":967739,"virtual_id_number":967739,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-20 04:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2025-11-20 02:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2025-11-20 02:27:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 21:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-20 02:27:49.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"fef014484f85bb6b2cd7110874f198df86c977eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-19 19:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"506a1b8126a8e6dacbc93f88c606772e92637b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 21:15:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/816cdb3170224fcc934d39b72d972090\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb2622ffccb48a38c61505d402d152d : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b198857b1254ba29592ed8463a8a684 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/289eb7b5095746b1b2cdd2a1d3eef2b8 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/889fbd2a48d245cb83f84398c7c63cdf : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/930c1685577049cfa546e6e46cee2f5a : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc30bfd1d250475bb5a0c6dcd9cc0db9 : SUCCESS in 2m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2514d919f6db480e885218e27ac6e50c : SUCCESS in 5m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a28a4d9a8c804eaf894d7070c27e7f23 : SUCCESS in 6m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b29b0743fa524817a220d5457b6fd1df : SUCCESS in 9m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1087f6d414014ab1b24476845cf7904d : SUCCESS in 10m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/591ee8ddd9d04df5ac527b1d93312e68 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4d68ce1f71554a5f9ddb180a67135ab4 : SUCCESS in 5m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/50fba573d0c341ee9e83ed8c265b3187 : SUCCESS in 4m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e1bf0bdbc50d450a9cc9f82ef3e43aef : SUCCESS in 10m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/29bb8a652130499a99f8b54e7790a959 : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d0d83a70883f40b4ba2a1b7b0b71d7ae : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c48a27421c0340ec8a492e14d20af0b8 : SUCCESS in 7m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/447f5e4d16f74acaa071557a400c6339 : SUCCESS in 21m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4035d3f751c543fcaf09086d007400bb : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/01d1323d8c09492a8ec1cf2d3ae09c92 : SUCCESS in 8m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4bb31e06b2b449e18b1b4325dd14bf5d : SUCCESS in 11m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7cc67219f5c04955907d42233ff6dcd9 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/54f19ece86464872b9def75bcb0455fe : SUCCESS in 10m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/71ac5a5829e548afb0f9d36ca04c04a9 : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b9cd64c564a240649e868ab1b129d365 : SUCCESS in 4m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9570faa59673477885e0e799c93dea96 : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ad2250a1ced0445d966eeb02a4b13432 : SUCCESS in 23m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d69bd5f66a04f68841033fbfc91c4db : SUCCESS in 9m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/38f8f44f40c64c58bed656ac091e6f67 : SUCCESS in 13m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bbbcfce4dc0040c189d1ba4e07d11609 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7710ccd175c84fe5ab5b540dd83db134 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27341679b67b4c3d8c164ef2eae20ec2 : SUCCESS in 26m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3244eff84a2647da85ee7365f1d8a4ca : SUCCESS in 6m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6a3ebfc21bfb437ca24a3070b701e162 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8dd74000a33347fab0c3ee2ed5fbdbe8 : SUCCESS in 5m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/684b8341f8074c4e8e750e756230f64c : SUCCESS in 19m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5c071281024a4f2fba1db2b528c390e2 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccf71ae37c3a4d23907cbc5c22c0cb00 : SUCCESS in 50m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea1fd58e9eca413781e05eaa93dadbf2 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b1edfeecde20a377f5cc46538790abd0427ec368","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:27:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14fdd77f5b817069b31ff72cb2c94e4df507c5f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 02:28:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"287a371fa2f08bc9017972d3e9818fc6cb4bfde8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 04:32:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264228cd2df042db812be36e092b872d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/941b78ccf26c4c03afd0f20e207d34b1 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9146a755b3944df9b812386404d051a : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/391676d6a7b84011af5754643fc35008 : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68c43d4a2d3242618825b9aef399b76d : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd10f6ecb4fb4e0b965ba63f6148afd6 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7d736b95fd1461ca5dc4445444bd14b : SUCCESS in 4m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1b985b82699c43788322841de8ab1e54 : SUCCESS in 5m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f5b714b0e43a4675b362e79aedb430c2 : SUCCESS in 6m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7eb11dbd7e342888bae38886633088e : SUCCESS in 8m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6b9874a770b5471091422ecc02f9c695 : SUCCESS in 12m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8869a9ac87ba48e38a0ab62437d7d1f8 : SUCCESS in 3m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4590c8cf73874172b7ec6045e3aadbb4 : SUCCESS in 6m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a8a7cef4206249fb950a501a2e1c51f5 : SUCCESS in 7m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f0cb48e1629b40f8b6014cb1a175c957 : SUCCESS in 17m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5454d81019af438985346cdcccd966c1 : SUCCESS in 8m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db6a61ad32cb4622893094968eec6a61 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/141aa946d2f941ae913aa0dcacdff6e2 : SUCCESS in 12m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/39f3c96784f042d9841274011da57d8e : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f686878fb35241e3a9adbb63451b1ced : SUCCESS in 7m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0a9690b6f4994696be81137cc4c3c5e8 : SUCCESS in 19m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e14aa76f19b460d9d3eec8643372e0a : SUCCESS in 8m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ce610fac161d452d88c3f3e54c449ee1 : SUCCESS in 10m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c1c918f6a0c245c1a524c0822fa95b07 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c87d7840dbc64949a536bcbf49fcbd45 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33eb3f9ef53f4a0e85f9bcf56a1aacc6 : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b15f9753fdcd434b9e1302510fdf361c : SUCCESS in 7m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8470d4c2c8ce4939b00d2dcd0e3f1439 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e32456562f2a4ef1963814c4685fc08d : SUCCESS in 6m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/313e9b0ddba0415ead58e5829718dc8e : SUCCESS in 22m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a234bd1949e34259b4dccd21301502e1 : SUCCESS in 6m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1d4da10311142f1bb2d322c6e50576b : SUCCESS in 38m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74a8be86c4a84f36aebf571f211862ab : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2676f766f65e3dac86a4d28d5870b9d4476e5526","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 04:32:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eecaf4f374dd564806510c3dba3c20a663c3a1a7","revisions":{"eecaf4f374dd564806510c3dba3c20a663c3a1a7":{"kind":"REWORK","_number":1,"created":"2025-11-19 19:36:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/967739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/967739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/967739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/967739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/967739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/967739/1"}}},"commit":{"parents":[{"commit":"fef0a61dfaed89d90a66f98781944e5a3fb196e3","subject":"Merge \"update constraint for oslo.rootwrap to new release 7.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fef0a61dfaed89d90a66f98781944e5a3fb196e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-19 19:26:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-19 19:26:23.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 7.0.0","message":"update constraint for oslo.middleware to new release 7.0.0\n\nmeta: version: 7.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ifbad95c6945ba51c4bf0809922ff9f6fea6d2af1\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I6e8d31af54e0b703a572aea687f09b0b6a95eca1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eecaf4f374dd564806510c3dba3c20a663c3a1a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eecaf4f374dd564806510c3dba3c20a663c3a1a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
