)]}'
{"id":"openstack%2Frequirements~967862","triplet_id":"openstack%2Frequirements~master~I31699f0b8a20c6e1b55de2d57e8392196e000da5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31699f0b8a20c6e1b55de2d57e8392196e000da5","subject":"update constraint for oslo.context to new release 6.2.0","status":"MERGED","created":"2025-11-20 15:33:39.000000000","updated":"2025-11-22 10:38:12.000000000","submitted":"2025-11-22 10:38:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967862-new-release","meta_rev_id":"da916f0f2519ee7e93ed020b2ba03c67077329ea","_number":967862,"virtual_id_number":967862,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-22 10:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-11-22 08:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-11-22 08:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-20 17:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-22 08:40:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1ac72120e869b8178589c6bd9053ab4d5fc09d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-20 15:33:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea58e8fb27fc4b1bbc82cf1f7cf53b06124f34d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 17:28:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3a513696ed460493e802093ef2ec34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f870de93aa764bddbbf98759eb138218 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a504f5a3614c9ca48accbd97bf6586 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e79247ae82244528eef63177fa34d84 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/833aa9fb282c423ca50374d1bcfdfd81 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/905f471715854d7497b0dd0d8c58724b : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cd93ffb6a574d79a9ef6457cef9e6c6 : SUCCESS in 2m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/41eea77ea37f43e19021ece6e77c73c6 : SUCCESS in 5m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cc09dccfe5c44cb3b64acc5cd1fb939f : SUCCESS in 4m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb14ec9d8775477ba96fe6e8a0fe6153 : SUCCESS in 11m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/55c34195a0704ef3bdca05c66fd2b178 : SUCCESS in 10m 20s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b32e7d40abde4ffdb94612e97ff8a908 : SUCCESS in 5m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b5ebc1a573f64d6896c11c3fbb7f14be : SUCCESS in 5m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/45da64568e30493b9548e8def002041e : SUCCESS in 2m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/21ecb90d4b3b4db59c7dece2390141c9 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9a36e226d4ed4a469246f8996cc5e355 : SUCCESS in 4m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c099fe7e6cea4f06b6844c1162032c1d : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0cb253d19f62456a8c20df6a578ca891 : SUCCESS in 8m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ec94736e6ef14472a39b378f609021e3 : SUCCESS in 18m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76b99aad5fe34ac7947ffe34188c37d3 : SUCCESS in 7m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/037eb0fbbacc4e5db602fabd056b3f54 : SUCCESS in 6m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c8a5f3aff11d41098a10df89239d0d85 : SUCCESS in 12m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/073bc7be7e3040bda9c2d41a9387665b : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5db6b75827c145f9af387b3783f3b3bf : SUCCESS in 11m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/94804af3ea5d481f9e3c43f3399910d9 : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ef2c9e2844c40ddb0b6ce2918191b6c : SUCCESS in 5m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0cb6b9fd3c1145e2b658fd3d13f06346 : SUCCESS in 11m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8ababc7f374f4c03b86f98c080ec41f3 : SUCCESS in 29m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35cdbe42e3c946b9906339bca9ea7464 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8d7fd09430a245bd88b365d141d38b40 : SUCCESS in 16m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d4641df5e89243fa9b433efabb57d7bb : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc0bf02ab9084b32a92f472acaf6a7f8 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb1f337867e74758b8dcc36b6db20140 : SUCCESS in 27m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/172edb5ef4b14766b1338f3de9008b88 : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fcfe05a2bb954da89ce4a9a76e8d83a6 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/635dc74234634d1b812d877243fd5b57 : SUCCESS in 6m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bd3e79f2764b4274bb6f5cf86e49e7df : SUCCESS in 17m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2277123d8c2148d99763020ef4e9df8a : SUCCESS in 6m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47aaf4a8dbad4101bd8752c8c19e75e5 : SUCCESS in 42m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0161513c5cff490d8fc14735b72f2b24 : SUCCESS in 1h 48m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f7fd59a74447aa9c3b3f036c279b770f51eb1f94","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-22 08:40:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cd64609af5b5d00ea962c276e5a9d60ed9b2697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 08:41:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2683d45c5847ce58ba4cbd8bff1783f717d6cdd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:38:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d1415678a8349c8b4e98ae9f53c8d41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101255d100874284ad1cb2f43031b448 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3bc2764d9a4d5fb820ec3f3c188e73 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c9b85d25c74603997e9e38e87af0b9 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9aa111c2ecf4447686989699c262f864 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb37a7cd23ac43e3ab9468276440c401 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a499030d10d4cc6b66de15811488aa2 : SUCCESS in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f3ae5f019d44de2926e7aa2e11fcd9b : SUCCESS in 8m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f029feca3bfa4762848d7f8d70be6717 : SUCCESS in 6m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0b0cd51f89364ee193c0c17d43cfe412 : SUCCESS in 5m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/07990a10219e4e3cbb407865e0c8d204 : SUCCESS in 11m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/99b51c47231641eb887e39036b1d7daf : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/76a07742f2f64661a946da53328ff182 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fece76afd5d045b98c9d1b6a45d02cf0 : SUCCESS in 8m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/af0d1ad69d3d4d54824c2acfa8151983 : SUCCESS in 18m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3a17c919e4b412186cd17d5d91746da : SUCCESS in 8m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/865a1fb271bf4d64a8dda752007d83a5 : SUCCESS in 10m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/713b0618ca83417f9088530cc53c0555 : SUCCESS in 7m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a3ec5fbd6023408ab983acd12f5e6ad8 : SUCCESS in 3m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/63e7a10e3bb642d79e52afc64a3058b3 : SUCCESS in 10m 32s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/43aca8322b3f41fa92c2a5a27babc4fd : SUCCESS in 26m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/faeb1ed9316d46e893bc961dc7ac2504 : SUCCESS in 4m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/01d6a9a857984a7d9a02798884735c5d : SUCCESS in 11m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9bdde75e087d4b92a7876b975e3bd455 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6eec8dfb01cf42c38af13b2e8600d86f : SUCCESS in 2m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c655dfa465ea45d3999383f76cb261bb : SUCCESS in 27m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/42effbfbed2b413ab8c0ea5f42322655 : SUCCESS in 8m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/082d6fa8d1004c9084b422226d3f46c0 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8973ab2c14b64233b00dbb80477db8ce : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5290097ca4504610bcc07d1de370373f : SUCCESS in 16m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fd2cea4ce1834a01918fd8ed2e8539b8 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8262df58aa74b78966101c4598c3e32 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc983ade57824c71a2759378f4264b77 : SUCCESS in 1h 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"da916f0f2519ee7e93ed020b2ba03c67077329ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:38:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"107e0a76021aa3f7be130a4298fc32427975954a","revisions":{"107e0a76021aa3f7be130a4298fc32427975954a":{"kind":"REWORK","_number":1,"created":"2025-11-20 15:33:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/967862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/967862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/967862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/967862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/967862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/967862/1"}}},"commit":{"parents":[{"commit":"6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9","subject":"Merge \"update constraint for python-zaqarclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-20 15:23:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-20 15:23:36.000000000","tz":0},"subject":"update constraint for oslo.context to new release 6.2.0","message":"update constraint for oslo.context to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I3115fa6436cd47057d891833df4804f1ae29a0a8\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I31699f0b8a20c6e1b55de2d57e8392196e000da5\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/107e0a76021aa3f7be130a4298fc32427975954a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/107e0a76021aa3f7be130a4298fc32427975954a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
