)]}'
{"id":"openstack%2Frequirements~968022","triplet_id":"openstack%2Frequirements~master~I10b662a41cd37552cf3cef350ec3ea4a97900472","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I10b662a41cd37552cf3cef350ec3ea4a97900472","subject":"update constraint for os-service-types to new release 1.8.2","status":"MERGED","created":"2025-11-21 14:16:05.000000000","updated":"2025-11-22 10:41:19.000000000","submitted":"2025-11-22 10:40:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968022-new-release","meta_rev_id":"a779687da118f22c15aa144bf09048fcd8e91d54","_number":968022,"virtual_id_number":968022,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-22 10:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-11-22 08:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-11-22 08:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-21 16:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-22 08:41:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"df4788c585890fe9371801404474f6b880de5d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-21 14:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa07d4dabfccd7e5c90055ebe69d4a57a47c9044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 16:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/207e6e9d5c694a0f889bb12c840babb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1706a1af11854e35bd178bb427eff099 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c5640c4ad8468799a26ed33e6df34b : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac8a0ad6e09443418d265235f8415ca7 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab47029f07df4527a5de16226ab5beb0 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da8298d6131a4f4581f16675393c6d2a : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/783cbddc17ab45f3b6ec61ee88e26844 : SUCCESS in 4m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/db8e48bdccd2431ab411ece1be2fd5a7 : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f7b0bffb5b5946678d889b9ee025c28b : SUCCESS in 3m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/40e959f07a8848e2899d4de8da5f37c4 : SUCCESS in 5m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1a556c6aa4c7423fad1f8be3332d9098 : SUCCESS in 5m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0ee010f8513d4969971d5120f0956779 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fa1030e975d74be886b5aab3b4b7329e : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e89a65ee8b314fafaf149e00f3e09f2e : SUCCESS in 4m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/55dce0db6b114d179251cf4980b06264 : SUCCESS in 11m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/104e49f3f797457199d453cd4a3e1a14 : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/125e386e0f134b8198811b5fcf7328ba : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f173e34f0e2e4666bc1bac35fc16dfad : SUCCESS in 8m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b943a3611f3e463bb726823a3bb9f56d : SUCCESS in 17m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b7cc5a3e9bb4350b7796ff955305dad : SUCCESS in 6m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9763d6bc559e457b9e520654fd5b0164 : SUCCESS in 9m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/98f12945fe4b4535a07926b6a461458d : SUCCESS in 7m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e3fd1096269642bba1a5d231133d80ac : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/08272de31af443d58b602ab3975a70ec : SUCCESS in 6m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3461a1c25db14383b44629af7e94ed48 : SUCCESS in 4m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/22ab170660e84f858e8968c1e2601ecb : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/10088cc6470a41edbfd1c75e796b7545 : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/04c12bb1b18d45ee8d11f59887c0c6b8 : SUCCESS in 23m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01786db15f774e40a4bcba266e193f6a : SUCCESS in 7m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c1eb553ff27148058f1ac75063cd7c8e : SUCCESS in 14m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5a7ce7bc10ce433392739b6769291552 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c670669787134165aefd82bacb18d91f : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16d8ab4efefe456683bba74878bf7495 : SUCCESS in 29m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/70dae51d26004354811959c0e78bdbb6 : SUCCESS in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7433242937ad4375b302c4b24d40b762 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6876d0a89cbd459ab622cfef78dfcac3 : SUCCESS in 7m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/adef591933db4985934bcdba768d2d1d : SUCCESS in 18m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0fdfdb7210d04e0c9e50b9fbcded1ce1 : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/128e2120691c49b2b4062de63943b43d : SUCCESS in 1h 10m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17245149737146a3b81b3c1231a1ee70 : SUCCESS in 1h 52m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a6633dda08f36d763c7171c5aaa33c353caab77c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-22 08:41:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4c3f8d5e970209f16dee6025513b32c7a438c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 08:41:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86bce0246f6ff6e421384a7398ef0f542ffa5cd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:40:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b47ab66a0f94d59b2b4c7fc4d1bc7eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327cba76c36a47d7a1204b3d0ebc4dcf : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24628f024f144e529ece0a02ed08b0a7 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08aa506bbb264d26b7ce9c4f2fa3b098 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8350770f7c604ca3a61e1ba5714304d0 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9eb6ddfeb1744f45b3000aefb14b2477 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3088e77a1f2840b488d698ad3972c519 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/22b54f16420b44cfadf4db6062540a2c : SUCCESS in 2m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/35e2e1a1dd7b4c89b29dff8abe63de1b : SUCCESS in 7m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ad7a06013d8f4216a5bb3729b89c1cb3 : SUCCESS in 12m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/236b28f8a42b442ea832914ee5ca5cae : SUCCESS in 10m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fc50ef9c603a448b9dd2a0aa5ddd1b28 : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/824dba9405014816bd15479cf77cd512 : SUCCESS in 5m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dbf9f580b27a4b81b14db2ce0ff3addb : SUCCESS in 7m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3fc249edb87a4c74acfaeea889825e13 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a50f0a0eb6442139f340ac57695bed5 : SUCCESS in 6m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ab267e5d33e5448f9bfa0fbc8e23e435 : SUCCESS in 6m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/49a9294188a7464b974b37973153cf01 : SUCCESS in 10m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9dcf4c2b826646a2aa84c8be03ac5597 : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fbf0109ca43843c2bcd11ed28ea556db : SUCCESS in 6m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/751e52223ee0457095de61e432a3b19e : SUCCESS in 18m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/14b89d72a76f4ed9bed16df0bb28e027 : SUCCESS in 8m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f361abfbf178463e9c291af2ef7d0b29 : SUCCESS in 14m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9c937ae293c8446e955f592b7fbaa3a2 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99e9b0aec8e54e9290a11348b9125289 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43db378511204878bb25f99b34cacd64 : SUCCESS in 27m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/35c50567cce64a229fb96eef686e2cc1 : SUCCESS in 6m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ec3491bf2d974aa881b3755858b04238 : SUCCESS in 3m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d02e5fa50454873b93c3771f87c7aaf : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e30f054c79224e61b9daa690f95f1644 : SUCCESS in 21m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1001fbcc707f4ba7b5c9fe27eb3e09dc : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d47efd6da0043d6a5311c8b7fcd1407 : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f033a0d8a06478da4fb3f35895f0c23 : SUCCESS in 1h 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"510dd74b0700df1449ac52cdb81ed1c6d8801e4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:40:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a779687da118f22c15aa144bf09048fcd8e91d54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d09d89db5614638af9ab06ec27ecc34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9f70393d0d4256b567f61dc8bc2967 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc2d07e7ba53b173c04cf24aac607b07a63ceaff","revisions":{"dc2d07e7ba53b173c04cf24aac607b07a63ceaff":{"kind":"REWORK","_number":1,"created":"2025-11-21 14:16:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/968022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/968022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/968022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/968022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/968022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/968022/1"}}},"commit":{"parents":[{"commit":"6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9","subject":"Merge \"update constraint for python-zaqarclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-21 14:06:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-21 14:06:02.000000000","tz":0},"subject":"update constraint for os-service-types to new release 1.8.2","message":"update constraint for os-service-types to new release 1.8.2\n\nmeta: version: 1.8.2\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I0d149b5ca22489d248edd64e5e6505776a3f10ef\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I10b662a41cd37552cf3cef350ec3ea4a97900472\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc2d07e7ba53b173c04cf24aac607b07a63ceaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc2d07e7ba53b173c04cf24aac607b07a63ceaff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
