)]}'
{"id":"openstack%2Frequirements~968444","triplet_id":"openstack%2Frequirements~master~I78a8f857a2e8b90c6f84978a45a7545d84ce1215","project":"openstack/requirements","branch":"master","topic":"pqc","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-12-08 19:48:04.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-12-08 19:48:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78a8f857a2e8b90c6f84978a45a7545d84ce1215","subject":"Drop scrypt library","status":"MERGED","created":"2025-11-26 11:24:12.000000000","updated":"2025-12-08 19:48:52.000000000","submitted":"2025-12-08 19:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968444-pqc","meta_rev_id":"415c971bb18823c0df324b647095ae72cb2dcecd","_number":968444,"virtual_id_number":968444,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-08 19:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-12-08 17:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-12-08 15:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2025-12-08 15:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-12-08 17:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 13:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 14:29:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-26 14:52:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-27 08:56:04.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"834397cd349a73050142d334dea5a3a4c00f2acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-26 11:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"808f09dd71cfde82f4071b83db67b3d7035274c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 13:23:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f761ec7dd47447ab67d42778732229b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dfedf93e734c5998402023bfed4a5d : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc43a64731d4a9f9ce03cee807d97f1 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cac4f2e046324e728ee0afc44a4b41dd : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/012cfdbc92e1432a9d01e917037f7caa : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c6f1e3d0e99479c81fd63a349bfc0dd : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81fea566e8de4edf8f5faed800aaa9c6 : SUCCESS in 4m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/02a764927e5546929c15240c10576d3c : SUCCESS in 5m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7a8c3f7bf2d1490b89d06e79ffd50bc8 : SUCCESS in 6m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/08275236b9a743eb85f101ca1c574ae7 : SUCCESS in 3m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/845bf1edcbb8415db3c99965fe8dcf8d : SUCCESS in 9m 45s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b80a202f120a43b8b86ef26bad6af1e0 : SUCCESS in 5m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0863e20aca6e463caf203b4a7a6f5bad : SUCCESS in 7m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e137a2fcc66b45cebfdc3284f10934da : SUCCESS in 2m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ee8aa903e91d48b88a198da96a1a8451 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/09c0ecafe0f44a38b7fc1d58767db441 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/adcaa8569e8b4ea29ccf4b51e6e50a18 : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/49e620ca29b647d7be2a1050e5927fb5 : SUCCESS in 4m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bab6530babad4b2a9e36777f05ee88e0 : SUCCESS in 18m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce5ea725b58e4b2b8e4e6cde9048858b : SUCCESS in 6m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/50ed894f51a4444f8afa70c8ec240efc : SUCCESS in 9m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e609a086b9cd49d89d9841a1c3ce62d6 : SUCCESS in 12m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/216801d32aca40189c24ae0b5b1fece1 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a3094533c68d4bd7909d14475dc35c8a : SUCCESS in 8m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/82c8aaba88d644579aea5c16b268a5c4 : SUCCESS in 3m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f2597c5f48fd458187d5df47602dd5dc : SUCCESS in 4m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/107d58346c174938b869d58899baf05f : SUCCESS in 9m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1ce9df7dcb1740a2b83d3058ca5cb823 : SUCCESS in 23m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed47d94c2e744be4a47aec1740f28c64 : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5104ef4bb6404d4e910c67ec360c4943 : SUCCESS in 15m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6da4cf720afa46de902c704fcd02b9f0 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b45d85b7f17402e9d79497f3d04015f : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa32dedf121746b2ae7b52b96a530cd6 : SUCCESS in 29m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d368e44b0c60469b8df63a0bdf5daac0 : SUCCESS in 6m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2cda7bdc20af49a591a533ee8b80f36f : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dff20660d7574506a073a4a2fe282e42 : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/99a6d898cad34c78bfa32414174d79b7 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0ba123aedccf40d8a90c58f8f22a3cb8 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fe9f7fe349a492386941c5d9f315189 : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9981a68e915c4f8a885d0b3b27fa2f4e : SUCCESS in 1h 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7081839b76ae585d355aa886a5a39f95ffbf27f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-26 14:29:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"917f3ca2b788ca71e52e6700728a19f0bafcc7ef","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 14:52:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"77234939a1667dee6108f1d44f393ad62b4d86ca","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-27 08:56:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e9463499ca11d4041423c94f5d41cfca61a200b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-06 13:14:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9b91ba8a59666fbd218ded582e0ad3e1fa5fbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-12-08 15:17:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f549ae3271722c1e85801054b7bb2f87c5f24647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:35:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90579cd5d6604f9d8250df2093a7a184\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a143e9e3dfd5469a92f553908aea8d55 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad8d120c8824297909728339e0da422 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c855e59a8fe4fd79ca05e09582c628f : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ef964ab81a24744a7c98cb846de57c7 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e0934fa2f1949018ac571be17d2d1ed : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8d9ead943a84a09a7b9e9d591c841ec : SUCCESS in 2m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e3e44f698ec44a5b19c8df5adda433d : SUCCESS in 4m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1bd2b815c93141cebfe13609a45a27b1 : SUCCESS in 3m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/16a456c93e1f4fdeb233c86d07edd69e : SUCCESS in 7m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e1080f60e3b34771bb3b853c9719ff4b : SUCCESS in 10m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/22c77757101a4a81ad696f867d065604 : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0e1214c37f4a4a9ba34e0615a6d28e4f : SUCCESS in 5m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d18cf1bb8dbb41798ee75e791877b412 : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c0c0df2c7cb04c158e0c7848a0f32c5f : SUCCESS in 10m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/310e950cc8324b9684acf1564e6a23d8 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e800d63c04514795a154c8f1920772b6 : SUCCESS in 5m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/30b9c7e44cf24201b822062e0f75e957 : SUCCESS in 3m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cafe8013994d46c7a3f4de6bb76c16f0 : SUCCESS in 17m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef31da65891547e1b36c5fcd02605383 : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fc2f8c43ffa74c97bca52209102f471f : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/70a4743bbf1c4a8981bfba55ea9ab61f : SUCCESS in 8m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b3024e20edc84c4caaa0946fe6944764 : SUCCESS in 3m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/33da834258604f54914c32cfea31af3b : SUCCESS in 5m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/88bbba629c46448cb4746d17a7283eba : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4261bdaf3d9146be87f592043c9f1843 : SUCCESS in 3m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2071783effc442c89ad9564ab7cc1b74 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4c281755b1df4e919f3eb020dc1b89bd : SUCCESS in 23m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0015e7a7a94d4cb293090adc4c4b506e : SUCCESS in 4m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/40d6dbf78e7e4e41bfe9afe5b3dd8911 : SUCCESS in 13m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/271e6770b3454c20bc27a9e50d8f11dd : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4630150fdfff4233bf05397341c93517 : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f09d8c8b87ce4584929535b112a5aa62 : SUCCESS in 24m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9e7d4409276f4438b12457d9f7d261b0 : SUCCESS in 5m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/47b167d4f9714d8fb7fc3d2000163ff6 : SUCCESS in 3m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41a1d4eb0e094b8db951b9d962d65a05 : SUCCESS in 7m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/459ee9dd6c8b473db74b99be5b1272b8 : SUCCESS in 17m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9e474d951edd447dbab0b9681aa0f7b4 : SUCCESS in 7m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3e75b9eca0a41e6a31bea65e0cb6c4c : SUCCESS in 1h 04m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9677f81244841deacaf13e97237a271 : FAILURE in 1h 12m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b08ae700bedd791ff4640e4ed77842c1a1e94f34","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-12-08 16:36:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f0a54ba270ed3ee34a7f5683255f402af0fe4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 17:43:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfeb7ec8607f4be69e42eb6804d729b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3820602f652c481e8a298b8fe503ed71 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f0014eb06f44aeb7e551dad26daa2a : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92d6c4501e23417d8a2fb84c5f382855 : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31e3e274edd44b3290404b8e1f083c6b : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3228afc143894364a9dbe100e6946089 : SUCCESS in 1m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d905ca31bf2f415085675c2112848871 : SUCCESS in 3m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/271897943d624a3081e2d3ca69730acf : SUCCESS in 2m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d31a4c6cd9c44e0cad2e626f4075c1f2 : SUCCESS in 8m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/657040f8e5244d589bd4f773cfc304da : SUCCESS in 3m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3dd78e4f55c0495f84bc9cbab5fd3bd4 : SUCCESS in 14m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2460df47603b45b2ad6d9afaf40705ca : SUCCESS in 2m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/05711fd33b134d17ab1314fa2db1d5f5 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/44a9dbe897594e3089071942bdefef7f : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2e2811ad02ae46caa4d9687fca0d7b79 : SUCCESS in 6m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a954e2d9cb214c519080770622b3495a : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/49ae07521c9a4ff2ae9afaf5272acadc : SUCCESS in 3m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/820c4358603e4eb3a5ee373ea3528719 : SUCCESS in 3m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ffdc3a237e054e1fa03badf1cc7cdbad : SUCCESS in 17m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecf702dca34b406dba66bc14d06de664 : SUCCESS in 8m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d4e72c0cef6643978a49002d64c6e623 : SUCCESS in 9m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9361820f57ae448d9bc8b967e1f0c254 : SUCCESS in 14m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0048c515dc37428aae16730bd6655d3e : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cb936f86873c432dac89b10c3510bb8b : SUCCESS in 7m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b1ac1faeb0b84754ae8421367361c10e : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b0f91ee0323f4ec4b921fafcecb84b5f : SUCCESS in 2m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f9d9010cde374ffeb0204f2d08599aa5 : SUCCESS in 11m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/19d1369c7c8c4683a304c6bfb34d8ccf : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4205015ae3e3407d8907a4e4d650c11d : SUCCESS in 4m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9fe1479242f144bab30f2abcc92b4677 : SUCCESS in 10m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/88c11ef0b74b49e5a473c8f2945acf8a : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd3c7310ede046b9a3aa001289cfd187 : SUCCESS in 2m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a559d9ec90834ab89085c142cd60b1c2 : SUCCESS in 24m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9cc326096c94402981befef29104615e : SUCCESS in 3m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a947f59945ba489a803b8b71916da0bb : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec718e3e95114e0c99ce953d9597e1fc : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f63f0b887e6e4d81830420aa627e21b1 : SUCCESS in 12m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a8f74c73bfa7404093ab729e7e85eca0 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1434c7da1ebc456c984583b4cf8c64d8 : SUCCESS in 1h 05m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c65ba4473d264a259567f89bd11d53b2 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":2},{"id":"957e535d83104374febdeea96a2a5c989392fd69","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-08 17:54:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b77a8d3814794e3075224d8daa2a60237c58b9f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 17:55:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7651cee43d268672ba18717f03b9a7f0ae73d3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 19:48:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f4c29ed21c40ab84c4aa777a6b71fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971eabc3f4b54223b11f8348068dfaee : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84c515b16d24dccb74346ccf856f3dc : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/695fa5b2eaef4b328f64fdf0fda10fda : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2511667f36f24385b79583c0717f32a0 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4e8f5e283bc4240bcc2c324f6210353 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ad49622ec574e1a9f5c32ad2440132e : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f75890a73e3e48939e249de6f664030a : SUCCESS in 5m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b227560a684a4d4e981574f614bdc128 : SUCCESS in 11m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7fe143f08cc949ce914fbe1860151460 : SUCCESS in 6m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0b330d668bb948d2aff9f22b1549780d : SUCCESS in 9m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8d7ac9f53d7048c3ae7c6571db285337 : SUCCESS in 6m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/67cd341d1c7e4a9398cd619f817e9994 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c496f16e7ca24e41bf8362584a86e13c : SUCCESS in 8m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9075464a858b41f592e064aff99135ff : SUCCESS in 17m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/222b56e8be0a45108dc15afedd0ec6cf : SUCCESS in 9m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0f1af8ebd0314007b07ee826fea81a37 : SUCCESS in 9m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e0626f360f564e879b0ac1e9924b45ce : SUCCESS in 12m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8465b6bc70244c31bc2bc3e20125a879 : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c1ecfa71f19f4c128ff318469fa5b4cc : SUCCESS in 10m 54s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/29f4987c79b64b16bec8a82bd0b8c6ba : SUCCESS in 26m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/07dda868c7c7446f8ccc1723dfa0c5fd : SUCCESS in 8m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/db9bc55e587c464f9e2ccf75a7840941 : SUCCESS in 8m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1171b03e4eda4c1c8c4e84d7af1cbee9 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6068ac47200843da97afec15ec4807de : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8370dd843b85439287d26f7ca2290e8b : SUCCESS in 28m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cd1943965a0c4739be5219061ff2deec : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/013d2546219f46f1b47cbd4140145b8c : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5806267c9e543a9a3a5fc27d5e58518 : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ca287da4612449e9b6e74de314f42243 : SUCCESS in 13m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1c8daf5380bd4652bb163883e4fcdcd5 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15b5b3e0f3774760a6be3473114787cf : SUCCESS in 1h 06m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00fc2dc16dec4c94934dcd3767e5b203 : SUCCESS in 1h 46m 40s","accounts_in_message":[],"_revision_number":2},{"id":"942eb4ae8a55153df36021d7eac5ab0a48dbadeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 19:48:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"415c971bb18823c0df324b647095ae72cb2dcecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 19:48:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735250557bf0463f87b7f231dffdcf39\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df51e1ccdfa045dcae5a6fca2c47c811 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36d32e876a0e1ba48a00160fe71026e9b534d6b2","revisions":{"04b7726fde63796668bd548362ef4b0456313400":{"kind":"REWORK","_number":1,"created":"2025-11-26 11:24:12.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/968444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/968444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/968444/1"}}},"commit":{"parents":[{"commit":"281d29299f1ca6dfbe5e0bf762e364ded93a6e36","subject":"Drop every reference to passlib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281d29299f1ca6dfbe5e0bf762e364ded93a6e36"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-11-26 11:03:15.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-11-26 11:17:52.000000000","tz":60},"subject":"Drop scrypt library","message":"Drop scrypt library\n\nWhile scrypt was revived after a period of no activity at all there is\nno reason to keep it listed when no project is needing it.\n\nRelates-to: https://wiki.openstack.org/wiki/Post_quantum_openstack\n\nDepends-on: https://review.opendev.org/c/openstack/keystone/+/968442\nChange-Id: I78a8f857a2e8b90c6f84978a45a7545d84ce1215\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04b7726fde63796668bd548362ef4b0456313400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04b7726fde63796668bd548362ef4b0456313400"}]},"branch":"refs/heads/master"},"36d32e876a0e1ba48a00160fe71026e9b534d6b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-08 15:17:19.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/968444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/968444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/968444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/968444/2"}}},"commit":{"parents":[{"commit":"f5707cc58f7152a82d93933a9863587af5f0ffa7","subject":"Merge \"update constraint for microversion_parse to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5707cc58f7152a82d93933a9863587af5f0ffa7"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-11-26 11:03:15.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-12-08 15:17:19.000000000","tz":0},"subject":"Drop scrypt library","message":"Drop scrypt library\n\nWhile scrypt was revived after a period of no activity at all there is\nno reason to keep it listed when no project is needing it.\n\nRelates-to: https://wiki.openstack.org/wiki/Post_quantum_openstack\n\nDepends-on: https://review.opendev.org/c/openstack/keystone/+/968442\nChange-Id: I78a8f857a2e8b90c6f84978a45a7545d84ce1215\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36d32e876a0e1ba48a00160fe71026e9b534d6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36d32e876a0e1ba48a00160fe71026e9b534d6b2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
