)]}'
{"id":"openstack%2Frequirements~968655","triplet_id":"openstack%2Frequirements~master~Icc0d7a32e56e49ab1d689a6280f3434c2e3d9f16","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icc0d7a32e56e49ab1d689a6280f3434c2e3d9f16","subject":"update constraint for sushy to new release 5.9.0","status":"MERGED","created":"2025-11-27 13:36:56.000000000","updated":"2025-11-28 15:05:29.000000000","submitted":"2025-11-28 15:03:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968655-new-release","meta_rev_id":"333e7e1209af9fdc8359c1d80ef2030562bbd60d","_number":968655,"virtual_id_number":968655,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 15:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-11-28 13:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-11-28 13:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-27 15:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-28 13:14:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cd9f0f3ca742aec9db94d5d6c4ae2117be4c9fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-27 13:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47c4f59dd111b4664bc8d2642bec5ad0e9232b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:27:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/069c143999c64897b557cb10ce30dfdb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4a5d2a86e14f9e81f4a75e6ce9c116 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221f375502644b769862bd8cf1e48dd2 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6349c74869a477d8c61269834b68beb : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b24372458fa74f5cb00708858124894c : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d3d9c72e65a4247877bf65de6843d08 : SUCCESS in 2m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2befb360cc61461bbba89a7ca41fc6dd : SUCCESS in 3m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dbcb4abc708746a694ffca4977e96c71 : SUCCESS in 3m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f871b10c5eb04719842db84016dbc86f : SUCCESS in 5m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a11eaa69585e44f395b811ddd0a47f84 : SUCCESS in 4m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/134498c1f13341eb83c14d3b6e90ab05 : SUCCESS in 5m 36s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5a206293161b4ad3a65f9d72b9c0f2fa : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/650bc1e0060142d9a7c7ae4bd3554aad : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b61be49758a649df808c940b18457fcc : SUCCESS in 3m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ca0f7da3864d4c4984a864a7d7c73a64 : SUCCESS in 12m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7408ab2b9e2040ef8ff5bd0dab458548 : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/169cff9dd1be4990a2cddd12effea502 : SUCCESS in 5m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4a5b31d782c247cd81b8c3f7808943bd : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/88fb5a455abb47ec9b415e09a77918c6 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2532db6507a640d39cf4500f31640547 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6fd7d4a061a34c2989e428712f627e98 : SUCCESS in 8m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0911cef55bc9457b8cce93ee3477194b : SUCCESS in 12m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5292f4f7c2064243a6da4cc7e871f532 : SUCCESS in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ebba1643cc754a05bb462eb939ba438e : SUCCESS in 8m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/71f4d5f83cab42d8b36fd989ad2e3e9d : SUCCESS in 2m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/785fdfc6df7444bfa548f35add8c9a96 : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/66b562f71a6945c2be268d3b350a0df3 : SUCCESS in 8m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c62edf83ac50447cbef10f316a7bdac2 : SUCCESS in 22m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dcb4a596867f40eda62a89142a488787 : SUCCESS in 7m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/de8f1fa6dc7845a8853bf93f6d40fc8a : SUCCESS in 13m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a0f7541caff485b9b9f53dbbadc44ea : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fad16822d994d0ba6123dedc26528ef : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52906fdea23f4e16be921d723b9e322b : SUCCESS in 24m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4baada2de74740d08bc1ac67b6c1cfff : SUCCESS in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4ba80520200c4e51825aaedf919824ab : SUCCESS in 3m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab6f57199f064871b936cdd3019bfd68 : SUCCESS in 6m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fa388eaf723748fe9ca84236ddd4a303 : SUCCESS in 7m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9e482f34952e4babad8c2dadf23e9ec5 : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c25347a94c444b8a058623fa9a379f7 : SUCCESS in 52m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10763dde27144f2aaf27bb527e9976a2 : SUCCESS in 1h 40m 04s","accounts_in_message":[],"_revision_number":1},{"id":"00ddd66200254906dc743c37834d25dec7b65476","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-28 13:14:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76ce9cbc1f70e2febd8ea0a0f84845ea90bfc57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:14:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63d470350a1aac5c9d01ed76041001ce0f6b2878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:03:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbe4a775cb945e3a97b52d224256238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa81cc7ef53413c9b17304c44eff72e : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922e812609884c4f8fec373586cacf04 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b2f934dd3374a0380ae4baba35514dd : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b63e32fb920f47229deb8df2650d8c67 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6b70dad844b4d9cbb318d0847f26186 : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e9d576033ed479e8e9c7c7088ad48f6 : SUCCESS in 2m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd6b1a24a7d24932bb9a3d4fb8540655 : SUCCESS in 5m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5c6172277c6d47878328dbc11f2e885a : SUCCESS in 9m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4a14288f8a0846e0809e2ce51af9d8ee : SUCCESS in 6m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7fbdb6efe47545008f27f42f5498d069 : SUCCESS in 11m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4af9170044364ae78cfaafd10ec1adb2 : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/83035a58e4704bab883f8c737d247908 : SUCCESS in 3m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d3cf314a619f4b368703ccfc9fa8ad4b : SUCCESS in 3m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/50b0840d890a461fa0481aad5e85b8ae : SUCCESS in 17m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a91e0ee466124bc79ee6ed18f77e57fe : SUCCESS in 7m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/43402f795bc349b0983c01ed55826f4a : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ab8ffe81cffe441aacc37ed5fb6bb69e : SUCCESS in 7m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/11bb77aa04de44a9bad5b64972297541 : SUCCESS in 1m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0e1be66a19c74e9b96e548383adb4f02 : SUCCESS in 8m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d7242c9bfcbd41169681039d4f4bad5f : SUCCESS in 19m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d85adeece0e048a197ce74f951b64af3 : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/583cbfc6ce15416fbff5451244e1ab32 : SUCCESS in 8m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/760baa61f8314c29b2881762ab64548c : SUCCESS in 3m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20540dc26da34c5994f1f5b28c0b789e : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a86fc93be4ce4d79afec06573330bb27 : SUCCESS in 27m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b74676c5d6374ffab50236cf294d09e7 : SUCCESS in 7m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aa1c94dd646c4366abb45017130b4557 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f2a55ae36f0474e98aad4adac4d495d : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/480f07b9d92d4cb2ba44addbc22cda81 : SUCCESS in 18m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/18feb6b68fcf409d9451f4ec65428ec6 : SUCCESS in 5m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08ee89d84e214d8bb2be9bfe6b14445b : SUCCESS in 32m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/954124da175e4d0c8d04afc7e707d753 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a975d8f3b58e8716577bfb4e87ec14348c349d07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:03:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"333e7e1209af9fdc8359c1d80ef2030562bbd60d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:05:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a6c8bc90294308bf0317e2b572c3e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebabaa9f0a84c0eb0564370c24be969 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92a8ee8b9a6fce7f316373b384ac9d3ac139fc45","revisions":{"92a8ee8b9a6fce7f316373b384ac9d3ac139fc45":{"kind":"REWORK","_number":1,"created":"2025-11-27 13:36:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/968655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/968655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/968655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/968655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/968655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/968655/1"}}},"commit":{"parents":[{"commit":"689ece5e1d9f3d684aa924209990b1e53c17c1fa","subject":"Merge \"Remove requests-aws\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/689ece5e1d9f3d684aa924209990b1e53c17c1fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-27 13:26:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-27 13:26:55.000000000","tz":0},"subject":"update constraint for sushy to new release 5.9.0","message":"update constraint for sushy to new release 5.9.0\n\nmeta: version: 5.9.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Iefbfd789f8d296de66d5c1808aa34491e74f9520\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Icc0d7a32e56e49ab1d689a6280f3434c2e3d9f16\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92a8ee8b9a6fce7f316373b384ac9d3ac139fc45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92a8ee8b9a6fce7f316373b384ac9d3ac139fc45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
