)]}'
{"id":"openstack%2Frequirements~968707","triplet_id":"openstack%2Frequirements~master~I8057eb6daffe61900132e5372729750210cd9e1b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8057eb6daffe61900132e5372729750210cd9e1b","subject":"update constraint for oslo.limit to new release 2.9.1","status":"MERGED","created":"2025-11-27 13:52:46.000000000","updated":"2025-11-28 15:04:40.000000000","submitted":"2025-11-28 15:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968707-new-release","meta_rev_id":"a8d9f670f5025887f5f228bdeea49fc76a3d8860","_number":968707,"virtual_id_number":968707,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 15:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-11-28 13:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-11-27 18:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-11-28 13:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-27 15:22:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 18:03:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-11-28 13:09:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ab13f238956f33a6fd7c37e87c5a6ef919e313b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-27 13:52:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b49af2a1bf000fdb3a857fcc33a79a2ba079a940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:22:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6514c8225a414c469903ee32a8fa9268\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c606a87975c747e3b446695ecd9582bc : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae1698dcf1e4294bd2c149f451219e5 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a17483c653944feb2e95b84078701e1 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa48857e1d984086928b9cfc17325185 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34cd8d3297de4635ba60872daed686f3 : SUCCESS in 2m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/189a19db800c4dbba21159c2e413d408 : SUCCESS in 2m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c41ea17b722439aba55ac0ab36e8184 : SUCCESS in 2m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/770404eaddcd485f8fe0f918dcc03bca : SUCCESS in 9m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/659da7e678144880a48fe3d90318a799 : SUCCESS in 3m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/97d798f5b6ad42eebd5467ed1691f6f2 : SUCCESS in 11m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4091b0f4f286463387726ce864b378ff : SUCCESS in 3m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e3c5ed05eefb4e55be8233635373e757 : SUCCESS in 3m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5835c42e77504579a8d55b4422eca7f2 : SUCCESS in 4m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3ad00883e9884ed7860f4ec4d5c6f335 : SUCCESS in 10m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8724247ecd81461db69afbb7f9c37d27 : SUCCESS in 4m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/42e5b636ee024aceb6d9a726839967a9 : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0538a1fdf7f648a488abfc9a158321b9 : SUCCESS in 3m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cee186680dff40088e5acc2e5ec8bc6c : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a564752560cd472b9ae856c787f8a1d4 : SUCCESS in 5m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a16f16acb26349dda30a1662f827682f : SUCCESS in 9m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e4795754a53344df9e620e7d90b11e90 : SUCCESS in 9m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/59140e6890d74b7e81a7d816ac848634 : SUCCESS in 1m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/daf3bedf4a9c4b27840a785e76639aad : SUCCESS in 9m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/36b99050b7e447fb9b05eba0f1c9cee1 : SUCCESS in 5m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/285490f35ac847adab7234e94cacdac0 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/77b34121a81c4cebbc17b91ed36c72c2 : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/dc9c24a1382848e782fa9cf3ca3d4247 : SUCCESS in 29m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8bdecf8d07cc4ee69efb3f1ae7a15783 : SUCCESS in 9m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b22d727376374ff49f4001e74a56322f : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4bfcef0af2794664af95a659595957ab : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/090c4facb17c4ceab1f988c1b3bab14e : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/413b9676d2454207b9b9cb1bee03c921 : SUCCESS in 27m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a52c4425b4ab45b79c87c8dd0dbc59d0 : SUCCESS in 7m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aff878a0087041d69761e76a04a2a158 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b910c3d6b6d4f538d7530a8ebb4505d : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/75573f53f914468bb006cf6ac72c3b71 : SUCCESS in 7m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e6238c561f284da5ac10dbc23289eea6 : SUCCESS in 6m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f8fb2c21dc74c828a18ea0c1096e52a : SUCCESS in 32m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa11af363acb416ca2d732a9e2b88b00 : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ef21d34e7717f4d804691d1c2807d0a7aaafad5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 18:03:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a80055418f8002841783a49210e1fca8563035a6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-28 13:09:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2dabd527b0c412497d0f0b451a7c203239fecfd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:10:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"216a72831baef8fd1b15954233115a5002b2d244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:03:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e031733128d34fb3803cfbda0104f10c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836d9893b35745cc9adebb44b9268b89 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed0db7c7ce474299b043d794479398e0 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1ea14c1172d474e9b779b0ef09e2f8e : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e88dfafa5797417bb6e0870c1cb032c9 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7dea14b4f574dde853ac732e67d2b02 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1c2767bf625455eabb64f12e51dda15 : SUCCESS in 2m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f9de3c8bf5944fe8ed00351d50cf790 : SUCCESS in 4m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5974cee2cd9f49d7aebe34e96fd71b1f : SUCCESS in 14m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/46ed5f27006e4e82b9b3488cbe404c6f : SUCCESS in 9m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/46b4b97f875749afa63d7d9fa8ddb92f : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/84b92e816d0e46fc98266aee96051eaa : SUCCESS in 3m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c55bbe47f0d44f6480e5699f780f0a34 : SUCCESS in 3m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1e3bcecb19174c4f86a21a4b7b303477 : SUCCESS in 8m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d80165d215e646ea880173450dd63662 : SUCCESS in 18m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a57f8fff5b44406bd4bf343fc3848b8 : SUCCESS in 7m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4849d9b09eb640a7952295ec477ea86d : SUCCESS in 8m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d0bd770240934a29b884ba85e577f319 : SUCCESS in 9m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ec7304ae83dc439f82ea6a553528e484 : SUCCESS in 3m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/887c412fd6294c649dbfe3715bc75e69 : SUCCESS in 10m 15s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/390d083a29e4446e853c8515b854c5b7 : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f99630eea99499f978bf1a17d3eea1f : SUCCESS in 7m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7901ce74c3b94c5ca6c1e7a38f729740 : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0cba5de88f294db2804385f1fd306ab1 : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/db29456fe90747979c0d9fe6d90cb3d8 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b14ec832e9274f00b28bfa12550bf042 : SUCCESS in 23m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0c2b28b27d714063925bf5a810d4991f : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8179e89e45a5475d88299b988a20568f : SUCCESS in 3m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/199c90c1c0f04694bbe0f00a5a080818 : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9b46a391158e4a81af20b8814deda58f : SUCCESS in 18m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/51fc11dc3f794273877f1d9d3a988fa5 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e71f97a6f704e6ab5f138ed1e40323f : SUCCESS in 32m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2440e3a380e4321b3bd1823cae0bb60 : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1fa99c5cb1c4fee959aab076eb8b49ce80b60ee5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:03:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8d9f670f5025887f5f228bdeea49fc76a3d8860","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:04:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65404aaa71e43a4b216c0814fb4d53d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f655b818018e43fa8ad6bdb19cc166a1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f30b92372fe446ea4c2dc9f9f75c313f9fd99a64","revisions":{"f30b92372fe446ea4c2dc9f9f75c313f9fd99a64":{"kind":"REWORK","_number":1,"created":"2025-11-27 13:52:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/968707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/968707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/968707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/968707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/968707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/968707/1"}}},"commit":{"parents":[{"commit":"689ece5e1d9f3d684aa924209990b1e53c17c1fa","subject":"Merge \"Remove requests-aws\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/689ece5e1d9f3d684aa924209990b1e53c17c1fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-27 13:42:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-27 13:42:44.000000000","tz":0},"subject":"update constraint for oslo.limit to new release 2.9.1","message":"update constraint for oslo.limit to new release 2.9.1\n\nmeta: version: 2.9.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I375dc15f9461a6b4e06a6e619c5f6e3fafb812b9\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8057eb6daffe61900132e5372729750210cd9e1b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f30b92372fe446ea4c2dc9f9f75c313f9fd99a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f30b92372fe446ea4c2dc9f9f75c313f9fd99a64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
