)]}'
{"id":"openstack%2Frequirements~969560","triplet_id":"openstack%2Frequirements~master~Ia81df004eed76f09a80ea51b16f43288872f9ffb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia81df004eed76f09a80ea51b16f43288872f9ffb","subject":"update constraint for cliff to new release 4.13.0","status":"MERGED","created":"2025-12-04 06:06:23.000000000","updated":"2025-12-05 16:09:09.000000000","submitted":"2025-12-05 16:08:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969560-new-release","meta_rev_id":"fe592da5b8f9acb781273ad01198dfe487d3ac31","_number":969560,"virtual_id_number":969560,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 16:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-12-05 14:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-12-04 18:25:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-12-05 14:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-04 07:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 18:25:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-12-05 14:07:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cfe815a258f8ac6b28ca18f217bece0f4460a2f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-04 06:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98d01ef5309830cade596d96cb89e29c07db9df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 07:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7eef131e7f4934966e21e2c27915e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407f5dbba1004cf1a7e89d5410d65606 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4dd676c3b749c2bd124eb224bfeef1 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b96e346d07114472991139ffda2cf005 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92e7397d2f5f4158a05485cb2eeb2dc8 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9953a09e79804fa7a6c75953c9de040b : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1228cd6de0d044dcab4464159e65884b : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e1cfea8f1d7546cd9a8be4ad6b0c3da1 : SUCCESS in 4m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9e05c194578349019fb96be341f4a86b : SUCCESS in 3m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8823a1c4ee6843f5b3ad3700a17e84c4 : SUCCESS in 7m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1a75dfdef564cd68f255f007d7eef6a : SUCCESS in 12m 53s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ee085c0f5850488cbec2af2ba4d5680d : SUCCESS in 4m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2ec9735794f547cc8ce038a59f0125db : SUCCESS in 3m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b0f0f2c4e7b04fe7878613cbcbf6f439 : SUCCESS in 5m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e8ed156f0eba4cce965bb046ce71dfb5 : SUCCESS in 11m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/af948a2810c3460e982edbb398a2b0e9 : SUCCESS in 3m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eba7f93dfafd471b85067b765b7c8c4a : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f85a7fd0bfbb4a70853aba0974090480 : SUCCESS in 7m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/eb03374cb9d04cbbab7a341f0b3ff131 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21ce392303ee4b65829199d425cd13ae : SUCCESS in 7m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/eaa56f897d1b45f8a6c73b329c378e6b : SUCCESS in 10m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d4b76d03d2f545e0b396cd3a38e1b987 : SUCCESS in 9m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/46ed1d5663bf45b1aa469e16f0ba989d : SUCCESS in 1m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/304a4a2200d7437d8dd15aca9561860c : SUCCESS in 10m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1ec2919f3e594da48db00e4e9d80afb4 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0c2ac3ad14e140efb2b0521df3c7362a : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/89bd25da59554fc1b1c097d247ba3fd3 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/593b38187b2d49cf9e53c590940b19d2 : SUCCESS in 29m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/732bd031e5b14429bff87251d81c79cf : SUCCESS in 7m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/87080a1379c34dc0ac473f30af3f0c85 : SUCCESS in 17m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2c966416dac64b3d96739133d77c71e4 : SUCCESS in 3m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78a540bc5d1e4f36b639f6cf71ebab2b : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18182c45baf743eca75bf4fd5cb286e0 : SUCCESS in 25m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bd4f41a800a3453c8ae7b6f4cd375f65 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/50e6dc403ef2455f9ec4e696c8ab9a2e : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/408ae342792444b5aaefc1c2fbb54be2 : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cffcc4ba2bb9458e9f294ecfbd388321 : SUCCESS in 20m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/408848d717a648eaa122451292c4c598 : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f40fb44c1afb4f0ea8a05f056851d742 : SUCCESS in 34m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf96723f1b964f129ac1f4f0df0e9521 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"00dbe4415ead23e7b7e50589cd0c3d465c2bf58b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:25:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f94d77c11979c5a8f05c24955a4100a03fe4ee59","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-05 14:07:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57880081a14b9de698d4de9c8a3b8ce5a245f07b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:08:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab8768d997bd4237d8a5d590494437af07442cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a352d862224bb6adec2c9e6003ba91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde7fc77087e4103a2450764ebd6c56c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd9a86bbb3cb409db96d51601e869959 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1aa84fb75b954c2fa8767e610326689d : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38cc0268adb24793bac86dff02b0120a : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c5e9a78f4164be084c190d804459a98 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eec053150760442d99b7975ac25cbdaa : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ab23d338ca1649a59b20fea14cf0ebb6 : SUCCESS in 5m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6f1df6e9f78e4575bda354e8fdcda96e : SUCCESS in 4m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/460853263c124188bbebc683a9c14667 : SUCCESS in 7m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b1863704a39148a494aeb1674a1c6390 : SUCCESS in 10m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/81b343257d4c4575a9d53a7e315bf8ba : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fbd2d40c2cfe4046803b93db6bec5900 : SUCCESS in 7m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b382be69163497988b81d7f100b7da5 : SUCCESS in 7m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2b335c73c174485291817c41b22b83f0 : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd8cffdecf0f45be8b1d0631c064f96c : SUCCESS in 4m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7fc0ebb3e9454dfbaa012180f09fe09f : SUCCESS in 9m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a53bcb85087b4ac7a644453ed29d187f : SUCCESS in 10m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/efd0fc1b162a41abb8a0aa1846dd0b74 : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cc0696f716ab4e569d77550fe6a341c2 : SUCCESS in 11m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/48df8c646ace4d39a086f009d33c8690 : SUCCESS in 31m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/140b7eea55054accba5b94edb1d3ad9f : SUCCESS in 8m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9a172205d0264eb6a9019521032f32a2 : SUCCESS in 9m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cc89f7f02950455b903c30bfbfe63cfc : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27433f2a53a84fec8b7b570b485c6ff2 : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf30c19230014345b219b5fcc87d32e3 : SUCCESS in 27m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/693fbd543b1e4a2180e777b86de08977 : SUCCESS in 8m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3d0843f0e90d4e21ab8ecc4f92309d49 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb094a73921c4557b454f60ec2300d83 : SUCCESS in 3m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3b3c2887e4b84b23a8b0faec307e04e3 : SUCCESS in 22m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a3be8d1b6cf443e8a021da0fa4da7fd7 : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff8bd9af6bdb4a558d543d0f59a7bfe0 : SUCCESS in 35m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4184b5802884bb7bc295892d25055bc : SUCCESS in 1h 53m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2d806b1b7da3f27a4b19138167c2c6a1a9bb5f5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:08:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe592da5b8f9acb781273ad01198dfe487d3ac31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:09:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3390e7d58d145c38c08d5f7ba696b5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2b69ff840341848b3f587c30fb0dd2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7949b4daa158c2c67f4ee9e8c0b946776189c207","revisions":{"7949b4daa158c2c67f4ee9e8c0b946776189c207":{"kind":"REWORK","_number":1,"created":"2025-12-04 06:06:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/969560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/969560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/969560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/969560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/969560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/969560/1"}}},"commit":{"parents":[{"commit":"44dbd103640a38272df3ad29048b4c5dece2c1c7","subject":"Merge \"update constraint for sushy to new release 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44dbd103640a38272df3ad29048b4c5dece2c1c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-04 05:56:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-04 05:56:22.000000000","tz":0},"subject":"update constraint for cliff to new release 4.13.0","message":"update constraint for cliff to new release 4.13.0\n\nmeta: version: 4.13.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I9c6d30e6341490a64faa5c4367715794f08bb2e4\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia81df004eed76f09a80ea51b16f43288872f9ffb\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7949b4daa158c2c67f4ee9e8c0b946776189c207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7949b4daa158c2c67f4ee9e8c0b946776189c207"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
