)]}'
{"id":"openstack%2Frequirements~970118","triplet_id":"openstack%2Frequirements~master~I01df5e3effdf03cfb79951895cf650ae81ee0a85","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I01df5e3effdf03cfb79951895cf650ae81ee0a85","subject":"update constraint for openstack-release-test to new release 8.2.0","status":"MERGED","created":"2025-12-08 12:40:46.000000000","updated":"2025-12-10 18:29:59.000000000","submitted":"2025-12-10 18:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970118-new-release","meta_rev_id":"3267a00402db856af1bb214bc8ca69d779118335","_number":970118,"virtual_id_number":970118,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-10 18:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-12-10 17:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-12-10 17:08:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-08 14:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 17:08:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"264c5dcd12e35ecec5abad8f111d7e1f3c6573ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-08 12:40:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62d16539aca0ea6f15044863299e8544365b6546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 14:24:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/069f1cef5dcc4e2fa09ec81023bde634\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9482cb8497f42da831683b12a199b08 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f577ec10b6b49cbb82a93ce156b58a6 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5ad0ca9521d4d62a48f6399169c5fca : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8415cb5109ec4f72b685a9c126fe2863 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d3228e331224da0972be792d212e9f2 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c836523a00b7441da6d0c25f9b241036 : SUCCESS in 2m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c349906d8d8747b1b5805285b1e74b0b : SUCCESS in 12m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8957f7b68f9745ca8ea100fa7b283064 : SUCCESS in 3m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/218314ed11bf4b2ab3c7cc0c1537a7ee : SUCCESS in 5m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/aa95baf152ee4ccc91007da673181080 : SUCCESS in 14m 29s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c6fe083c4d574135a8f3990ac88e384d : SUCCESS in 3m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/407531e7ebc24f80a69aff659d09c65c : SUCCESS in 4m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fc622ade254c4a83b17adca5a2b39879 : SUCCESS in 4m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/95201b9a54264974be05549bf393e9bc : SUCCESS in 11m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9ede9806fc9348a9bbfc17cda7c07f05 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/683b53b1fa88403eadfd1a161ee1325a : SUCCESS in 2m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3958fe4ac55f476d83fbfe9648375d05 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dc53964f4ff346ddba320f682ddd8c14 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19a1617eebe9441f8547dde00988efe3 : SUCCESS in 4m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7fdc67540f394c018c7761a0f24580d5 : SUCCESS in 7m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2570a475f6554eb8852e90ea599afcce : SUCCESS in 9m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/427a7efde20b441c9b4ddc2bf5358699 : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3ff9c80d78934695b42b75a0a5b772b7 : SUCCESS in 5m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/79b112504413410c9092e5ba2e16e7d9 : SUCCESS in 3m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b4c150f8738f44a394945e35aa614006 : SUCCESS in 3m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e38a59bef89b449490c2aa1df982e12e : SUCCESS in 9m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/921f367f8d574aa3849a5f50c942f8e2 : SUCCESS in 23m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/499b90d3eac941ac9dc264a242c37828 : SUCCESS in 8m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/95380ccb0fff4f9a8fd8e199310a49ba : SUCCESS in 11m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1be02ded240245ccb4b59c2b30d1cd37 : SUCCESS in 2m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9dfce15191314eb39c37b1cce572c46a : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/302c6e5f607945d8bb22b68396ea1e7d : SUCCESS in 23m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/181f8d49ae4a40b88df450307e0a8896 : SUCCESS in 6m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5bc801f9c73a473099b1aa7707da97cf : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af33a8d2ab214dc996fc25493f91b012 : SUCCESS in 7m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/28513b0c078b40a792a980e568f44042 : SUCCESS in 19m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4677bdfda3954c298a185bee1cef4503 : SUCCESS in 3m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/440d54ccf55846cba63863d673961314 : SUCCESS in 1h 03m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6b6274549c942779cf4659b73029336 : SUCCESS in 1h 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"521122b63e6d24646c79519a95d7e740c8b4a040","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-10 17:08:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3f2eb03fafecd5f417164896410931f3f47de70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 17:09:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"beb5b6b12074b9573a21e7ddd96ee24bafb6d5a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:28:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba29ff195ec4df1995e378096372aa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732d10e99eae42d4b5a7b2f4ed23421f : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd58ff15c8f0493ab1058917311f25f0 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9c3118e093a4ae7a70dd3ff4445885b : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43de5ef1bd034ec3965e34503e12803a : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73cedeae96674007a5ae0469faecacf1 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12c19e1482c74f0a954de3b7a0677fb7 : SUCCESS in 2m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/05aaa3de338e44328c3da268298be327 : SUCCESS in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8716ca7305f9407587b20b50e31daca4 : SUCCESS in 11m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1b07fbb97b414db987ea2062c962a415 : SUCCESS in 11m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4afe0f37abc441a6bc18a48ca427d112 : SUCCESS in 7m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1c4b63a28fe545e598fd77143ae646ea : SUCCESS in 6m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ec6df37bd4254c5c9b42c6394ec07b42 : SUCCESS in 3m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/011e2d3267964c7288f8b518c90c2cad : SUCCESS in 7m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/99db89915bc7419583170fbb1129e6fe : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e558ab6ab1af4939879e72865db1416e : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dd0f48103bad411a9507ad3a60629837 : SUCCESS in 9m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e557fc321a154f3c83ea346331c4a5d6 : SUCCESS in 10m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5f392b1c953d477a83e2ad7795bb6eb6 : SUCCESS in 1m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3485e3f0706941fb8ded61885a9c5240 : SUCCESS in 6m 28s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/073679cf526d468db4bd6aa884fc2d1a : SUCCESS in 18m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2bb6caa376a64b9084583ad5343ae013 : SUCCESS in 9m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ec17549ff9dd4892ae54772fe626068e : SUCCESS in 13m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6173218f21334eb6abe70832826e9351 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e14cdad8168b453598fc4dbeaf7a9dd3 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2583371262f4909becd259e8267a7e1 : SUCCESS in 21m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/aac5c65df44a480ead592e92e226b1bb : SUCCESS in 8m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8df7569830fd4a3ea2452c869bbb2c29 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a4cb61df21b40a8bcec7d5e1dccffd0 : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1aa19145382f44bba88748dd741f0bb8 : SUCCESS in 6m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9719b4954ca244e780176dd3cc3c963f : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f38cf06bc3e74b51829f259c9e9ee0ec : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b5569cff685499a9f22f2ea8dad6069 : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ecfde2a6179f5b07d1fe51f0ed7f6c254b6f6a7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:28:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3267a00402db856af1bb214bc8ca69d779118335","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14f7019856ec400aac77a4f7bb1504ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4011107e9d8420c9655bdea2c457d1c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa7b907d5c924c0d9a57f74222a27d0aafa04454","revisions":{"fa7b907d5c924c0d9a57f74222a27d0aafa04454":{"kind":"REWORK","_number":1,"created":"2025-12-08 12:40:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/970118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/970118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/970118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/970118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/970118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/970118/1"}}},"commit":{"parents":[{"commit":"f5707cc58f7152a82d93933a9863587af5f0ffa7","subject":"Merge \"update constraint for microversion_parse to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5707cc58f7152a82d93933a9863587af5f0ffa7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-08 12:30:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-08 12:30:45.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 8.2.0","message":"update constraint for openstack-release-test to new release 8.2.0\n\nmeta: version: 8.2.0\nmeta: team: Release Management\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I722626664f1f3d67be524a499ab766b220067d22\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I01df5e3effdf03cfb79951895cf650ae81ee0a85\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa7b907d5c924c0d9a57f74222a27d0aafa04454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa7b907d5c924c0d9a57f74222a27d0aafa04454"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
