)]}'
{"id":"openstack%2Frequirements~970200","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I913c2f6a2331cfdcffd6108a8bfec24b978d8e6a","project":"openstack/requirements","branch":"stable/2024.2","topic":"monasca-retirement-stable/2025.2-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-04-28 09:35:59.000000000","reason":"Change was submitted"}},"hashtags":["monasca-retirement","retire-monasca"],"change_id":"I913c2f6a2331cfdcffd6108a8bfec24b978d8e6a","subject":"Drop Monasca","status":"MERGED","created":"2025-12-08 22:38:52.000000000","updated":"2026-04-28 09:35:59.000000000","submitted":"2026-04-28 09:35:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970200-monasca-retirement-stable/2025.2-stable/2025.1-stable/2024.2","cherry_pick_of_change":970197,"cherry_pick_of_patch_set":1,"meta_rev_id":"b37d3532d1db6f00d6b60669b8c4e92f8d255d89","_number":970200,"virtual_id_number":970200,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 09:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-28 08:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-28 08:02:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 00:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-09 09:33:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-28 08:02:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"92abfc477df7669d62594ce3c7513740a2fd3c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-12-08 22:38:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"f649215600d84bbd3e69ade3c7afeb068d6ae9ca","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-12-08 22:39:25.000000000","message":"Hashtag added: monasca-retirement","accounts_in_message":[],"_revision_number":1},{"id":"29efecfc89a7916fbf8b90c0b13f30634f29b05a","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-12-08 22:39:32.000000000","message":"Hashtag added: retire-monasca","accounts_in_message":[],"_revision_number":1},{"id":"b593adba0115b3d10388852d52fb43fb631cc768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 00:26:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8809e0aac1e94e948da73fb40dbd265d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a95fbf4e1b49aa8218e435182e3bb2 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dc0728deff6412da142f1aefb630b9a : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99328a4f5ff54f45a66377f3d09bf1f2 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4171fed625e4277be137651eac24f5c : SUCCESS in 2m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ded20587346040938f9f29d13031c08f : SUCCESS in 1m 58s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/916917f9d94148f28f0ee977940cde2f : FAILURE in 2m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/132a44d821504dc7b08ac2948d8a4d6c : FAILURE in 31m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60a918ff51744b14b1e29e4808c5fa35 : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2683227dda943b24b7440cd76378707d9f475abd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-12-09 09:33:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65cb9efd3f00ba2dbbe1885c58b1e41be6a0533b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 11:42:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e61964e6014b8ea68fd7d9f4b61f43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134b5ab3ac124be19d54d5781708962b : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ecd64003f96492abdc63331016ea33a : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71a0d72a311f48d9a63907bc08cbcd29 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b704aac85b4cf0b1e3683fd01f2f6c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e2146745fe947079d88754565459acc : SUCCESS in 2m 56s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/a4dc931bf20846ffb6b9aba32f56ff90 : FAILURE in 4m 00s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8724edf590b04f44b2ca9a41351731a1 : SUCCESS in 46m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d176587a54246a898b43201592c9452 : SUCCESS in 2h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"52d00fe050d6409bfe98287547b7a2716e5f8ade","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-28 08:02:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68424e8ecfb0a144f6050f7a930acc50a72572f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 08:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce8e3b7f5ae66e2e72442bf2c2bd47a94bddbf4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 09:35:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/282d2e08d83547c5a4ff960bc36f7e85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5eb8edda60461cbe042d092661d51a : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51fabc66625e40d9966dfbc56cbb1718 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4820097fb5f14bf8ad366586a62bc69e : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02409f4e24e54a14aefd6eb748579894 : SUCCESS in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/411a6e489ed541b58ea61b39457d5c8d : SUCCESS in 57m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/348463d1b92f484d885311ea64b7c99d : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b37d3532d1db6f00d6b60669b8c4e92f8d255d89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 09:35:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22bc86bf2e01f02b443cba31ba867b9c57c70661","revisions":{"22bc86bf2e01f02b443cba31ba867b9c57c70661":{"kind":"REWORK","_number":1,"created":"2025-12-08 22:38:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/00/970200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/970200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/970200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/970200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/970200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/970200/1"}}},"commit":{"parents":[{"commit":"2c31269cffe9d6e5414822882d6158934df536ee","subject":"update constraint for ceilometer to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-12-08 22:34:24.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-12-08 22:38:52.000000000","tz":0},"subject":"Drop Monasca","message":"Drop Monasca\n\nThis project team and all the deliverables were\nretired from OpenStack\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/953671\nChange-Id: I913c2f6a2331cfdcffd6108a8bfec24b978d8e6a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 0bcf04932ce2ee8140070664e41dc71586627acb)\n(cherry picked from commit c8cd9e9c57d27b8a8dd86697922cb40cf74ebcd2)\n(cherry picked from commit a242107c709043dc915afff78a44b80698431320)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22bc86bf2e01f02b443cba31ba867b9c57c70661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22bc86bf2e01f02b443cba31ba867b9c57c70661"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
