)]}'
{"id":"openstack%2Frequirements~970354","triplet_id":"openstack%2Frequirements~master~I66a4f43c04e4235a260c2eb7da8929c2ea670766","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-12-11 12:38:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66a4f43c04e4235a260c2eb7da8929c2ea670766","subject":"update constraint for glance_store to new release 5.3.0","status":"MERGED","created":"2025-12-10 10:32:20.000000000","updated":"2025-12-11 12:39:41.000000000","submitted":"2025-12-11 12:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970354-new-release","meta_rev_id":"c70cedfa4816bc360bc85a7f234260cd6b3aa6c8","_number":970354,"virtual_id_number":970354,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-11 12:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-12-11 09:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-12-11 09:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-10 12:27:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 17:10:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-11 09:16:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"31fe18c4f943cdc588e9e9761abf486ac33906a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-10 10:32:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca6525c7751940fa611c91d4d0df0aa095aa87c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 12:27:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d232b06a986d4879b667bf7619dfd652\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47561ba1dbda4a91bf21297915a6ed63 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1966c6480db340f7accc6de303dacf5d : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0eacf64c7f76477893d2d04c0401e27a : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44c1fc736a0f4b31805e3ef265028adb : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abd2e21a9cfd4b19ac102e7f918a62b3 : SUCCESS in 1m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cf70b2760854b54a452d08dcf654643 : SUCCESS in 1m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b57dddb62a0542cbb6a83daa12be55ab : SUCCESS in 4m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8e6d16cbb083458680f71d66cf47d48d : SUCCESS in 6m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/50c2142069de44d9b44c78cb9d641da1 : SUCCESS in 10m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6d143195590f44c6b0e102f9225257ac : SUCCESS in 9m 35s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/22128b4fbd544a8a944a291a89f12b07 : SUCCESS in 4m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71a9164b3fdb4166b0d5797b383eff6e : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2ca496572670418bbd1f8538fd7b24e4 : SUCCESS in 5m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9d13ac9985f545e494f83d295b7f6d26 : SUCCESS in 7m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/46c9c3fb92fb40d694af7eb9b34ad3a2 : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/265197f5ae8d4f0e91c381341ca9059f : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1ffc56b8efa743c2aa9cec6f021ef22c : SUCCESS in 6m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ca60f1b2101b4d669e96c9ac53b7eda1 : SUCCESS in 18m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24096b034f944adcbe8d4696bcac93fc : SUCCESS in 6m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/085af0428cdc47dfbe90274f900a0698 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3ce19799a40449baa5800ced57c1a0fb : SUCCESS in 12m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e1ab803c12ae41ff88c29b16f767b29f : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/55752ec46ec047e7925465a5f6b574db : SUCCESS in 9m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5bc068feadcf4f39ab7c54db3a992c25 : SUCCESS in 4m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/421ca85ca72347acb9870509257ace56 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d7a7502333184be092349f5ea648df8c : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0c3befcf1af24fcaa8d3f193ad6718ce : SUCCESS in 27m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4a4d3caa19e74ebfad4efc062f57fdfb : SUCCESS in 4m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/dc3ac1a32dac4e52a10624a2e3c681b6 : SUCCESS in 10m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bd738c46420f46c0855342013117c08c : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1996a56b38c44a2b3c0a12d5401a802 : SUCCESS in 1m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13145ccbfdcf4edd9ec559966252649e : SUCCESS in 22m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/15b74e5fa827401fb1a5a3f68c4b6e2c : SUCCESS in 7m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5bc0a853c0734c3eb8cdc4b86d4374c9 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6efa72aca59a4b6499857d0da5e8856e : SUCCESS in 5m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8cac9c67340e46f28bf7606698060606 : SUCCESS in 7m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ccf0f366ad124b929de5a1e0150d889d : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97e6d1b6fbb141bda52be7cc358b84a9 : SUCCESS in 35m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17fec2302519451e85fb8ed6a686aaea : SUCCESS in 1h 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1533dac97ae63acef882fac90d5bfae61d0f1086","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-10 17:10:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99e1781ad2850679461d98e5570f54a737b98202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-11 09:14:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ecc7b5ee98c9ace1e936cc822b406fd957c8e3d9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-11 09:16:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ce37406c728fe44860862a8a8e44e5a2ef66c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 11:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc53f68e978450db3894c06f35640ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d86d88998446fbaf44e8f42d23f3ba : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc42c5026fe4585833f371f7c113e0f : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e2e5cee55174ba7950809f87180ca59 : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24f70cd828754b68a88ebafedf732d04 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dd6d4c3186f4950bafd8640e835fda9 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1a783ea122a4b26839e6ea01a3f2c2c : SUCCESS in 4m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d82997f7b2a24d8eb2f73ccdb035f914 : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6d0de5db58084f229a8d4862e9c05a9f : SUCCESS in 6m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/52d18e089ab1451abc96a250f13268c7 : SUCCESS in 12m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/286184453d2f4c398c088cd406f3bd33 : SUCCESS in 10m 05s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fde9c2f0525343c69e63a3044f3f9996 : SUCCESS in 4m 36s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fda53cb61ec44704b5d6abbad4bdd0d6 : SUCCESS in 5m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0d653a61dd2d4797ba3eb2775fe790fd : SUCCESS in 4m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c2b813bed6b04000aaec1b451d67fdb2 : SUCCESS in 10m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5627332a85e04a868bf15af76578cfe7 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/723e475f2e68431cbb0a19f9937450f6 : SUCCESS in 7m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3a54844f2d14479b8ab7290719b4bc6d : SUCCESS in 6m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9003c918647f484e8c4191b158a4d1c6 : SUCCESS in 14m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e60cf53483144f0097caa4cfe7fa06ac : SUCCESS in 13m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ff3e5fd283fd46adb52d7e5f9f4e1bfa : SUCCESS in 7m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c442d970d9c248b683e3853f658ff760 : SUCCESS in 11m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c0d6afab58f54394b1fed248ec8a1626 : SUCCESS in 2m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a59a286aa1c3477fb32f83e6a17ab4ad : SUCCESS in 10m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f204d58224f64d81aae002634f6260ff : SUCCESS in 2m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1db57173f8f74a48b345653081438f2b : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e1d39550c7944bc7ae7040ad60d96a30 : SUCCESS in 9m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/824278ad19b34721a470fa5a40b49698 : SUCCESS in 28m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6fe62e67368046839a8eb4de3254fccd : SUCCESS in 7m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ed8dc8380f9949659fab18f4b8e12256 : SUCCESS in 11m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3111a29f5a04426e99c0614c1a792486 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a02b1d1208c467f9c514e9a41fa0b49 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60dbb3c89a844cfa8d54d515e2f3951a : SUCCESS in 31m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0e74964fe6614d1e9b9c14afd3e2c1b4 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bd9687f239674f3f8db90bd133e2762c : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c61d68282744d9cbd64fae08435baa4 : SUCCESS in 6m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c217b1d0024a4d4885e691493502edac : SUCCESS in 19m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a1f789e3453749659d819c4db7171291 : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d90814019a0433fa53893526a19768b : SUCCESS in 1h 01m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10a00aaaf8ff42e2b9d874bb991f91ee : SUCCESS in 1h 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2bf3c9ec10f9a28446d5271ff099c0cc124e06e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 11:07:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea17be54ac47eec51417ec60ff3daee7cccdfb6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:38:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f417983de434958984222fb02a8edcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7e673e2bed4842a5a601c1f64057f3 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c135fd3a61456aa878628d67a8a12a : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a308b2a0a46a4bbdb1f502e50743fdaf : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52e3a059671442af9bd0905c89ade19e : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8422921fdb024d32b8a5f8ea25b8f724 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6abea5481f024b25b6e9865fac62e482 : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0a21a720fe740859742d353f2a02bc9 : SUCCESS in 7m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4044d61973b14d5b811f5f733a7f5e95 : SUCCESS in 6m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/53e29a646fae4e2c86b975a55dd72c66 : SUCCESS in 3m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b5944039bff549e7811d79ebecda274c : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/13afa7c4d1684fcaa8320b0f422d162f : SUCCESS in 6m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/67758d178a34481c8cd0b898e89a41a0 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/68e9a5cc2f224d15b0ede18abe6a36c8 : SUCCESS in 8m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/143ebb900aeb439ebbc3acdbec85bc72 : SUCCESS in 17m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8128b69c54c44653a76e790511294040 : SUCCESS in 6m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/667b6f89b4ad472b8ab606107d938276 : SUCCESS in 8m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2ffb3d5889754e2389e5a9a1983d60cd : SUCCESS in 11m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3248182817cb483f9f46a0f75f9cf73f : SUCCESS in 2m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d72f60e45b874bd9a80732b310ae8bdd : SUCCESS in 8m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4c7471e4845c44e0a2780d330a6bd0b2 : SUCCESS in 25m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91c35f8fadfd477abd20aab41d3c46de : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1199a91db1af4af5bdf18df294fb7998 : SUCCESS in 14m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/78b7413007a640ff90d271c0d5b4186b : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d75cc712e3d242e6b9d386cbdec955b7 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a15bab5485094950809b1aa5fe3133c4 : SUCCESS in 21m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1a2215a2ec3d444abab5830fb9a1539b : SUCCESS in 9m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b8681f0887ed4dab81426cf48581484e : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ef895c9ce444c148b1edd8bab76678f : SUCCESS in 5m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/353e4c7535c74aae80219217952c4a95 : SUCCESS in 13m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79a7f0fd2ac94a0e9bca96c2707acf97 : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37f6fbe3693d4175b773b91ce49c5701 : SUCCESS in 56m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b066000ceb54847b24c55f9256244af : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7fe23031052cdbaed738e32ea1567a38b0f662b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:38:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c70cedfa4816bc360bc85a7f234260cd6b3aa6c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:39:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e31b1f5cedd4cdca1b151c19854657f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d194a9192f4c7b87e47c697b5585fc : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97aa40c2ce36dc032c2580060b50b335660e3a7f","revisions":{"f8f0f96be37758a1fe2ff03af6ad70ae0b7d6f25":{"kind":"REWORK","_number":1,"created":"2025-12-10 10:32:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/970354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/970354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/970354/1"}}},"commit":{"parents":[{"commit":"e0433b5aeec97367454bb4a110a85800c1e65958","subject":"Merge \"Remove python-gnupg from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0433b5aeec97367454bb4a110a85800c1e65958"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-10 10:22:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-10 10:22:19.000000000","tz":0},"subject":"update constraint for glance_store to new release 5.3.0","message":"update constraint for glance_store to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ida20ed7c9505a688bf26a5c1192082fbdcdfd3b2\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Bence Romsics \u003cbence.romsics@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I66a4f43c04e4235a260c2eb7da8929c2ea670766\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8f0f96be37758a1fe2ff03af6ad70ae0b7d6f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8f0f96be37758a1fe2ff03af6ad70ae0b7d6f25"}]},"branch":"refs/heads/master"},"97aa40c2ce36dc032c2580060b50b335660e3a7f":{"kind":"REWORK","_number":2,"created":"2025-12-11 09:14:45.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/970354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/970354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/970354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/970354/2"}}},"commit":{"parents":[{"commit":"3e0d656ea82b0c25ce8cc75de4c6919affad67ba","subject":"Merge \"update constraint for openstack-release-test to new release 8.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e0d656ea82b0c25ce8cc75de4c6919affad67ba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-10 10:22:19.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-11 09:14:39.000000000","tz":60},"subject":"update constraint for glance_store to new release 5.3.0","message":"update constraint for glance_store to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ida20ed7c9505a688bf26a5c1192082fbdcdfd3b2\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Bence Romsics \u003cbence.romsics@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I66a4f43c04e4235a260c2eb7da8929c2ea670766\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97aa40c2ce36dc032c2580060b50b335660e3a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97aa40c2ce36dc032c2580060b50b335660e3a7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
