)]}'
{"id":"openstack%2Frequirements~971430","triplet_id":"openstack%2Frequirements~master~Ifed9af5b3e89e129c3919416fcc748cd28447941","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-01-21 14:45:28.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-01-21 11:31:59.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Ifed9af5b3e89e129c3919416fcc748cd28447941","subject":"update constraint for cliff to new release 4.13.1","status":"MERGED","created":"2025-12-18 14:29:05.000000000","updated":"2026-01-21 14:46:22.000000000","submitted":"2026-01-21 14:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971430-new-release","meta_rev_id":"73d522212e44d4381b8c52c5673eeed3eaf17e33","_number":971430,"virtual_id_number":971430,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-01-21 11:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-01-21 11:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-18 16:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-18 16:48:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-16 12:14:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e6969b5df4097e11c130ee545694219a105ebbb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-18 14:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e4f532887cbd89882ce7e5ee5e8f093a13ec380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 16:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650e4b0627ea47708dadfb309d80e1d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4a569601e9429c82c8c819ba7c60a3 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb2f08545d8a41c79fca0e83539e9f72 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/171f679f88754c5abdf9bb1b1d3b5d37 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcef0ec65d914998b050c5f712230aed : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81fcffe0522941bd967898d6ec7345af : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/950beec6fd7d447188a6d9f60e5faf6b : SUCCESS in 2m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c4bc736dc204999a3888afc67962df3 : SUCCESS in 4m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f7bc385f1e5146f2a18e25b8f052ae3a : SUCCESS in 6m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dc7e71064299424a89426f743f5d8861 : SUCCESS in 6m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2ba537fb32254feeb55ffbce7b62bd03 : SUCCESS in 10m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a210ade442fd41b189817110571d62f8 : SUCCESS in 3m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/96d49b889e8e42ce9b2dc1b34cee787a : SUCCESS in 3m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4d30de8ee6144be69e3b7019588d3e2f : SUCCESS in 4m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7d4400bb89404b7c93f148583da8e7d5 : SUCCESS in 10m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2cc140eb85674b2ab80c4b50cb65addc : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1518303b4b7e44fa933cf1a46f36870f : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b6e9cd7326542c283b757b69bdf5959 : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9e0e52fbf3754d77ab05896fc61a7ec8 : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f326f742fec04c7bb2b3df11fc88c8a5 : SUCCESS in 8m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b745f8f448084ca1b8c6c423debba862 : SUCCESS in 10m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0315c724b4c04a97aab85203b3423b4b : SUCCESS in 9m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a697ebef5c64421eadfd4551dbb4b75e : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f0aa27846dcb46d78883c64d8e731f31 : SUCCESS in 6m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d57b59a36a9b4f34819f2bb419f80671 : SUCCESS in 3m 58s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c04386e2aa50482c81cf0fe9699b4288 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/55777ff6455e4f458e38901d52c70816 : SUCCESS in 12m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/76472e1eb0f947379e4c38a6b6a40e9b : SUCCESS in 28m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2234106e7a104f69bec375d022362e00 : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ea45eff5d8cb4c21a2d28ce287431a69 : SUCCESS in 11m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1bda20757f4e4c6ab88e9b013aee4354 : SUCCESS in 3m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99c822bf5d7e447cabcb75ff2c452877 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e4821878f1e4bd692d89271fb7952af : SUCCESS in 27m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a9d4dd4b38f84a38b1e96affe4cfcf26 : SUCCESS in 7m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/37e9062c1ba949009c2d71e03727fe49 : SUCCESS in 3m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e99167413ec4b21b0c64d993ff4269b : SUCCESS in 5m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c2a3f731c5c24e12bb516d1171436e44 : SUCCESS in 17m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9bb2756743154fbd96af408b5954b380 : SUCCESS in 6m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68f459ca96e9489cbcf6b0ba071e99fe : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b237fe80b2734d70881ee06da39d60c0 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":1},{"id":"060c344b47198fed0617fe2fd6ddf13f9f614a69","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-18 16:48:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c6bbcaa5519003c740aeafd8c18aad534867021","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:14:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c7ff63ef752e57158f150bca8c100f0008c80d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:22:22.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 971430,1","accounts_in_message":[],"_revision_number":1},{"id":"11e5e10339bd71ad249f11840c93d5a2dd49ef8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 15:45:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e15f30f3606de1d8e7e6ca06f3666ab34631c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:22:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64e880d48874b78ad9e6b63af6c74e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cb4f51acbd4fc984b66ebc1f3322b3 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fbceebddb754ddb8976f3b008fb9ff1 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0938a2840324eacba3ec9f01a91dca2 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd4fd25e25d4489fb81983e06ddd1227 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c1b61f7060d45b6aa4d3bd3bf696846 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98b3a5df7f1f442f884f5abd0d6cf704 : SUCCESS in 3m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ed3520c0ae042a5a7f92b2200c3f6c0 : SUCCESS in 7m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0598193284aa482894182ec548fe59ce : SUCCESS in 6m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db12256cf83e46fbaf516cb5e79840bb : SUCCESS in 6m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5483d00a464b47c18314ad11d95de148 : SUCCESS in 9m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d4efaa30e641404bbdfccf67c5f4601d : SUCCESS in 6m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/241f192026554104a84a7de264efa25a : SUCCESS in 3m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9c3267d4a46543eebc1926184bd31592 : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c01afe9012144993aeeb758a03834964 : SUCCESS in 8m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9ae953dd266f4f119b457c33d9a740de : SUCCESS in 6m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0e441ce71db64cb8b856f1cef1ff5fd1 : SUCCESS in 3m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c798d86b910e4ec7b8a7f3b1aaf60a9f : SUCCESS in 7m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e54c97df82ce40d585980b486bb11115 : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd119b1486e041cfa855765c40e2c241 : SUCCESS in 8m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d2b7d3431a04ec0914c9561c0f6253d : SUCCESS in 6m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b70016e0764648ac843deed8f21dca3b : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2e49b751667b41be8bde32126c735357 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a243e23131dc4c6c84ad058c2375f710 : SUCCESS in 13m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/419aab6621bc48acb49beede6935be08 : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/26d3a6833835451db053ce149f8ceefa : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/874e641974c34cd69dea4291ca604afe : SUCCESS in 10m 33s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/98ce039dc9074df5aabe0e3b4ac006c6 : SUCCESS in 32m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/41e30112126946e1ae7cbbf7532c2e60 : SUCCESS in 8m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89ef6b7ee07a42d69ba03f588d19ccf1 : SUCCESS in 15m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e3451d5a8c6140199059dc03fe72ce72 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/081ce94d1f494bcb9d172125dcf76abf : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/728b4bc6095442de8a5aad3730dba2e2 : SUCCESS in 27m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/434d605cae7440b4a39ea0d1456ba730 : SUCCESS in 6m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8909597f70364603971e90c416d3f068 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e62a8f816a3a4374b12f83bc8bcbf6b4 : SUCCESS in 5m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3a1b7f9f77cf4bf48fef9d013ad0b6e6 : SUCCESS in 7m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e23d2c546c4b4eada7b375c86c9437d9 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4657efa23a7b4a6194be01ace0c33336 : SUCCESS in 1h 02m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fb76d77603a45aca292978d3f570c3a : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a32ae608ae64470ce534cea9202501ce9853deb4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 11:31:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bffdd8f1190f5cf33049b8a3b115e6d94c619fe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:32:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e0c7f21dc66fa39a4abd20c560e208a426378229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:45:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9224ba02ad5440296eb2247b717d377\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014241fe40fb43ab8895d73f5b8f73b8 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4537ac2e6424a8994b6a3eae9bd5de1 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cd24bfc8067402e805f30946c577c99 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aafd6057675f4ffeb55a4cb8e1900400 : SUCCESS in 2m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5073e5a47858489b846d4f9a0742805b : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/632e55525e754cb9930f5ca03e7314a5 : SUCCESS in 1m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/63b27999e9754ae495cb8b6f4de3b80e : SUCCESS in 5m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b89bccc1d1c84fd3a6402e3de16a4502 : SUCCESS in 5m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a43c1d57edf4bfe93bfed9fdb840f2a : SUCCESS in 7m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/509b8b382c044094966c445cea4cbf3f : SUCCESS in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/38738d5326f5431b99047fcbe9c46bd6 : SUCCESS in 5m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/af47e885842f4c139c28168bb8e9ea37 : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e2c0f5b61ce449c39b762405876e8e92 : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/66f4d7ea83ff45f890c321d68f58e98e : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f514c1aa13ea4613a711b93053fd2566 : SUCCESS in 6m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/17c33b0adcb84871995fc78cd1406787 : SUCCESS in 7m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/eedb3f7a309043f78adfb72900a675b1 : SUCCESS in 10m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3dc2dc87040a46dca22e145d0338cd06 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/972fb424cf094507bce52932b30c4381 : SUCCESS in 7m 00s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7ff8aca7637c4dbab3fffae466fe5741 : SUCCESS in 28m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/444e317ae3784fefa77ffcd5243fa4c1 : SUCCESS in 4m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/99444eb77bbe4cbba77ad11e68cb44d2 : SUCCESS in 10m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f45e5dcc80aa4b33b558b81551a677ea : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32b9df5bb6054a5b87b64b2c93aa1327 : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/205ae51107f242fe98299c787dba90b5 : SUCCESS in 27m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/855893461779453f897024f6a0656552 : SUCCESS in 5m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bafadb9eae6f443eaffbb8fa02977d73 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aebec215c274fa1a64a22fde33b1ad1 : SUCCESS in 4m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c56e4787eef7433e8e88c334f7c32502 : SUCCESS in 18m 07s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/de81a3e92ac54dc596aa460db469b319 : SUCCESS in 6m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb4fbc6286c9439a899ed636bfe8c45a : SUCCESS in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a44e6b2d7f974f0fb419f82e311616af : SUCCESS in 1h 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e23b838647a2957bf9d6297d9ccf1339936c41b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:45:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"73d522212e44d4381b8c52c5673eeed3eaf17e33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:46:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd4f524be354516b61ade8c4db2cf21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa5f806905444cfb4a4d8a8f0c0156c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c04fe8ec12e0bec5d71b2607beab4db6ad832375","revisions":{"ca72cd5a6185a74cc031d8f647632e13212d1065":{"kind":"REWORK","_number":1,"created":"2025-12-18 14:29:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/971430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/971430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/971430/1"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-18 14:19:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-18 14:19:04.000000000","tz":0},"subject":"update constraint for cliff to new release 4.13.1","message":"update constraint for cliff to new release 4.13.1\n\nmeta: version: 4.13.1\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I21c0f8c3a633277df38188989697d0fa7bfb6f99\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ifed9af5b3e89e129c3919416fcc748cd28447941\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca72cd5a6185a74cc031d8f647632e13212d1065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca72cd5a6185a74cc031d8f647632e13212d1065"}]},"branch":"refs/heads/master"},"c04fe8ec12e0bec5d71b2607beab4db6ad832375":{"kind":"REWORK","_number":2,"created":"2026-01-16 15:45:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/30/971430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/971430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/971430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/971430/2"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-18 14:19:04.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-01-16 15:45:07.000000000","tz":60},"subject":"update constraint for cliff to new release 4.13.1","message":"update constraint for cliff to new release 4.13.1\n\nmeta: version: 4.13.1\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I21c0f8c3a633277df38188989697d0fa7bfb6f99\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ifed9af5b3e89e129c3919416fcc748cd28447941\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c04fe8ec12e0bec5d71b2607beab4db6ad832375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c04fe8ec12e0bec5d71b2607beab4db6ad832375"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
