)]}'
{"id":"openstack%2Frequirements~971544","triplet_id":"openstack%2Frequirements~master~I58311ce6ab939168b4218094fedc131074476a00","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I58311ce6ab939168b4218094fedc131074476a00","subject":"update constraint for oslo.log to new release 8.0.0","status":"MERGED","created":"2025-12-19 14:18:53.000000000","updated":"2026-01-20 20:32:50.000000000","submitted":"2026-01-20 20:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971544-new-release","meta_rev_id":"fd79f5962432584c847198e33491184a8133a14d","_number":971544,"virtual_id_number":971544,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 20:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:13:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-19 15:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 12:13:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"5de95a6c79cad26316d2495c8f5d8de8b25a9b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-19 14:18:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08cb2a0c165116f6299a0cb9c933dbd02e902b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 15:16:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c602cdd4c1af4377b869833a4d8ad75e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc3416222de47e289d2f34b14b7fc9d : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d30d2884786a420489edc9743a9b3b6b : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/920431c4601d4f77966598290a030f66 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/843d905d86944db59df616ba7034fa71 : SUCCESS in 7m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1007e16ef8844156bb1ae86481a24032 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37c1fb49fec14b86bd8482c3abb47b64 : SUCCESS in 4m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bc021abb71cc4150812677ec1013b197 : SUCCESS in 3m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c7ceba087054d9cb146a6baf12ff166 : SUCCESS in 3m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7854fd63ee8e4da5ad92a82309fb8317 : SUCCESS in 3m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2466bf7233af431daf6b9f92157ac0d5 : SUCCESS in 5m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/83d8530cc18c4e24b7fd6d5898516e97 : SUCCESS in 4m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f11ef9c3a63b49e4bc3a8e9a281dc302 : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/14c0cef1ef834c369e2c4e825e39b2f5 : SUCCESS in 4m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/effd1989cfe14919aa708dcefbd9a4da : SUCCESS in 10m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c452a9e664cc4bc9ba5fd2c64af7c51b : SUCCESS in 6m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/68154ef563b846549105e0786677908e : SUCCESS in 6m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7cbd827502fc459cb25ab1ba08bc4793 : SUCCESS in 7m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e45f98414c0f48c088e5c8bb013c03fc : SUCCESS in 16m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54ae68c0d9594a13b2c7bd05c82baef3 : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2b6ee41e52ae46ffa8a416fe2951e575 : SUCCESS in 7m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8d733177096c477bb2eb5b3d63ad6401 : SUCCESS in 12m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/757b170d3fc04fd4a2d3900b11139e81 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/467a8d54e0624f6991330894c162ec9d : SUCCESS in 9m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/dc416f779ec1439db73740a2fb2a6445 : SUCCESS in 3m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/30f1d02f6d2e47399ceaa4d4c72741e7 : SUCCESS in 6m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/686ceaf6b31e424881270a7a4cf8af82 : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8a62ab13c33f4308bae282cefed32b9b : SUCCESS in 25m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8eec8bacd36d457b8860dc02d2706c42 : SUCCESS in 7m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d3064988429745dfb416abe20514dc02 : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0d6f6421a9da4534afa4da91fb281605 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/10cd5f021783442d8891d23f76e26a15 : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22caa278d8c149c5bc7fb1eeb23001e9 : SUCCESS in 32m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a98437366f0a432390d27c9d41ddffb3 : SUCCESS in 5m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/046dda8a0aa847648c09afba016a37a8 : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7c218a0260742e1ad526d1e31dbebb2 : SUCCESS in 5m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4aeb5ddd8a8840a585c7011e9208977b : SUCCESS in 17m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d776fc0e80b945358114064e9a0186bc : SUCCESS in 7m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87e795e2016c4c24bfc5ef8e3e1eb167 : SUCCESS in 52m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a3638eb34794ce5a483935cd84149bc : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7e29e822af1145c0e0129ce054356a442fb5f307","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:13:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"837ef88537d52d6bc2c6d9bea3861992026ffd72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:14:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86bda096da1748c4309a74d9a8d4ab0bac253312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:22:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b89c44f352594b47b33ef323add02063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e186d9b9e3b34ffba305965e4d77e846 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5abe1383cd584125b2ee716678bfdf6a : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99a2b08023c24aa3b2bfbcc146f92b8d : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/863f4ed3555c406983763d917a76227e : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc26803e13b64262934a2f844a63c7dc : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9297d5aff88a4ec7aca42a97844d6f8f : SUCCESS in 3m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/72f62834cabd4b38b39e0c26f7d34295 : SUCCESS in 5m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/633f4cd51f794d3c96340e95dc7d2620 : SUCCESS in 10m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8527b3ef5c6a47a2be99f2a6fcd97965 : SUCCESS in 13m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f63685d9257e4ca5977481c6832d09c8 : SUCCESS in 9m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4ca64f827e9248d09b631f2f2071e869 : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b55dfe459f994070acf9a4566e346b7b : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/509feb4b41a14e8bab27ae723b82983f : SUCCESS in 5m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/570b21e3050f4616974acdb99b0d2a46 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3475f1903a4c4932ac48ae3821096b71 : SUCCESS in 8m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5bb94254c6f645d68738af04da0e0d7b : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2a3a8403cd4f48f5ad8954116c57dd83 : SUCCESS in 12m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a0b80698521e450eb68197d7631c54a2 : SUCCESS in 2m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5ad794ec99174d9e9d873dab7aa6c253 : SUCCESS in 10m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d73d869f727047adb8dcaaa877ab55a4 : SUCCESS in 29m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e9aebbb0d5464c09a01697885f0b2ae9 : SUCCESS in 8m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cf1590ce157b423083c41eaa4e700ccf : SUCCESS in 12m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a4dd884ddbb84f2eadb904c3c8fc05d9 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61502d277c5b4e7399091e489eab22ab : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5a1bcba0ca447c39569ba527cc8b5d6 : SUCCESS in 29m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ac25f48a12aa4b1f995478a2158e67f4 : SUCCESS in 8m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/31a701b6e2aa4707a1f9191a460fae3c : FAILURE in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98ab2eb0b0294c9b9188ba3c56658d99 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d2b4abd768014e8f8ed8d13c8bc6d81a : SUCCESS in 18m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/59d356cd67694535bf099f5bed85648d : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e64cf02be33943e4af9dbec3fbc0dd42 : SUCCESS in 1h 06m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b234f97ff1dc448ea83be21973650a76 : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ed75a11937dbd6f5a8f16140269d5e32e4a15c06","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 15:42:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"820cc2c30997fed771457dfc0447df053837c342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357bfb90ca2e4e70af86fe3190ee4c2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0190434a2c9f4ef5890ac28b7cef1fa1 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93de383fe0549b3a93a94cc1b2098ec : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ec090e11b1d4208b8bb54e470aedeae : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/498aa5afd357427cab91a91065bcadb7 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/399b304868da439cb1689e3b1567ab03 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/26627fb5b0e5432883877dd1070088c5 : SUCCESS in 3m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/08e9c6ac41db4aa2ab9a2b653e5883a3 : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45d6c048a7ef411f912750ac9fc74f77 : SUCCESS in 11m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8203ec5e117e45048bcb13a7173f47a2 : SUCCESS in 6m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f8628abfa75e49dda9a5f6e25812c9a7 : SUCCESS in 9m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/53188e8b511b4eebb74219842bac8d22 : SUCCESS in 3m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3cd002b54671494390a361a2b2432f7d : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3092333972844867b3564a930fde8279 : SUCCESS in 3m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0f3b6ce25d554271b4732bcd6292b75e : SUCCESS in 11m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/521a1181879a4545badb63bd7d9c567d : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90425646c0554838b13e9800cef111cb : SUCCESS in 5m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b84d6956d9e4fd9a3cd1c26aa336a78 : SUCCESS in 9m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4e27f5fbc70647f382fe7585d6bbf859 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73fed1056e63478cac95fc1ec55c9c39 : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/86e0d7f7233d4c02a5f4e13efdc962e8 : SUCCESS in 9m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/db65f95606c44c30aa5ab07e1dfa30b2 : SUCCESS in 11m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd6e99d8f9064b8fa6bb4b1ffed0ed88 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5a4900aa08964c40803de3a864357174 : SUCCESS in 6m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bca367adf7d3456798e2f028387c46a8 : SUCCESS in 3m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/79b63af8d43e4fedab963dd133e95f20 : SUCCESS in 3m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5f200d92c8ce42c2ba4ea6c3a63c79dc : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1abf7dc49c934718a72d2090ecd419bc : SUCCESS in 24m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d55f084a5444c52914a8b715761de2e : SUCCESS in 7m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d22c281bdd3f4fc7bb4cb890e9d7977d : SUCCESS in 11m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/801c2ff64249421197f55447859782a6 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f3ab36d9c0e4691ba583ef9a36a825b : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a083947a891049a387ccdcdefe41c5ed : SUCCESS in 26m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/18d551bc41cd4ece8215d9ef81032571 : SUCCESS in 4m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/73c8ab73f48245a7a7003aeeef1f4b76 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39e2c71f216d466f9bbfd5e1e75ef435 : SUCCESS in 6m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8f2a8a74f92a4309a1e9248345af084f : SUCCESS in 13m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/93328a9197764ed7aef0c6e3ac35403a : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37371724149b49d7bdfe64b97f0ba9b4 : SUCCESS in 33m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d4b3679583e475db68f3fad18fa4c28 : SUCCESS in 1h 43m 33s","accounts_in_message":[],"_revision_number":1},{"id":"58bfef2c32248a7144c72029c6596649de0256ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 17:35:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0c783908b07e0fb031b425f4b34787f332384ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:44:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/907115653be24bc9871e4cea9d50f86f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2505e6ce991e449fb0fc8179f9507d3c : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4997eecf3fef4c1fa3333b4a94b0efb3 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b57c789eec794861bf3627701fe0ff60 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4b15c8f4e7a4d81bbdcb387dfe936f6 : RETRY_LIMIT in 1m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d4b141f406c402db4334accaeffe20c : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed55ed57ff2042f9be3220c00a836177 : SUCCESS in 3m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e241cd0a575843828922719f4d2aeaf8 : SUCCESS in 4m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fdb2575c6e394d0db837c3fe64fc7041 : SUCCESS in 6m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a6e3cd2c0e3645d295b0f96e5306e803 : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4c0e11d20a194ec0996e6aced7a663d8 : SUCCESS in 8m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d607f3f2cf1049698961db98867ea744 : SUCCESS in 5m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b643f5d299ce45a1b3718e7346b8b261 : SUCCESS in 5m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/46b7e369fca148b8b0bee5d210152ae6 : SUCCESS in 5m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f16ae8fc6d544064912c6bce02416d56 : SUCCESS in 17m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/668546cccd0346359249cb95131f7b88 : SUCCESS in 4m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/97459d50ef9b45fbac018a5d067647db : SUCCESS in 10m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e4d90e1de1cb40dd9562ccd6c41fa179 : SUCCESS in 7m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/609bc2f81ec74ee0aa6faa0d388f4fd9 : SUCCESS in 2m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1b7d02f12c1b430b808ab54441c2c8d0 : SUCCESS in 10m 20s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d28ad727cafd4a989eb5f3e30c929273 : SUCCESS in 24m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a63382affc014f37b9cc5cf4a02bb58a : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f0df4ebde90f4de9ae8d2ceb4ef0e1ec : SUCCESS in 13m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/621fcab1814b4286b268730e9757a273 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46a4fa4c8fbb41f98dde6cedb4807e47 : SUCCESS in 2m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a10e8b4cba0940f6bcf52161f03a0121 : SUCCESS in 26m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22f44d3099cf4d5ebc446ab4817a340f : SUCCESS in 6m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0ac60e26a5fe4790b27f0828833e111d : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d544712fba28497c85894bd72ee5fedd : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bf7e967e9ab14499afaf2337a36aef94 : SUCCESS in 16m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/84063a8bab064f068af5c235b7c07dac : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1534978ce5149a1992ca3e6f14f8c50 : SUCCESS in 49m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/422bba1248b8433e9d6e5621116352be : SUCCESS in 1h 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"dc470705994ebd232007e92e7417025222c0cf1a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-20 16:36:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b34f5f3a10c0eb5b57df938ef75996b4686f608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93db696e0f4441ff97fd8deca0f62436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dae4ff4498d443cb4db0384f23c5acd : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a0de13929d4eed8711bfe551f39b0c : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cd31086549b4cf6a4007801fb2e1a1a : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f71a1869c3d4da0b78d9e581cf6b2c8 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebdee9194d6e40ac8a853d79bdf6a287 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39602a4c75504c7fb12498533af4b34c : SUCCESS in 2m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2b0186354c3947b5be0702afe183fc3e : SUCCESS in 2m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a80a3bf63bc946fb96d9b2a10e94aa27 : SUCCESS in 6m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e15d9c390e01448f8d8050f313b4f727 : SUCCESS in 8m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/794ac1cff1a1449f85abd14322940489 : SUCCESS in 11m 09s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/743efa6d19c24c3fbcd313f84545dde8 : SUCCESS in 5m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dcbaefea4b87428e987021c3a3f04166 : SUCCESS in 4m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cbb72fe0b5fd4a34938f63cc1698aa8e : SUCCESS in 4m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a3cc0cdb003b41b982a05dfaf35d0e6e : SUCCESS in 10m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/22b3880136a64226b214d91a1262e969 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d5114978f807417eb42b17e3e4e3a968 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d29bd578a5e549c8bfa8b7e5733a293a : SUCCESS in 6m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/89ce8e1585de4182a0496354534b4a86 : SUCCESS in 18m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b940f31f2ae34d16872704ee6d2e2464 : SUCCESS in 6m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/815f61ab17ec49c2b33cb9b04c893f54 : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2d5eb8cfc4e749b8a7b2703efea31eef : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/77ea8e90768947689f9e0f85f15b62ba : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a28783463b0f4bc4a702d25a215c9dca : SUCCESS in 10m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ce237d8a71124d83be80ad08778f598e : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2f2aacef979646a9a1ed79fa78b18494 : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9c208e6120ab4306a74e1ec6b3831c7e : SUCCESS in 11m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/439f0948d4534b358d134b07a7438627 : SUCCESS in 32m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a940c3e6fd704d3a87479e966a3684ee : SUCCESS in 7m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/15afae5bdf2549e792e65a136e5c1f1d : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9e5397e233e64392beb83170db09bf5c : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be3dccce5bd54d648d28ea3185960a64 : SUCCESS in 2m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2a9ae89b1a3423eb481c215cdf07cb7 : SUCCESS in 31m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/83bfa11ff8c0464e9816ae73e74def52 : SUCCESS in 6m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7df19735933344d8a0dd4814c97629a9 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0baa6c4e27e845179084fc5ec5b70111 : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5366f1c9ed5d4e13b1e883326deaa1c8 : SUCCESS in 18m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dd39d7bf4ecd481d874eeff59104ad5f : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63e1f29367b045d6a1b12b21ae34d86b : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a480b1ee914a461f929c0f4e548a47b7 : SUCCESS in 1h 42m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0b11b4143b1bfc49db9ff484b6e903b6b3b565fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:24:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ece6c14b63c2bbe2942e2b330f4452971996cd27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:31:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78c6fb5aed2a4d528571cf309caf51a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ab4293df494911a4669688bc6857cc : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74fa2b104a2f42de8c88392afd0a7c84 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e52eb9bcb6b4630bf39ef33864ded2b : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a35739f9486040c78ed69c5b4589497b : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8f27df4ba414c79b5518124d50e53ce : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92c289c47ce34f7fbb045dfbe07c3fdb : SUCCESS in 4m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ad76a9ee18244fa59b34d684b49edb70 : SUCCESS in 4m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e3db3730be8f48b38f5c7e995180bfcb : SUCCESS in 7m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b501770d4f744a889c2db5cb66868b5b : SUCCESS in 12m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b1c656f6341f430cbae01466d9fe26b4 : SUCCESS in 11m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/67e356e04a884a5eb4e32adf5022f19f : SUCCESS in 3m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e8f986083d1e4509aeccec31ccbf5166 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/74aa31d938974d5facda508fab7ea43a : SUCCESS in 7m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7d3e3225ef8e4eabb20fd1abdde33409 : SUCCESS in 18m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff300f8d43a8448898b5a3ebbced5ef5 : SUCCESS in 6m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e6ac434eab6948518aba3c6c0863dd3d : SUCCESS in 9m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2a463756e5f74664852223ee2e78bfee : SUCCESS in 11m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a9785fd5d2a843579ecba6cbde5e31c3 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3c3c598aabb2437fb75b6061114ffb11 : SUCCESS in 8m 29s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fa9ec1b3f8434e4d83f0d869671351c5 : SUCCESS in 19m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d0552ce964e44859dc7bd03c5ba855b : SUCCESS in 4m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f731040386f443c0a46bd9621f4ad2e9 : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f99b76ad6b174495a335929ba1c44886 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/581222ca14dc4a3b9a0e9f6ca52ac027 : SUCCESS in 2m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d313857ce014671b51d877c17967dc2 : SUCCESS in 27m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7071876293b646788fcb52ca205c04f4 : SUCCESS in 7m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/847002fcd39047ddb57129e61c5a5d20 : SUCCESS in 3m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6acd08abc15d40b9aaf6451a763c4e48 : SUCCESS in 7m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3e63f39e8e284a4e85e22d50adccaf50 : SUCCESS in 14m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c1113e8a54f1451b90ffb639db1891e2 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb01947cc5aa4f6ab4dc21b4f2a4efe7 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfa99dc4e66440a0a082aa9c86fb98f2 : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d15bdb1daed6e2fc23e531ce3c37299788df7cc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:31:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd79f5962432584c847198e33491184a8133a14d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:32:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab38f17cdb04f0fa2c5733d272b271f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1067baca3f37443c8088d4220af48f5a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee91d769a633c8ef0ce4bfd811229fac22f15622","revisions":{"ee91d769a633c8ef0ce4bfd811229fac22f15622":{"kind":"REWORK","_number":1,"created":"2025-12-19 14:18:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/971544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/971544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/971544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/971544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/971544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/971544/1"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-19 14:08:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-19 14:08:52.000000000","tz":0},"subject":"update constraint for oslo.log to new release 8.0.0","message":"update constraint for oslo.log to new release 8.0.0\n\nmeta: version: 8.0.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: If631c5d46efa94478620704eba5ec222b8ba10db\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I58311ce6ab939168b4218094fedc131074476a00\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee91d769a633c8ef0ce4bfd811229fac22f15622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee91d769a633c8ef0ce4bfd811229fac22f15622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
