)]}'
{"id":"openstack%2Frequirements~971545","triplet_id":"openstack%2Frequirements~master~I3df755608d037a0d999ee941068c836eb4678949","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3df755608d037a0d999ee941068c836eb4678949","subject":"update constraint for neutron-lib to new release 3.23.0","status":"MERGED","created":"2025-12-19 14:19:38.000000000","updated":"2026-01-16 14:08:48.000000000","submitted":"2026-01-16 14:08:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971545-new-release","meta_rev_id":"528a0dc88fae093ec0449f3f575739c9309b67b7","_number":971545,"virtual_id_number":971545,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-12-22 15:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-16 12:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-02 14:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-12-22 15:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-16 12:05:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-19 16:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-22 15:31:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-12-22 15:32:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-02 14:28:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-16 12:05:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ee7ce1ed2ef1122760ca646e01e2a276ef80fe91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-19 14:19:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b75f00a8f2a382793f78f1418ae9db5c6c160fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 16:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58a50c8827347a992f0df2f0354d1e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7c86a5b8db4254a552396ae97b2564 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb994c9731bf446884fc2a7bfe3d2b80 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e07f394cb0c4e138883bc9fe79bb7e1 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/951610b0f4ad4d7686c03898f2c39e41 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80c37504b2f24daa80cc32f36056a50a : SUCCESS in 2m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/678f7d244a3248cd90d937e0a3193a05 : SUCCESS in 3m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae3834aa4311416c98d250c3e2d6aa5d : SUCCESS in 7m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1e9c490898954ab598b0c33ea9065018 : SUCCESS in 3m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7ca53c46416b4872a3357f2e57e9ef6d : SUCCESS in 3m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4cc73d5b506142408b3220fd9390b275 : SUCCESS in 9m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/11344b2b5bb740dc98dba12249447af4 : SUCCESS in 6m 11s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5aa3f0e15f2c483ca99c5b704bf0c670 : SUCCESS in 3m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/55ca069626534590a204dee2a823ce4c : SUCCESS in 4m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/85253b2c1aa44f2d911b85812b761417 : SUCCESS in 11m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/86fc608e713b4ef2a17a74ff3e3486e6 : SUCCESS in 8m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/732cfa7f405a4ccfbce21f5aec0a0570 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bed9b694d5e04025b44888c0c40dc91d : SUCCESS in 9m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5c343b361d4742d7ad6d5c7d4ba8d833 : SUCCESS in 17m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6235acbc280d416e90543983f38873e6 : SUCCESS in 9m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/538f3714598a461fa93bba2ca03d6e48 : SUCCESS in 5m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c5d396ebfc354426841df0a807cb9bcd : SUCCESS in 11m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b691c7d2725d4b749430022ec091ea2f : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2e927d4e32d141358ca639893b54045b : SUCCESS in 9m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ea268e6715f840d3ba33a023fd080503 : SUCCESS in 2m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/495df3cbd61a4c12ae9d4536a52795b4 : SUCCESS in 5m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ebc15e2dcc974c238c5275030aba0c11 : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/775656e412f44372b8de04df11c980aa : SUCCESS in 18m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ba59d9f1964b4a5382f146c77f01c8a3 : SUCCESS in 4m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/db14f37cf07043c49cf72832d050a181 : SUCCESS in 13m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bf250317844e4073933c372462427de6 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c73459cc81114c829bcaadd7463df9f5 : SUCCESS in 2m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6c44cd897904b3fa94ba1e3f0118ca8 : SUCCESS in 23m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a626a7c0fb8e41dd9cbf0ce95953bccd : SUCCESS in 6m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f0d3fc1edd8049f88f7aad409ebc6264 : SUCCESS in 3m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a42b069ccb1d45c0aa6d9981b406933b : SUCCESS in 3m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/14751e6f33094ba081e691524beb1107 : SUCCESS in 15m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9ba2d59797ff491ab70a875cf1e1efbd : SUCCESS in 5m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36155dbbd6334123b9b338455128abdb : SUCCESS in 55m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/056607ec08724282a9130f1dc4bc4ccb : SUCCESS in 1h 45m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cbdf168f886537e068f9b0b2f3a0fb21888d8276","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-12-22 15:31:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"023476d1d6e862c9089ba9482d22bd830446a10d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-12-22 15:32:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66a516f4f3127dae896faab9001860003c7cc745","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-02 14:28:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d5562d20b1a40284aa68dd0524214fc813675e09","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:05:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55bd5a2a984be05a5ab8cd29661cd6157278dd94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:06:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1f4fc53c77b08e9f478bf18f148616ce1a55be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:08:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2bd7fc392bb474bb27dbc2a761f8fe2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b121ff5ae7f64c8fa8323c9cc6ef0c81 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95ba45f53ff4c4989c712a7eecf69d2 : SUCCESS in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0fd2647e8db4a31b53e410871a6dc84 : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e4d7574e9d9418aa22c05af12fbbb51 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38843488d7e94c25ab0bf615340c7ea1 : SUCCESS in 2m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f9600c3ed0e4a4cacc8e8a00a962a6d : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e3a0415f8fc34ad2be878e458c7b78bc : SUCCESS in 4m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/50d9d7ab129e4ada8d1b01c40a92ae41 : SUCCESS in 10m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7cba459b8d874b30af2007366da335e1 : SUCCESS in 3m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6fa6ec249cf74bb3b5684a37a949ca38 : SUCCESS in 9m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e8451ff3827d4248a229f15103e0e17a : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/da9596df6b464f7c99bd8857a200ba1c : SUCCESS in 3m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/19a61ec15eeb42de82a2e91de6ae5acd : SUCCESS in 8m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6666f000c58146fd8d34b1213dbcabdc : SUCCESS in 7m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/819cbfb4fbb544b9ba1e3c59b397cc57 : SUCCESS in 7m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/692bb2152fc14214a3946d73eff255be : SUCCESS in 10m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/96968f9b92bb40f5948a20dc88c14da7 : SUCCESS in 13m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/14cc6918a75a48af9f0e4f6a4b20cc2c : SUCCESS in 1m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4354039fb4204c0d84f9132a17a0578f : SUCCESS in 11m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a606fc86bd1d4a279eba89f0f124b05a : SUCCESS in 32m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3fd6789bb67c4126ab11188431204b0e : SUCCESS in 4m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0eee06d08cb940f091cc34ed96201fd3 : SUCCESS in 14m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/118eaf01152649a6be2d2fe6377a39c9 : SUCCESS in 3m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90dc0f1e500a45dc8c89249af8e43b73 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e360193fac946c18b1f0a1c995df6db : SUCCESS in 26m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/906687446f7f46b093ce4c324d7df308 : SUCCESS in 8m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2faae94efc694dac89d739473f00b190 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4451325c984644f0961e00ae4c8dd356 : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8f4ac312723847d2b10278e05d901a60 : SUCCESS in 19m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/eefe9e6767854edf9f435d46e7d6003c : SUCCESS in 7m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9831bcb40f0c4308b610607ba4b36df7 : SUCCESS in 46m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cca98950e6e4d5d930fc8f18653aa45 : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"528a0dc88fae093ec0449f3f575739c9309b67b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:08:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14c9d4eb89fcb047c075650106b6767aa95221c9","revisions":{"14c9d4eb89fcb047c075650106b6767aa95221c9":{"kind":"REWORK","_number":1,"created":"2025-12-19 14:19:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/971545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/971545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/971545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/971545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/971545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/971545/1"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-19 14:09:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-19 14:09:37.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.23.0","message":"update constraint for neutron-lib to new release 3.23.0\n\nmeta: version: 3.23.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ic2d13f300264cbd3dac6fbd426b917ae44852b78\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3df755608d037a0d999ee941068c836eb4678949\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14c9d4eb89fcb047c075650106b6767aa95221c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14c9d4eb89fcb047c075650106b6767aa95221c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
