)]}'
{"id":"openstack%2Frequirements~972261","triplet_id":"openstack%2Frequirements~master~I65314bd7747a55c8fffc65b25cf1edf5053f77f5","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-01-08 07:06:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I65314bd7747a55c8fffc65b25cf1edf5053f77f5","subject":"Bump diskimage-builder to latest version","status":"MERGED","created":"2026-01-06 08:43:36.000000000","updated":"2026-01-08 10:07:00.000000000","submitted":"2026-01-08 10:06:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972261","meta_rev_id":"720c98c3121fe0c647b336fbef4b0208acaf2127","_number":972261,"virtual_id_number":972261,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-08 10:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-01-06 14:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-01-07 16:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-01-07 16:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-01-07 16:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-06 10:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 14:32:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-07 16:36:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-01-07 16:44:05.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"64c3a31e4fe28201931a5dfb4bdd0843574e30b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-06 08:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86661200b6c8eea994f09642544c524b63057df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 10:47:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273ee48ecd8442d988edfc921677041a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c539ceb7abae4505a8c25ee690572ea7 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d26fc20342d480aab89fe79d34711b6 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/214b793546b34c4d909ae93f6b988da9 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f49bd558dd874bd884a3968dbb2bbdf2 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c941d20a7bde47ffbc7963112ce1da1b : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f81cd53e3239415d8c5f4d5f413abd37 : SUCCESS in 2m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2abd8cddfc7e401997d36f07222eb5c9 : SUCCESS in 4m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/16ed0a5ccbb748d995060ac0f0279eb2 : SUCCESS in 10m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b18b734ee42247b2ba6c2259a0cac6d7 : SUCCESS in 7m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b8f5771bd01e4b0abd5dd23d01e4db97 : SUCCESS in 9m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ec163a273b2c4519869031466f9f8e81 : SUCCESS in 6m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/50b9c43601194ca0a46b9e0ee649c8f8 : SUCCESS in 4m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d6ce00e54cbd4811b9edba2d9350a061 : SUCCESS in 2m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1b83fc04f465447d94048553fb11b549 : SUCCESS in 9m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bca1dcaad41a4cfa955e11e50838c206 : SUCCESS in 4m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ddda34a4df2d41fa8a21b8507246c96c : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a50de856e6c246fcbc8aabdc5afc354f : SUCCESS in 6m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bb7132a0181e442fbd4341f676268fba : SUCCESS in 17m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0ab2e5cb5f646fbb48c7e17ec851191 : SUCCESS in 5m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/970d22f2ac4346729a52f2fe0574c27b : SUCCESS in 9m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c2e41a900849463ea37ed4bd15018956 : SUCCESS in 12m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d62aa221cb7b4deb905048d8613b135a : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fe216e076fa543c1a00019bca3d65851 : SUCCESS in 10m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/180982d48c624e5387c6edefc7dcf93f : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/443e50618fdc4989a1700c7e40332893 : SUCCESS in 4m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fbfd35e3ec8747e29280b63a32cc0cde : SUCCESS in 11m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/16a74209411744e4ade75f6972a9aad4 : SUCCESS in 21m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fc71fab3014e49cbac8cfc797a886fb7 : SUCCESS in 8m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5d1f3fb2e7d64d4e956eb2707f85622d : SUCCESS in 14m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/59e20562ab234045ac538c735a65d60d : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c0a3b9b337a437bb515997541df74f5 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/502e002640de4d2f99cc9917c76c4854 : SUCCESS in 30m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/32f7894223874f94ab5306b9d90fe88a : SUCCESS in 6m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2878a8eaa2db4e768585585f0f705299 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30280431485a40a289daea074b438dcb : SUCCESS in 3m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6c7ee36edd574c01ba1e8112980c4d3f : SUCCESS in 13m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6ee4e073bf5d494d8e6246ea2c7d7208 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a0d90b11900444f8f63a753eb6b8926 : SUCCESS in 43m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/018a245d5ff14466a145f6ac20900e09 : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"74b799cc137bc64688bd35153fc903f8643bf81b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-06 14:32:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6048ace39a0d5062de40e5d77f3ae02c77c660db","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-07 16:36:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"179a1d673d0aa4ede0a381121ebf24e160bbc237","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-07 16:40:25.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87c06a5420d2784397e8d8d21cc0c30f312c42f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:40:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7af6a34bd587c2cbcd296568a629f0db8d6286e9","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-07 16:44:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a68d9a953ba9a3ec9101573f843a5f6ea7f180b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:58:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42e25e6668e54c21b0876e817ca4529c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf5b3ae6cdc432cac6b593878744da0 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6c5c9c36584fcea9a2c3c0e878b29a : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d94b1eec80f44025913017f419cb7e7d : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8883f4ddde7945d094af7c5f82529732 : SUCCESS in 7m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b50e6b35c3684cc7aa975dafaa01d415 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5fb2b93e9e242fb9b58c95f513d0727 : SUCCESS in 3m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cb08cd5df4524014b0fc5c2c1305cf77 : SUCCESS in 5m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c56875c911784605951967d504900910 : SUCCESS in 6m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d0cb1bb4f518425b868c8c618ec2da56 : SUCCESS in 7m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c373cfcedf0c487791f2507d28b61e7a : SUCCESS in 11m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6598303293df4af1a07ad150384d1045 : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1eb1090e72c84296b1b001d92c425418 : SUCCESS in 2m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/764292666d574c1cae710ebacf842039 : SUCCESS in 6m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/40eceda6af6f40b394a7ed32e4071ff3 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eaa6ad027dfe4d7084319bac1254d880 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1ae73be1a8b64b9c8f274c18d981b893 : SUCCESS in 6m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e21c10769e9341f598b179f7d6f6fa82 : SUCCESS in 12m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fb24d45462a943a9921325333a160d58 : SUCCESS in 1m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eb495e61897c4d49971bd9f3a3724dab : SUCCESS in 8m 31s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a99c6b1db7544085abb7c254d6ea08b3 : SUCCESS in 28m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/24383af0d29d4c79a6c47499d9060342 : SUCCESS in 6m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7b594fc453614904abe82a259f53a20a : TIMED_OUT in 41m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/72b507386a0d419da62795b3a22bc126 : SUCCESS in 3m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0add0e7d3ba4952828c1ec940278755 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/804ce45a196d49a2a7ff0e189be8278b : SUCCESS in 25m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1350309223964b96be0b1aca6950cbbb : SUCCESS in 4m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/32f137f6e253460f869d1b0cffefec52 : SUCCESS in 3m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/754b90377a264ccea50378e70b027263 : SUCCESS in 5m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1bd5f05eb0d2460b8729ee7b013904af : SUCCESS in 15m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ec64ea3be28f40f986b2581ff4bcdff2 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/095f13f28eea41929f00866cc570d423 : SUCCESS in 1h 14m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa7fc625fbce441bbee55f1882446ffd : SUCCESS in 1h 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e036feefc4584fe1d45fb8e6e6c90f8e1a035c2b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 07:06:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c551e357c012bba55d2a03c2256fd34610f78a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 08:15:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564630b8eadb4f9aa2559155368ff173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451f54eb3a324ffda1935ae7f1b1670f : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c48f0223ae4f6492df8472464c945a : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e9f332f23b44d20b6eb652260fb92b8 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddbd0544299e41019e6b2ce359a77a99 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbfedf99f2304577995f7e1f8f80be98 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e009653a04f41418b87e5d8cd37cfbe : SUCCESS in 2m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5f7b6081ce88490f8015b0044395b79f : SUCCESS in 4m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e9598362c4784cdbb9c553a8e7395b13 : SUCCESS in 10m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f2c6cf599c924bcfad64282aed9f887b : SUCCESS in 3m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a04136090dca4ebd8387f64de3c11ed2 : SUCCESS in 14m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ac680e4033f249b19ad008ff28c9e087 : SUCCESS in 4m 28s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c601112182c545bc944015a29c9953f1 : SUCCESS in 4m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/82096e720c2a4323814c97d6487fbf6a : SUCCESS in 4m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6045bce20255450a96ac59a9c2bf9acd : SUCCESS in 12m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/186268b1d70a4cf5b9df019998dda87e : SUCCESS in 6m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e6d52592d8f04cbf83624bff0c4e34f1 : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/91af4bfe23d8429b854e58b69299da4a : SUCCESS in 6m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/475e73925cf94fb7ad980c0db553d216 : SUCCESS in 18m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9791d24c6b14a85b2144ae714a03dde : SUCCESS in 4m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fe537cea75df4f45a7104a66876c7c7c : SUCCESS in 8m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/68cee6ff95924e98974f7dd23568d581 : SUCCESS in 12m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/57e37a34115f4c96afb3f991231cc6cd : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cf75bce9a434487fa7b77aa4e7958cb0 : SUCCESS in 5m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6e2965eecc21444ab13f1a5653dc3706 : SUCCESS in 2m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7a5531f25a9d42ebb87d1524b93f15bd : SUCCESS in 4m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/af52c1379f884f29a6d060516a6ffd78 : SUCCESS in 11m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/257af44f57124c8fb77b73987890ac3c : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/033350fd579849e3b27452cc855e39f4 : SUCCESS in 4m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9f6aef1f1ed74350929817080acad811 : SUCCESS in 10m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eb1d54d1620e4402bfd717fef7da85eb : SUCCESS in 2m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29ad0ccaf9ad4209a188c8a13c52540e : SUCCESS in 1m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4989de47a774011ba07665a84320f4e : SUCCESS in 23m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/94eeb6fc77414e3894734ec7153f9836 : SUCCESS in 5m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/422970107c0f4411855479c75da155eb : SUCCESS in 3m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b45d33d87394849889c21b409bf2a84 : SUCCESS in 3m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/37832a6c584f4f078ddff641234a22ce : SUCCESS in 17m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9730a74c0c544f619744b33f09b34d91 : SUCCESS in 5m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e80139932a22463eaf38a8c0c73f9af7 : SUCCESS in 1h 00m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4464a755006427e971de00876068950 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3689061c70c6f4ffd1705a5b30d400f7f2a5f9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 08:15:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"849d288b62911aaf64e90b508896b48168588163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 10:06:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c43049fbd4154986a4d8cb4506a700c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c610c5423a04825951dbf2eb1c95943 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9883370054b47818c8f0d2ef64118db : SUCCESS in 1m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0d3f640b98c4097a816fa7bff55cec3 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44008dd818a841749883afc2b64e1c80 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11a85519e8b541c7884852a3266d7605 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f1c037e791e46b9ab9a738bedfc6e84 : SUCCESS in 1m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a79b077715a644d29462100312e5b50e : SUCCESS in 7m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/145fe12dd7054dc498b7372e1e3f2e7b : SUCCESS in 3m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/17bebb6ea39644919443ab251175ac0d : SUCCESS in 6m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/062f13533b5940049a3f426d4732218b : SUCCESS in 8m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/73caf82eb21d4d6990c930b47f6f658d : SUCCESS in 3m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8734cb1880014c9387dcee82bc70fd3e : SUCCESS in 4m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a5f326ab835f4e4d8760ac9fb0073906 : SUCCESS in 3m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6e428478d8114008a9cc574562d932a3 : SUCCESS in 17m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bcaf58fb802749d98f4e8193c389c494 : SUCCESS in 3m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1b7eeed86b96451db2176283bdca1b4e : SUCCESS in 6m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fdd6d149738f41549af0036c31f1873e : SUCCESS in 10m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f14428bd0c6a4e31bb3f48c0cf03171a : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/56bf81b1bdb248f099d8c587853ad086 : SUCCESS in 9m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ff224ee7a07145c6ac935764bfa309ba : SUCCESS in 24m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b78c7e0ecfa940cdb79e99d424780e2f : SUCCESS in 4m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a967e602627948e9a7bc5972ee462363 : SUCCESS in 12m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/06e9ce74de9e4e35b7e41923425128ba : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5413e8626c024d00afd618fc55ba297a : SUCCESS in 1m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20a43471f51b431986c5dbecb83284f7 : SUCCESS in 24m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/47d2a1ac24c24e6fbfd541d953b4e15f : SUCCESS in 5m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/77ade7fd63f94852bac5831253dddd87 : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/465a6cb9bcbc44478c01427e6642d8e2 : SUCCESS in 5m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/296affcbbbf2430fa7d22f2609a399c1 : SUCCESS in 6m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/771d812e069e47eb8d7466ea39090051 : SUCCESS in 6m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbc11db11e2e4678a0c21d3c5b9f451c : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eda84638ad07460e9e563e0d60e5a249 : SUCCESS in 1h 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b28c1f7c4f7a3d787d4bfc6a4aa7189d13e1298e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 10:06:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"720c98c3121fe0c647b336fbef4b0208acaf2127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 10:07:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af5401a36ebf4509b2d8b1f7f4e2ffc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56100689d29a4d4ab339710fa3ed7625 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","revisions":{"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83":{"kind":"REWORK","_number":1,"created":"2026-01-06 08:43:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/972261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/972261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/972261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/972261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/972261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/972261/1"}}},"commit":{"parents":[{"commit":"40cae06c9fe1f1302cf3129e402ccb9525a32278","subject":"Merge \"Don\u0027t include type stub packages in upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40cae06c9fe1f1302cf3129e402ccb9525a32278"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 08:41:17.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 08:43:15.000000000","tz":60},"subject":"Bump diskimage-builder to latest version","message":"Bump diskimage-builder to latest version\n\ndiskimage-builder 3.40.1 adds support for centos 10 stream which is\nrequired by Octavia\n\nChange-Id: I65314bd7747a55c8fffc65b25cf1edf5053f77f5\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
