)]}'
{"id":"openstack%2Frequirements~972472","triplet_id":"openstack%2Frequirements~master~I2a7daba6433d28dca35f46a26ed03a9dc6037fd2","project":"openstack/requirements","branch":"master","topic":"urllib3-bump","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-23 11:42:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2a7daba6433d28dca35f46a26ed03a9dc6037fd2","subject":"Bump urllib3 cap to 2.5.0","status":"ABANDONED","created":"2026-01-07 12:40:43.000000000","updated":"2026-02-23 11:42:19.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcf780ec1e302881ad6a6d99ec0752e071605eeb","_number":972472,"virtual_id_number":972472,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-21 20:30:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-02-02 15:01:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-01-21 18:40:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 14:07:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 18:40:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-02-02 15:01:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"32712880fa58de1411db8668af909737448783b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-07 12:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b789dff0277535037226b2b4f2fcfd9fab46529","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-07 12:42:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"320e4e211d4682bf3191873ee96404e418c5e8af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 14:07:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41d21602ee384a65bf8315a5ac1a3451\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9189d49583245f69a9293f38df4c8e0 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23867e482e54bafb423bd6950271979 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deef4443d0e746f7a68c641a705f8f86 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f30880a858e141feaa65a83a7f27885a : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a0c679d69af4ba4a4c88148bdddf595 : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/021498ab4e8b4a6ca93d4ce9d3c1b8be : SUCCESS in 2m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2b738e74d8c24b39b34bb026947d9150 : FAILURE in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/303dbc39c20c43f18b26d4116742ea8c : FAILURE in 3m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e205c67f8c7f4c28af48630eb00e29e8 : FAILURE in 5m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/578aa5a5cac14ba58ce684dcc62fcd05 : FAILURE in 12m 09s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ce268739d3af4e178021fecfc4a0838b : SUCCESS in 2m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2ba35deecd0b42aa9e85b63c87612633 : SUCCESS in 6m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ca948475785740fd9003893cc9cfd93a : SUCCESS in 5m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1141d1fda5594dfbbce3af871fa30f75 : SUCCESS in 11m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3bf7c85d1dfb47a6bc19acdb97b8703f : SUCCESS in 6m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f4ec866fc393401a8431c2489f79ebe8 : SUCCESS in 4m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e106c2eb650c4616b1bd28ada1938ae9 : SUCCESS in 3m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b87f3363abe74dfeb73bc63d7dcb436e : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c67e749487154e1db719f13276002502 : SUCCESS in 3m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4b97cf1ea1284998a915793eca8f37ce : SUCCESS in 9m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7b2b087c6a57406fa822bf15a388ed30 : SUCCESS in 49m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/98453b2365a14cbbbe76a0665be5a0e4 : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4bc4d1b278a546179b64b45f32acf8be : SUCCESS in 5m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eac226e55da843c0be8464601fcecf9e : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5240280b8c6249f19852cfdff303b295 : SUCCESS in 4m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/34ebd94331774c5f828f21a66eb2d3e9 : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/44771f60b7e64d87834c562b82a4a31a : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a2a770b861748a1ba96dc32479b6490 : SUCCESS in 8m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7ab8004d45374f7cb8aa7b400edbe74c : SUCCESS in 12m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/28a4d862ac354d57a9e0004135e8875d : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f541e40fbcb344ae9a56fb0b01a3d679 : SUCCESS in 2m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19111ef942a747c285e58ef8846a6c1e : SUCCESS in 21m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/385d761946fc47d4b1077afd0183d27b : FAILURE in 4m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/09b1aa00727a462c9ad3a977ceb350a6 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed6c589e4eac4814a975c20464fe1d8b : SUCCESS in 6m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5f74fbd5a2204278995f407575ea8e49 : SUCCESS in 18m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aa4033a0e7a44dbdb2785aa201f6a8ac : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3494e7d16434d2591a5f87f3c691f8c : SUCCESS in 1h 03m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65e24b34c6be4a6eb24a94635592895d : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"41a37e8c29e391d9e059863ce8daadbbd20519d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-08 14:29:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec98b77f9580eed89dc0d36c2bdf5f673d33eaf7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-08 14:30:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f7d2aecb343108290f7ba9c19628f521de78a0db","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-08 14:30:35.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"85e2b0edee079ad1aea387e89d566886e1fc8f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:42:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb8c06af4aeb45e1be140c2ccf437364\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbc95fecd0a4d36921a825da2930230 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef957bdce7b4e48bab55ba77c10cbfe : SUCCESS in 2m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e3f46daba224ae0a5ed03f6525e9700 : SUCCESS in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1374d5dae29746428a909a6d83a07db5 : SUCCESS in 7m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/520457e090b54884ac5cb5dabe91ae82 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aabc4156da43438bb250841fb7fefb40 : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d5f1f8857778456d914c20207115f228 : FAILURE in 3m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f23089b03f2940eab43c6041a813fb0a : FAILURE in 4m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b299c7b3434341249aa91bd6d5bf5cbf : FAILURE in 5m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ca8e8f63727a49ffadd80778545a56cc : FAILURE in 5m 05s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8f354cb4e912493ca00ee6d16a41dc76 : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/084b16f60aae43c98311453fa71df330 : SUCCESS in 3m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/30fafc5fbb984589a1b3483f537fcadd : SUCCESS in 2m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0812be2f7a104f8e9bcc21078091cc52 : SUCCESS in 9m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2805eb8ff6cc4b5c9e1c1e7703476d2d : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/571560b89f8c449293e98121bb26db57 : SUCCESS in 4m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cff27a49c10049c292ced80500fc3fa7 : SUCCESS in 7m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c571197a596a43acb2d2263d163330e9 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dacab57989f74d67b79ef613126f32f7 : SUCCESS in 7m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dfc78b7bfe5f41988802f5e29f0bc124 : SUCCESS in 8m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/640a7c2c99c94ed781272c96db4e7ca9 : SUCCESS in 12m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/92b4247637804726ad80989243bbf309 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c187408ee94f4e9a990e971f4480cf4c : SUCCESS in 10m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8f5c02798dd44ea796560e18deb81375 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fa2c0658cb9745bb8618096745172244 : SUCCESS in 4m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/251868ee2c0f45179e1cdbae57ae89c2 : SUCCESS in 10m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/87614e3e51ef4d44a20bd76619cd8c41 : SUCCESS in 27m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/aabddbdf32eb4ce1ae8825ebb92b69d4 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4d76f051dd004960a4d0ec7a9f4d7d8e : SUCCESS in 12m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d64fc287207d43c0bf540751a075e328 : SUCCESS in 7m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dcbd22e9589e469a9a341cf164f9924a : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e72c712236184dacbd53382cba9b9e56 : SUCCESS in 30m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c94b8c286c604c8f8aa42c475a5b66cd : SUCCESS in 6m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5f8ca2a6a3fc49aea840f72db583ee9a : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80af018d2a8644879e915bb433afa21e : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/606af005c053415e943c62ba6210b9ed : SUCCESS in 14m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2a43d82818644e2789b7197dd23c7024 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e45e4855624406eb0a376ea40a6e043 : POST_FAILURE in 1h 02m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fcc36cab1a94c47b12e4796ac149e86 : SUCCESS in 1h 43m 35s","accounts_in_message":[],"_revision_number":4},{"id":"69134b78480f76438ca1e47532fde4f184439afb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-19 11:06:35.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"810eb2e1b921bfb6eaa6a8860019cabe10f6a6fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-19 11:06:59.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"d935eb0e656672be5ca2905588a2f1dc214792b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:47:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21ef999f1164bfaa9daee25060a49db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c44067e73d40ceb981c69d4b54a49e : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/311a745242b9449d815b2cda02177901 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69d3674cd39e483a98daac2d9794ad97 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d45dffa927d543c4a5219a6aee9b2834 : RETRY_LIMIT in 1m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/368efdb6657046ca9729c08e236ca822 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d031cb7f23349ed8e7c7299a06ecda3 : SUCCESS in 4m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7cf6ab1b7b9a4762b7ed8668c1422428 : SUCCESS in 9m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6c7e6df9acc246e18910f5b88af15c4a : SUCCESS in 3m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/42469960ca6b47f29dab7f93a8ae1ec5 : SUCCESS in 9m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0e0ab997c3d448c18e70246cd8c252be : RETRY_LIMIT in 2m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ce99bfba31b2402ca4d90c84bdf28628 : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bf723a50bd51473aaa89f861c5e1179a : SUCCESS in 3m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c6ec11d9317a4980ae6459cc40518c2e : SUCCESS in 5m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/da0f1c2185834353a8ed616964424e3e : SUCCESS in 7m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6132cef7197d41dc942feb370e7158b3 : SUCCESS in 3m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/21d0418298d44dd1b13e578696acf76d : SUCCESS in 7m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4a9ef1fefb2841709a8eac9366aee5e0 : SUCCESS in 4m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9319ca821087444489294c180f45cf1a : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53fb3d80529e4430b917168b33937953 : SUCCESS in 6m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d9c8ba3bc674eb1a6be1d818be506d3 : SUCCESS in 7m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1e892c1d53134b7f991f4bad003fedb9 : SUCCESS in 9m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/30aaa883c1de47569bc2873fab4d87ad : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/86823be4c50041ed8fe18e5fa8735f5c : SUCCESS in 11m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/156a93d134334d16a128e6be1dac8531 : SUCCESS in 5m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3408ab96e5d24855872b60466eddb882 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ab453489dc91408d9390548c7fb2d85c : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9e2d11d1ef814519b76c053dd7c40f30 : SUCCESS in 22m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c24206214954dec856689c0b7e0ccf4 : SUCCESS in 8m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4d00288c6afb42ed984f661263f44766 : SUCCESS in 9m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2ca7147c799940cdae7b95204e17cc4b : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4918c06bf3ea4d96b3906ca8cd24f123 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef9004733d6d4f129c0c64d3c5dcb6e1 : SUCCESS in 24m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5abf907f5ecd43f78fd49ab1989d67c9 : SUCCESS in 3m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8a65c28fe4d54e58a8ab8271db2e3fcb : SUCCESS in 3m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1245b05f9944244adad9f08afbf994b : SUCCESS in 3m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4a50b0969b09464e90367a7c7804be89 : FAILURE in 18m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a5fd40db43754d4dba56be91729669a3 : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fbc66bc6ca54066819b0de2b3adade6 : SUCCESS in 1h 00m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/944d002e2ddc4dd6809aced49f69a21d : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":5},{"id":"a7bc555311aa63b8edaf1310db6fb5f3386614b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 12:43:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab4ac16ab2dc8c3eb18ef9767d53f690b9cadbb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 14:39:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7becf73dad83469f969a20f517d67818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb8044de4834b55a570100949292c1a : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d8536d5b2d4e128efe6d7ec9613727 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3bf2c9531254a9f978ebe90c2b03750 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2453a1d7a24e48448afd9eda1e1bd76f : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59697432ca0a4df8819985bd9d688087 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d563da38826e43c6ba46bce722e34fe9 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a7630bcc5cb347da9ee5dd4295df9947 : SUCCESS in 7m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f531e35b5bcc4a938b989d4f807d9cf4 : SUCCESS in 5m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/471c9431196748cc88ee0cae8b283ddd : SUCCESS in 7m 23s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/53113b96f3604ea6a1356819eeee018c : SUCCESS in 11m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ea9f6f9dfcce434b81ff86d390fbb067 : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1bbfd34ce4f64b508e970390c6245f80 : SUCCESS in 5m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/37ca9f23ed4f4d249300836d2a0f6453 : SUCCESS in 5m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bf883e06c5854cbf9ec05d1c929f1fb4 : SUCCESS in 8m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aada0876d66740a1aaf6f795f8847463 : SUCCESS in 6m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/83e49345b22e471db16d124557818ff0 : SUCCESS in 3m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fb7bd266b9844c08bccfcb6ce5f96151 : SUCCESS in 7m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fce1b05ccb29462f8d03dcafa26e7e9a : SUCCESS in 17m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f1b701097d54555baecf62f8d1ecd32 : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/51549182c78b4a7b870400bea372f52c : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/96240ca8e1a84a649f3089f7581fdc9c : SUCCESS in 12m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1c96f33f6a6643bfa651e5e5c1141228 : SUCCESS in 3m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d54fa4ef82af4d18a2fed1450a9b575c : SUCCESS in 9m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/254f27a521484b15b15856bc56722cf2 : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f2696f10101e4cca9c087ac13cc9d4fc : SUCCESS in 3m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/65336b21ffba4fc2967be4c03152b3fd : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/055b16ec25cb4f039bc0c2affcd4b6f2 : SUCCESS in 32m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0a6d9d306de541e4b5b73dfa86be2b71 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f8f894ac6a0248a1a330f092eab95c3c : SUCCESS in 15m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3657594b3de94297bd0025fd2d290abc : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f4590df04924eb4a4554aa6c6fb4e8a : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a3d4fbd06244d52b7b63686e45e3495 : SUCCESS in 22m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2d0a59e503cb4fb49905e670db6160b4 : SUCCESS in 5m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/22fe9da06c634c6bb224742464f8e8a0 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9966eecdf35c4880818f977eb801ae96 : SUCCESS in 6m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f25232c5d495402382ad421935637af2 : FAILURE in 13m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/02ef569b4d164a26a0d584edc41ee842 : SUCCESS in 7m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa435f3202d74e88abb539fe5ab81db6 : SUCCESS in 58m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/766ab3ac6632470fb71f88340ce18f7b : SUCCESS in 1h 50m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f55c5d84f692715f2c06f4b9ed6475022329224d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 15:37:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73755b36ca95484898d010d3dfba7b8982293520","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 15:37:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5965d646c14f0e45bb56beaaa8484e229212e707","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-21 18:40:48.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"726b4113099fde2591260f1cfa86545caf2c7c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 20:30:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d950196b28154156a68dfe83899d3a78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200df589213f45debd370b1ca11c663a : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca77c9c38f204c7b8d5e16c6faac5fa6 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2627db07a3ee4fa18f5d715987863002 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30884df3f796420593a00d6724bfa7fe : SUCCESS in 2m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/923dcfed81444b5fa095103030604ec3 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2011085e767943acaa38ffd6246f86ff : SUCCESS in 2m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/961a039ca9644d028923268fde74d32c : SUCCESS in 7m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/906857d6033f4b29a744bfcbf374e159 : SUCCESS in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0f7e4da9b5b0493ca1645ad065b2f48f : SUCCESS in 6m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/92da674c2c214a548eb7c5cce14f025e : SUCCESS in 3m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c59a35db75f24e3098d90370e1c3f505 : SUCCESS in 4m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c7af2feadaf249a589f2648d0fb871d5 : SUCCESS in 4m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a172be593cd34870a1b706ce42618430 : SUCCESS in 4m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1191bad0f6174f5f9d5c6e55fc1c6e37 : SUCCESS in 12m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/34b8df2ed5c94d3bab0047c3cc495138 : SUCCESS in 6m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c04660d73ff04005ac4034b6eae8e861 : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1612c20dfb74466780fcdc11c6b8aa8a : SUCCESS in 7m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d93465316f5e4d04aa8650a313b519a4 : SUCCESS in 17m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/457c42fbe5df496cb90c647e8f149205 : SUCCESS in 7m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fca0b3da1c3d4ba8ba5a76502fdd3849 : SUCCESS in 8m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/17cbda8e73ee41049a6290aec9b151bc : SUCCESS in 11m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3a889715efa94629bd5ca89c687b8595 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/71270202f0b0413e9f967d9728fecd57 : SUCCESS in 6m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/270a289a3d21459c9c68c3029c0a0738 : SUCCESS in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/878d87f935e14efabfec33c1e04def55 : SUCCESS in 6m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f41dd73eb2aa486caadb9f3518e844a0 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0729925a88d44db7b2ee631e37309a6d : SUCCESS in 27m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6239215c6200462aa24a2ee81ccdcf36 : SUCCESS in 7m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b0f29cf903e4704a71525f6ab0fae39 : SUCCESS in 14m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1c40e431578741139a876940eddb7f70 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd4c49dd074f4dfe9a634fe2c2e62f3d : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87b215463b1f4a199f7c6730136e929d : SUCCESS in 27m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/55b18dad180d47ceaf07402858539547 : SUCCESS in 3m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a899227a78024051ac5e48c445b2db3e : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4071006a93c40f4b03eb5b64c265f92 : SUCCESS in 6m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/849dd84e809d42d8a7f69d3cfac2fdb8 : SUCCESS in 21m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/49d5e8d0286848cea821b4589b6468ae : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7a333fc721b49a98bf1acce094f3e65 : SUCCESS in 45m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f21b48473be403c8467cca2975945f6 : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":5},{"id":"1ed1b4e5d95c1b519c37d71509c05f011eee3a97","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-02 15:01:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bcf780ec1e302881ad6a6d99ec0752e071605eeb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-23 11:42:19.000000000","message":"Abandoned\n\nfolded into parent","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"82485d97610632b8cae14b4170a0a37802da2b1c","revisions":{"227154e38971da8580cd1db4cc23117ae017a45c":{"kind":"REWORK","_number":1,"created":"2026-01-07 12:40:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/972472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/972472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/972472/1"}}},"commit":{"parents":[{"commit":"c3756c7d1cd47b027a3b358c1bea30fe55b58639","subject":"Bump urllib3 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3756c7d1cd47b027a3b358c1bea30fe55b58639"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"subject":"Bump urllib3 cap","message":"Bump urllib3 cap\n\nWe can now use the latest version since kubernetes has removed the\ncap [1].\n\n[1] https://github.com/kubernetes-client/python/pull/2496\n\nChange-Id: I2a7daba6433d28dca35f46a26ed03a9dc6037fd2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/227154e38971da8580cd1db4cc23117ae017a45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/227154e38971da8580cd1db4cc23117ae017a45c"}]},"branch":"refs/heads/master"},"801af216601bc1935cbb9b493111f32cb0baa4fc":{"kind":"NO_CHANGE","_number":2,"created":"2026-01-08 14:29:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/972472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/972472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/972472/2"}}},"commit":{"parents":[{"commit":"f83e97fe612afcef7e73c24f8910c31736b37750","subject":"Bump urllib3 cap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f83e97fe612afcef7e73c24f8910c31736b37750"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-08 14:29:59.000000000","tz":0},"subject":"Bump urllib3 cap","message":"Bump urllib3 cap\n\nWe can now use the latest version since kubernetes has removed the\ncap [1].\n\n[1] https://github.com/kubernetes-client/python/pull/2496\n\nChange-Id: I2a7daba6433d28dca35f46a26ed03a9dc6037fd2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/801af216601bc1935cbb9b493111f32cb0baa4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/801af216601bc1935cbb9b493111f32cb0baa4fc"}]},"branch":"refs/heads/master","description":"Rebase"},"61ab332811047d4906727e4a1a3d4f5688517578":{"kind":"NO_CHANGE","_number":3,"created":"2026-01-08 14:30:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/972472/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/972472/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/972472/3"}}},"commit":{"parents":[{"commit":"2d209dbbe7da8abb08dc289921c0ec60235af0ee","subject":"Bump urllib3 cap to 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d209dbbe7da8abb08dc289921c0ec60235af0ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-08 14:30:30.000000000","tz":0},"subject":"Bump urllib3 cap","message":"Bump urllib3 cap\n\nWe can now use the latest version since kubernetes has removed the\ncap [1].\n\n[1] https://github.com/kubernetes-client/python/pull/2496\n\nChange-Id: I2a7daba6433d28dca35f46a26ed03a9dc6037fd2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61ab332811047d4906727e4a1a3d4f5688517578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61ab332811047d4906727e4a1a3d4f5688517578"}]},"branch":"refs/heads/master","description":"Rebase"},"e263f81da3eb76d686c6f03ca666744cf3b90ed9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-08 14:30:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/972472/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/972472/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/972472/4"}}},"commit":{"parents":[{"commit":"2d209dbbe7da8abb08dc289921c0ec60235af0ee","subject":"Bump urllib3 cap to 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d209dbbe7da8abb08dc289921c0ec60235af0ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-08 14:30:35.000000000","tz":0},"subject":"Bump urllib3 cap to 2.5.0","message":"Bump urllib3 cap to 2.5.0\n\nWe can now use the latest version since kubernetes has removed the\ncap [1].\n\n[1] https://github.com/kubernetes-client/python/pull/2496\n\nChange-Id: I2a7daba6433d28dca35f46a26ed03a9dc6037fd2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e263f81da3eb76d686c6f03ca666744cf3b90ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e263f81da3eb76d686c6f03ca666744cf3b90ed9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"82485d97610632b8cae14b4170a0a37802da2b1c":{"kind":"REWORK","_number":5,"created":"2026-01-19 11:06:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/972472/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/972472/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/972472/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/972472/5"}}},"commit":{"parents":[{"commit":"2d209dbbe7da8abb08dc289921c0ec60235af0ee","subject":"Bump urllib3 cap to 2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d209dbbe7da8abb08dc289921c0ec60235af0ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-07 12:40:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-19 11:06:35.000000000","tz":0},"subject":"Bump urllib3 cap to 2.5.0","message":"Bump urllib3 cap to 2.5.0\n\nWe can now use the latest version since kubernetes has removed the\ncap [1].\n\n[1] https://github.com/kubernetes-client/python/pull/2496\n\nChange-Id: I2a7daba6433d28dca35f46a26ed03a9dc6037fd2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82485d97610632b8cae14b4170a0a37802da2b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82485d97610632b8cae14b4170a0a37802da2b1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
