)]}'
{"id":"openstack%2Frequirements~972691","triplet_id":"openstack%2Frequirements~master~Ib67395fc79716d747146a4b4b381deeb5e39122b","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-01-29 12:47:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib67395fc79716d747146a4b4b381deeb5e39122b","subject":"update constraint for oslo.concurrency to new release 7.3.0","status":"MERGED","created":"2026-01-08 13:54:29.000000000","updated":"2026-01-29 12:48:21.000000000","submitted":"2026-01-29 12:47:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"972691-new-release","meta_rev_id":"a70d55a4befdc79d41e1579ae7e05d01bd41a743","_number":972691,"virtual_id_number":972691,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 12:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-29 10:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-29 10:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 16:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 09:20:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-16 12:08:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-01-16 16:56:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-01-16 22:06:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"0f5ef55eaf757e3affd03dd885df2f95b1a55964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-08 13:54:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a709d84ae3ceb2e2a6696210216fb8d662b2447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:07:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82da43485e71410ca93889e39e4c7a92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f06ba976fa499bab846a3f62927fa4 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448a0a31afed4c769524d4fd8d210cac : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6abfcc4c877b4ed49d870852f6d3af50 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95f9c3292e8d42cba81c88fb909ec2c0 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d90f6d0099aa41079c58824edad1444f : SUCCESS in 2m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/621aebba58804f9e9970ef86a5f048b3 : SUCCESS in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9cf1aa82c9943d086376d7d94c962c6 : SUCCESS in 4m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5e6e6943380e45f6b47d739b6cd376a7 : SUCCESS in 5m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a5d43653cdd34869882b4f0f923501c4 : SUCCESS in 7m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d31836f53e6f400bb8776ba1d645d108 : SUCCESS in 9m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b67d037ded17489593c19cd49a60c818 : SUCCESS in 4m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e210f927671841a4876d253a930af077 : SUCCESS in 4m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4cad8d3f4fb34d69bb942abe26d9c55d : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/59a128f54068424e8d28fbd2e3d48be5 : FAILURE in 15m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c0ad3e7da34c4efe8b760be680dd26a6 : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0defbdf96dc24c6fb8a4897bcfb21f2e : FAILURE in 6m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/588479b3273e444ea84b6afb1f22f955 : SUCCESS in 7m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f7ed29d5e51b46b986a85f5587b18437 : SUCCESS in 17m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f1aec3e08774441b3a5a01e8b920f7c : SUCCESS in 5m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6eb3726cbf0e4d13b4dc1a86f1ace031 : SUCCESS in 9m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6b84624427674658b75b60ee9d252508 : SUCCESS in 10m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/675d5448a3da4d2ebb6fdd1a88f3ea69 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9a0ed16f9cab4a138f38568aa5cd29bf : SUCCESS in 9m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d2b142a2f32648438ec3d82d32dc2e95 : FAILURE in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/77cfb2c6e8bb437dbe7dd6d7cc0a74f1 : SUCCESS in 5m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c7d46361166f4ecc94a3ce151a161982 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f38f89c41719461ab6e04f3fef2c152c : SUCCESS in 25m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/75fe5fd552c540cb967f2e19d9027f4d : SUCCESS in 6m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1c2718e19faf447189f8372f1414b90b : SUCCESS in 12m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b2b8a6fe740b43ae907f5867111ecb5e : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24b70ccde8fa40149ceb5530c35e3eaa : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dbb881bc3c44acf97aeb7c1377fe6c8 : SUCCESS in 24m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/20a1315e09a943a286d7a63fa59c4e21 : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bee120ae394c42cfa71159c6b1787b65 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b940fb816884ae3bd4351c174929207 : SUCCESS in 5m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0cd47ec54c5b4e19a8ef8b09f4a50c8c : SUCCESS in 21m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dd738e3d529045139c37b0143cc6f1e5 : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96d3376427324f3c98898a329d1ecc28 : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e41cd0f5b454ca78ceb9f1404410588 : SUCCESS in 1h 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7989ef8f1223a54809c323bb18f410c96d0c5538","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:20:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77a117ffa1d16e709342b2ab612ab4d75874091a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 10:38:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03087bd015cb470abb1b4e300c7bd18a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4918dd26567a428bae97c731f223fe19 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93aaf5f0b166422cb0774a27757b9856 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0365db402cb14514a26b97088c71110f : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73b9a669e6d94dd49c39ae4c0fc221d8 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95f370b6640940e6a582669a3be194a9 : SUCCESS in 2m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c12810fcd8aa4d079d5bfc25b25c2d55 : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6d9224ee55f049af9a8ee6a1d9ba0fb9 : SUCCESS in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a04e86af65174ff497ac60349c61085c : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6a08d52518ba43ad88458ee6393c76d6 : SUCCESS in 5m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5281e590cfa74ee28d4a3119fd6b1e77 : SUCCESS in 10m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c1df4e653f404bd89e042bf039c3a12a : SUCCESS in 2m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1657dbe3362c40ef9b9aa19f7a589ace : SUCCESS in 3m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e8762d2c993c42e1a61c322f0562767d : SUCCESS in 2m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3a2d87932e8246deb06fdc9f136ecfb0 : FAILURE in 11m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/45748a78e545478cb0463c76821dc3bc : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e09eb047998e4672b65d9b4515d7097e : SUCCESS in 3m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/797f63a3a83d4780b573a389c93d7dd7 : SUCCESS in 4m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c4a61349c07d481291deaf8065d7a4ef : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95ef9ed4ef1b4bf4870e1e72f066a747 : SUCCESS in 4m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0a5a04c939a94fae91f76dddead3d66d : SUCCESS in 11m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e8a1dd173e264db2a0e26ed94a478642 : SUCCESS in 8m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a52b685c14804f22b5ce7130723355fd : SUCCESS in 3m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/963075bc102d428492744947b1ddd1c8 : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/095d009699294e58af074481291347f4 : FAILURE in 6m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3d78dbde0bbf4024b0107f5e92d7dd64 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ee3b0a55a6674412bbb1a49db8a16aba : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cae3725ffc604975bf3570893ffa6efc : SUCCESS in 24m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a14c68a4b0c43e680bfba797427b291 : SUCCESS in 9m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/983ec47ae29a4ba29d1f41adef31771b : SUCCESS in 14m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/35b2f30352df423f936dfffff72527e3 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7e087bf8a24495a81ca00e78a341b1b : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a153cb33bdb242da9203ee08e649ef22 : SUCCESS in 25m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b7a9581b0c924458a6c8037522f4e8c0 : SUCCESS in 7m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e539914f22ae4b6e9b9c872f4c5d8f2a : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e06b8143bcc542db8fc07525f8c15091 : SUCCESS in 6m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2e55e3d34b6b4966b200cc4e46b5aae0 : SUCCESS in 15m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ee4c6ca3693447f848ca3e5fe6fb6d1 : SUCCESS in 4m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/998574f4a0974161a9bda6b509e8e049 : SUCCESS in 34m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8a492f353ec4285a4beaf852b62ae81 : FAILURE in 1h 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"35f81513077dc953ddbc4ad5efaf7976f964e992","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-15 14:04:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e59d3173d6d7152384cc6b073fe9254ae24f2232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 14:57:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86bce5bebfa747c193d7f75091927e42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa9ad9b3540454fa10888f9ebc5a26c : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be1b9219ae8439a9afc87b4dcdde80d : SUCCESS in 2m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/328b27f38f0344c9aa1c8a5f0d05fab2 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a453bed37e74f60b87141d370d07cb1 : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2097aef753a54edeb63bc5e497a424e6 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/036ab4d40f1f4b8bacde7ebd1e205699 : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a0cef56ec7e04d849181c19866e53db9 : SUCCESS in 4m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7c6a7e4eeb3b4638bf49315ebdb850f4 : SUCCESS in 6m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bb2a18486beb4eb185df62a0accb534d : SUCCESS in 6m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3e3170d7a2694776854f095fa9163151 : SUCCESS in 14m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/224b01959c7047019eee6b0652543b75 : SUCCESS in 3m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2e1d29c59b6e4987859112a30632273e : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f63f285d39014d7b8fb22ae5e5152711 : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3d9ad0245ad2495caa0a0ec12d2fff61 : FAILURE in 11m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/278e87b96f7f4eb79c32a46a29bbe1c0 : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/24e237cf5a064b0688b685fc2b72aa19 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9400820ce56f489fa908020484ae984b : SUCCESS in 8m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ba8972eed2084507b1846c220aeb7094 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61ae2df864b54768bd2ea3522458f732 : SUCCESS in 7m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5fca0f4806cd449aa05bfc9a29424e7d : SUCCESS in 8m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/98a2dcec5c6b4d6d8dc93667eef1a2a2 : SUCCESS in 7m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3aed9a2b95024bc4a03dca4d2d5959ea : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1d4e3d1db4e0472388f08d265ea0f7e8 : SUCCESS in 9m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bbfae6b9d8b946d186c0fe0618dbd189 : FAILURE in 5m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7bf79989558543d887a78aea69d6bfba : SUCCESS in 6m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f105342f46ae479dacb69db4d00395ba : SUCCESS in 9m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0d08b56ad2e54d70aa50b75a92280076 : SUCCESS in 33m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b9875ee0a1c4c8d8c608201b2228407 : SUCCESS in 9m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2ff8fcf606944b36bdc300ee8e8b16ff : SUCCESS in 13m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7046fb77387f4e4ebb6d598448c18578 : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e5e07cf42a9f468f9bed33528aa4f416 : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba4cc60ebac24389a2f60e3073196770 : SUCCESS in 31m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/71fa7e82512f437492c4da8e7b07e5b2 : SUCCESS in 4m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0240415760f04be6a4f9040a365e8815 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46b9abc3ac034c89a00296ab8ae03b3e : SUCCESS in 7m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5f4bfd78991042bcbd0672aef53c0073 : SUCCESS in 19m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/091138e5ebb846f5aa47e0c0a8f4fb86 : SUCCESS in 7m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e962470dcc0d4de1884c87f73937cbe6 : SUCCESS in 50m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80613e32c24e4480b8ddc0f6e0bb6cb3 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"04367c726f1f6d59ec8e31399e501b646767c8c9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:08:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bca31f5479f4c45bbaf9524beb6cdc48251f506a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:09:05.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5acf83db620d8cad3bb5ac2b7bc5b12f8f8ffde8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:39:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21ac0ad6890b4acb96694d0ee478fac7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d084dafcda4a1395d0594271850d84 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d78451ae2de429b8651a34a83d9cbeb : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/082a54d0ec3743869848360e5440c823 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89d6f0d580dc48798c7aae6700345900 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c42dc7ce44aa49afbbc05fec28790379 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbe3ef91bb7240f4a92c65900cb46970 : SUCCESS in 3m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a0825cdf19440a3a3bc6394f7fb0680 : SUCCESS in 3m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4d5a13689b4a4cd0a014bba32e2e2dd1 : SUCCESS in 6m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fd849ec516504399b93b93c626251999 : SUCCESS in 12m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d64e1697ca4e4aa1bd15e28f7a8931a2 : SUCCESS in 13m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6c4d44e86d144a928faec27a44318c9e : SUCCESS in 5m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/df1c4bd40380442da8832df682a7e825 : SUCCESS in 5m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0ee69e7d00094798b8b2b5e0ea44be48 : SUCCESS in 2m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/637d4c27c04a4715b935c4c8bdaa152e : FAILURE in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/87a410b5ad2d4b4dafda1158f5bb1062 : SUCCESS in 3m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a1988c6386b84569a411953ce490a33c : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9abccad246be4f56a1cc284718b035da : SUCCESS in 8m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/460b1f80417e429e9afda23238917671 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1a60dabd67b446497474e1a15ebc39e : SUCCESS in 7m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dd6dd782beee4396aed541a5a0e1b78d : SUCCESS in 7m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a9f97494329e401987145f45a5c66d97 : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6db292fd75bc475e9bcf7164abd8ea5a : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/353077f4c3e44e6ead4b6be1f70d2112 : SUCCESS in 10m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9a7a24c9cc4c48c88ed42bdd9e630d05 : FAILURE in 4m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2932d629212546bf806ad68e4db7a6d7 : SUCCESS in 3m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/42ddb3ed84964cc1887210d6c12186da : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1baa3334dded4ee6851310fd9313debe : SUCCESS in 19m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6eaa38db5816409abcc6c2d6290db4cf : SUCCESS in 7m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/10d72e4690334d33b93edd07df7369d3 : SUCCESS in 12m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9968212288c54717a2c3e6e116dc23a4 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee65aa4cb96440d4840bce18859f712f : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fce6e59010d4c868129fe16664c0995 : SUCCESS in 25m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c0e45707b23240fb8d217520c077622b : SUCCESS in 8m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/33af667a47ee4fbb8a11298b237e8465 : SUCCESS in 3m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc7643e9fff64566aba2ef1abe90b4cc : SUCCESS in 7m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2a90a2b2238a4ff0993ff02e855b7d77 : SUCCESS in 14m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cc7aba469dee4d42a426bfb991b62ef9 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb49fd32351e44bd8371346a9b88e307 : SUCCESS in 46m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83c04429e1d04251b4480c4094738c62 : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":1},{"id":"adb282540ac4b6c02f81dbdfac3c526c0afbfcb0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-16 16:56:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c0327dba13216181058f140e04339fea36bc67f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-16 19:46:32.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"745f60acfb63ac8c8ae5e4b449addee04990ff8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 21:11:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf12f9b5c1c9463792cf1685a65eab18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a07d97f75a64583a7e5303cc990e88b : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c4a369858b49dea474b8060219607c : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/029964c8eaf74156876ced92ee5e926d : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be16d56347564edb9957cea879ba848e : RETRY_LIMIT in 1m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62a861cf858444f982f7d08e3c815ac0 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5de2b826c6f47fc835853f31fc9ae6d : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c996d9ea708a483da72dd4bf9d8f0da2 : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c5bdaca2472f4a58962a369bdd3328ce : SUCCESS in 6m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9b19bc47e04046afb6eddc847b46f86d : SUCCESS in 6m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e4909a63c70d400ea988ef44d68e9ebf : RETRY_LIMIT in 1m 42s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c5c59a492985400995240e47ab625467 : SUCCESS in 3m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8ca9c7da21294cb8ac33ba57290e8425 : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6ff8fc963e9d4af7bb53940e22c73689 : SUCCESS in 2m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3ed79a79f6ad450282a69f4bba82eb9e : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a852d2c5a0aa46c6a8c748a3a067484b : SUCCESS in 3m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c151a02ea6c4416c81eb25cf151ffbff : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fde1ada09fa64665b5277606fc4a8f9b : SUCCESS in 8m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3dbd582f0e6241f08679eeab2f5b521f : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29f736b9c80940df853656c586723c38 : SUCCESS in 9m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9a78b7a77e9d4e1c899ecd98368e7f63 : SUCCESS in 8m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4f8cf48850f1453a8407158dff7d8ab9 : SUCCESS in 7m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/332ba1fae6574d2a8a5742cdc004caeb : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/be6c2b27f7084430b16f9b7655b38d72 : SUCCESS in 9m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/85217fed7ca74ded825ef6ac48234c33 : FAILURE in 5m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d354be82fc334b91ac17ab20c54724b6 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/83403925bfef4e0ca36d5c6343390ebb : SUCCESS in 12m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c6f82f484201452d9103874f0f40fb29 : SUCCESS in 23m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9c7e5b1151f04dedb9bde3d245f4edfc : SUCCESS in 4m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89168b2b2afb467987b95193d0732588 : SUCCESS in 8m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c5b45918f61e45e3a510c98a5930ae97 : SUCCESS in 3m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4aee520d6bed479b90a4f57fea043099 : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3fe478f553b84c46976b1d9a731d63f7 : SUCCESS in 25m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e1ab4d0c7a28464da87f9fd818145f3e : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9910941b8ad74df49e31334c6b1e6162 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba8f1aaf84bc49bdbe0b63e94516f26e : SUCCESS in 4m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/329ea836254c42a4a3831f93d629b34a : SUCCESS in 14m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9926bc65dec146abaf0042eb67ad3f6d : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a8e0032051b42f8be838dc97651b42b : SUCCESS in 54m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff2be1764ccc42ccbd41a6c9439ba2bd : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":2},{"id":"59cfdfbb7fb028d31e0a6ad26b62d00ba3c12d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-16 21:53:23.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d3b52d85f1701d2f6edc680fca70ebdc01dcd11c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-16 22:06:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"48920b2640f75f3dbfd72f4fd37e38e74445a0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 23:51:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5b5321cdddf44628438f4bde585900e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6ec2a8271a4f0bb4b744fb9e43e284 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5845ce9929914c1091943f2b13bc523c : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65f28dec613e4d0cb57f8f968ca12bd0 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dc01370777d406ea0ee1355e7c4c7e2 : RETRY_LIMIT in 2m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fec703cf8e0541b88e3090cf520154e4 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bb599d7ec384967a5256a97d83d758c : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c54804ebb28a46d18224fc61c70137eb : SUCCESS in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d0a71f15286148be93d3b1961d6e2e9b : SUCCESS in 3m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/403d8a2ecdd640bcafb920c47da5c1a5 : SUCCESS in 7m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8ed18bf1c4e34856990f9bda3d77c1cc : RETRY_LIMIT in 1m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bad160058a374ee28a999c05017316c3 : FAILURE in 6m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/66699d15178846c4a82406967c8045ac : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ceb927c294504e3288bcd006634c2e0a : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/df622511c453441c93a763aecb63ca66 : SUCCESS in 12m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5f6c2946599a47598c7e0ce9f8dfeb7e : SUCCESS in 6m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/15ac51a16fd646cf93f0eab0bf8686fa : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e1b4592179014dbbb34c6fbf99a90c9f : SUCCESS in 7m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bf80e7b287d340df803d6cf40d07ff00 : SUCCESS in 7m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/384d2fe5fab14069ad7eaea248a3f078 : SUCCESS in 3m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e07ceba5931941c2bf0e2364bcb5070b : SUCCESS in 10m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aa0c4489c60c480ba9ae2d2bd895de83 : SUCCESS in 11m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/41db6c34d48f4addbc5364ee96bd348b : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/14d7e18733c5417f9725f6f3b34f0f5a : SUCCESS in 10m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/230090855024418dbbdd3909cdd76779 : SUCCESS in 6m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/14d422194cd04f3fb7b2a67df929a631 : SUCCESS in 6m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2febb6747dd54c4ebcee6345abff9657 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/952aa12fd87b47359cbc6b9b99967a8c : SUCCESS in 29m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0df0682f74da4a5ba94976743e76641b : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f5e81a458a8d49438170d56c254d3c3c : SUCCESS in 8m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/990a4a403bd443789d0beb4e27c34d14 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f092258eab3466087c7ab12d327752d : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/424fec1c20694e72b8a790e10423caf4 : SUCCESS in 25m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/88bfc777439a4e2e88266ba4739453a8 : SUCCESS in 6m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/20582f2924894b74acd7c5258010353c : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12592f9cc2cd4570be59abf3c714a49e : SUCCESS in 6m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/398605b0c1fd4ae1abbe8a9d3312dcf5 : SUCCESS in 18m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/438d163254f74429ac6072f897a5d8b7 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/658ebdfba2e740b8a842200a4bc96d90 : SUCCESS in 1h 06m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4e33a0a78bd4fbc8267f06de0048634 : SUCCESS in 1h 52m 54s","accounts_in_message":[],"_revision_number":3},{"id":"cb76aed49846cb455c9b1f9803b66b02cb0b9bf3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-20 16:38:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2b863c30ac5634063121d03b4de5a3344d7f883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:03:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3557c1b81864ac48627cbc4ed25acd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476e2d9fa1b340908d5eea90e93b4d9b : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab2dd65445949aabefe7a0bedfaf13c : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acb7f72f84d341b39ed4df487fb7558f : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84fbb3fe151a4a80a57a19d1c76527ac : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/388884dd4ccb4e479c6a8a54d4d09d69 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d9abf0ace194f6abca770a9563a2e55 : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3d0d1ebf350545fa8bdf764ac2ddf4a1 : SUCCESS in 4m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ab1b60ad287494f94870276035cda85 : SUCCESS in 6m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/367a88b33a42419786c63e06b8ca43fe : SUCCESS in 9m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fb4185311d0f4fcf807f6e5b6efd3744 : SUCCESS in 7m 08s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6b658a672e144fc1b5ad681579a1c618 : SUCCESS in 5m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1aca36cfc2674a06b96a248920229634 : SUCCESS in 5m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/43ca2b45669e4380ba4bb3637fa08ecc : SUCCESS in 5m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e5b30c6a78fd4468a9634bccb15b761a : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/154266bfdc4f4914933ef1f4ab02e983 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/203fc5d503174764bcfe99858aad1294 : SUCCESS in 5m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d85d0dbc923646e49ccfe6b667e1c634 : SUCCESS in 4m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/88337081dff04c569e6c9780bff5d166 : SUCCESS in 17m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ab01b81296941cea6bc51014f6a4b61 : SUCCESS in 6m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6e237beacdc5448796ac70431450d3d4 : SUCCESS in 9m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fe0f23b993d345a889ef6c9edbd4b966 : SUCCESS in 12m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6866d97cacd2419c8121e418513b82c2 : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ebfb839e5071463c931f4c30fb51cc73 : SUCCESS in 10m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/097c218210ec45549483e19288b732ec : SUCCESS in 4m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6e23435282c54fc29e5832d72168e99e : SUCCESS in 5m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/41890e543a784976a6fc34ede7ae100d : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ea4c794df37b48ea881a530847019053 : SUCCESS in 19m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9ed91bca0a1f43bd8fca3380e9d977a5 : SUCCESS in 6m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4df28a93e53c436f8e7d46f214f95bed : SUCCESS in 12m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e6f17b49a433481186dcc217657bd463 : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5ca1948940a4eef9681fb9983a3c19d : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1071a1d7ca14f0ebfa08ab58e1a2030 : SUCCESS in 21m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b4b43962ba14431fb5b81a5030171f6b : SUCCESS in 7m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0240e43472ea4eff9755d380f7101fff : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bc3ed323be443e88aa4afc84fc4b776 : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/24b95bd073b346b1963c8ee755e7baf7 : SUCCESS in 7m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3233b5b8ded14184912d029f19a822d2 : SUCCESS in 3m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b268f3134d46cfba24c7d2a649cc62 : SUCCESS in 37m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ddffafb4d5c42d7a63d269f4e99b70a : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0369c96ffc38ea4545a117e6a4fd8b3e25a83293","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 11:33:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4998f23f9a40d3abead0b726830fd3d16ac8f533","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:33:36.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/726330024e2749a3aa89b5439b3c8798\n\n\nWarning:\n  Change 973723,1 in project openstack/magnum does not share a change queue with 972691 in project openstack/requirements","accounts_in_message":[],"_revision_number":3},{"id":"01d28dc808cac555486be7369d34743f513b1343","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 11:35:35.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c657b675d1bed8226b3ac056d471a537731031d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-22 01:23:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"774efb519162f9579ab1ea1a3d0fed9fdb20c0be","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-23 09:08:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"966f6bc584ccdf509561394418ed885928c62a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:49:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a92634e66c14c4f90ae38628a27c1a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f37028a9034aed89e20aab4a98f18c : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42cf3093e5d487f931cce9287070a15 : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fd77070e31b44249dd1894f77c32952 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ea2aee3c76f42048427eb077a492c9d : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c55052f9132f4e3fb286b3953b884bb5 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3ac17b900f245ac9f385ba1a8b8b074 : SUCCESS in 2m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d3eb3b431bad4b55b867e6d229d77d07 : SUCCESS in 4m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/50b34a5db46540e39eeba9e67550855e : SUCCESS in 6m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2bf97b204a5e438382ddad47a02f9da6 : SUCCESS in 10m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67b2547123494e2bac33caf6984eec4a : SUCCESS in 6m 57s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/94a3cefcd45b442694487cd05dc695f5 : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0729355ee1104680ac3de3dca5d5b6c2 : SUCCESS in 4m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/be90dfa7369d47c19401f3c49ef4c20a : SUCCESS in 5m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0ec31a85f6a646c0b089234d923d2594 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/30f38c491f444c2f9a111b548b5d6044 : SUCCESS in 4m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b13ff826a750494dbf86be508e7a6def : SUCCESS in 3m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/50fca22ec9be4661a53b180f4f1671a5 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ed33583c85f342528dce69c9404f135b : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/444125e0efef4771a39827c439523de9 : SUCCESS in 4m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2e1c0d1400c14b8bb98cd606bd55f415 : SUCCESS in 9m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7eb1d163fa2f4a14bd30453c17de49a7 : SUCCESS in 14m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8853d09350884e71a846877aee5f064f : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/71925b8b697f4599bef400b60b441756 : SUCCESS in 5m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d7b19d3d180540daa3d1c02a90249fa7 : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/45243ac89f884024bc4f496d44eaaa30 : SUCCESS in 6m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f45dd309ab0a4751bc38ce49c0d60336 : SUCCESS in 11m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ed119c3c8b5840d0afcc839ee3b01670 : SUCCESS in 27m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/77c3494f619142afab04d63d5a508410 : SUCCESS in 9m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6ecfe4eb8bad46c59d30a4d7c6c1dbf0 : SUCCESS in 13m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bc57e68bf91a4bffbd76a0465496c52f : SUCCESS in 2m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f80d0efd60174e319186a8c3830bd6cb : SUCCESS in 2m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24ef57cc5c804c43a6fb4d0aad54bfd2 : SUCCESS in 25m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3831fdc74c8646f99f5124c9afecb50e : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/61f4bbebfe1b4c2fa3aaaca9db12fc84 : SUCCESS in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4fa270cf47d4612b68982607ae1430f : SUCCESS in 3m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/72bf1e8e7dc04ad6ad809d528ed592fa : SUCCESS in 7m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ba1aeca3a572465886d8a2c7f9ad1fd2 : SUCCESS in 7m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63f7ea0e34504ef0b49f4a711dc787cb : SUCCESS in 35m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8eb121f32b5433b8f122d92ae7f6ef1 : SUCCESS in 1h 38m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e4d3baaa65ce600a31f1d145e27b7f3fa68c0193","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-29 10:26:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f069a7b9cea2eab4e365a7e8e2ed6bb93e6fed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 10:26:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05632508acb3264b5e86b8d21d583cf55b794197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:47:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f64c190a9f4993838f3759a5c3b389\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd50c71d025a43cdad2d652ce956a6bd : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0cf55c3ad8420b870e6474f8f4221b : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f8683fa52414f87a882a654666426b7 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36dc0d5df94c4c2cb868ae75670a2d26 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee726067e48f426cbeb2815a97389a2e : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/489fed5443c5472ca57dae116d0979f1 : SUCCESS in 2m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/41100bae7def48fda82853a18257145b : SUCCESS in 4m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38cad5a2d319425bad00f2372132003f : SUCCESS in 12m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5c5a157ccd1840c098d5ff1201979537 : SUCCESS in 7m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/50920da3cb464319bec9619ac65e5397 : SUCCESS in 11m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/75543437ff604dfbaa3c19a45d871a46 : SUCCESS in 4m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/139a93984b134ce89fc67adfbeb1cc2a : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0c4c14dc4eb641cfbd05c2fff7cf318b : SUCCESS in 5m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2e57d116d25240e38824840685919e38 : SUCCESS in 17m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa7bb9fc6a264bfdbf3f8b05242fd68b : SUCCESS in 7m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bda4afc506b5441db19cb811fa34eb84 : SUCCESS in 7m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0f2ee15beff849fab91e50e407edab1a : SUCCESS in 15m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1764e8ce0263479eb5fc124eabfab9e0 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1dcdd83067e2488a8d94a1f48edb5d4a : SUCCESS in 12m 58s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e34b8bd13b544564b9d33fa002b0ca48 : SUCCESS in 27m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac2b1298f0384ef1ae5db9fdbfb12973 : SUCCESS in 4m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c17b070f30344563b0e2ea30e64dc563 : SUCCESS in 7m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e825b6d0eb5b48e0a01b44ce254403ac : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba322b02d64c4d58ba40ada6e12f581d : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56140c98afa84ca58dc1767aa36d429b : SUCCESS in 30m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c6b37c286a9b4ee6a20ecbf129960c6c : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/994469f2d2e34166b9fe8e7adc0e6b72 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56ec97ef928e4ab398cc096f059416ef : SUCCESS in 6m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/efbe1f90df9744dfaaf6053a54e01dd3 : SUCCESS in 20m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1b67a44c65154cc88df4fd88d44a6a42 : SUCCESS in 5m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef11f67a363c44b58ed34ca3590b6cac : SUCCESS in 52m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/562b37c920824a529a3b28c2c4aae0f2 : SUCCESS in 2h 14m 43s","accounts_in_message":[],"_revision_number":3},{"id":"56b8e2609a60a453173383c2dbf9754706ae2c21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:47:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a70d55a4befdc79d41e1579ae7e05d01bd41a743","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:48:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/472742430edd4fa9ae5e5517811081ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72ab86734e149429f405d8d195f99b1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a31cc245ee46982dbc69ed5266d4af3ab59f9c11","revisions":{"e144bb97d5c453a95b435bab0edf61b485433bd9":{"kind":"REWORK","_number":1,"created":"2026-01-08 13:54:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/972691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/972691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/972691/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:44:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:44:28.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.3.0","message":"update constraint for oslo.concurrency to new release 7.3.0\n\nmeta: version: 7.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1774035b33181269a7b36e913343906144f8de15\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib67395fc79716d747146a4b4b381deeb5e39122b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e144bb97d5c453a95b435bab0edf61b485433bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e144bb97d5c453a95b435bab0edf61b485433bd9"}]},"branch":"refs/heads/master"},"dc061936d1567f664388acf7abac1bebe53da555":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-16 19:46:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/91/972691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/972691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/972691/2"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:44:28.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-01-16 19:46:32.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.3.0","message":"update constraint for oslo.concurrency to new release 7.3.0\n\nmeta: version: 7.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1774035b33181269a7b36e913343906144f8de15\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/973693\n\nChange-Id: Ib67395fc79716d747146a4b4b381deeb5e39122b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc061936d1567f664388acf7abac1bebe53da555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc061936d1567f664388acf7abac1bebe53da555"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a31cc245ee46982dbc69ed5266d4af3ab59f9c11":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-16 21:53:23.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/91/972691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/972691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/972691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/972691/3"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:44:28.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-01-16 21:53:23.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 7.3.0","message":"update constraint for oslo.concurrency to new release 7.3.0\n\nmeta: version: 7.3.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1774035b33181269a7b36e913343906144f8de15\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/973693\nDepends-on: https://review.opendev.org/c/openstack/magnum/+/973723\n\nChange-Id: Ib67395fc79716d747146a4b4b381deeb5e39122b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a31cc245ee46982dbc69ed5266d4af3ab59f9c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a31cc245ee46982dbc69ed5266d4af3ab59f9c11"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
