)]}'
{"id":"openstack%2Frequirements~972693","triplet_id":"openstack%2Frequirements~master~I962e310c97206fa88009cba1c9df6338360597e1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I962e310c97206fa88009cba1c9df6338360597e1","subject":"update constraint for os-ken to new release 4.1.0","status":"MERGED","created":"2026-01-08 13:56:30.000000000","updated":"2026-01-16 14:09:32.000000000","submitted":"2026-01-16 14:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972693-new-release","meta_rev_id":"9a3689c2eaec3c65d490ea1ddb4faf1bb8c30695","_number":972693,"virtual_id_number":972693,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-09 07:52:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-16 12:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-12 20:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-16 12:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 16:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 07:52:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-12 20:27:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-13 14:09:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-01-13 14:09:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-16 12:00:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"9875cc84927a7aee0be7161a4cb5aed1d1746d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-08 13:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17e91ef4444da09fb6beb7c81fb1ea73f46229a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038203d1293d4adeb5e34f70b5d52f3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277eb0e5397c4e9fb1eae868113de11d : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b45369fda0246a8b7e0da0d05a76b0c : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4439c248d263418f814038089b1162e7 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bf17d30d25d4b5fa0eb818871229a30 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0aa7632c40b431992362b008e7e5d38 : SUCCESS in 1m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99fa53153d4843a1a7e8c00c6e3a6671 : SUCCESS in 1m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a0ede192442249e097f6926b93d95738 : SUCCESS in 4m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1cad6f24d5e04634be50946a98c0976b : SUCCESS in 6m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1d9ba2ecad1e4c6f9ebbc30ba6df733d : SUCCESS in 3m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f673d4cb303d42e88f38840e53a7a7f0 : SUCCESS in 15m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d780fb4fe56c46fe9d803101f5af2815 : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/608a9975e4714a07810e60434aefbe15 : SUCCESS in 6m 43s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fed05304fb604a098a69a45d2a6300b8 : SUCCESS in 2m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3a66eef3bc154618885a2349c19bfbe2 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/48ac35439c3d4568ac8a614aa8f6cfa0 : SUCCESS in 5m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/817b7ce0d3c9457ab49486e8fd9e2c8b : SUCCESS in 6m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/baab3f036e57471e9f14460ea4b8efe6 : SUCCESS in 7m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/71572675707f4ee0943ab3478b318e43 : SUCCESS in 18m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a009912abe7b4a409a18ac9c54f28043 : SUCCESS in 7m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e0e4716adf9d4319b445ba0d570c130d : SUCCESS in 9m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/891f28c3b94f4d03963482a6e62c1a79 : SUCCESS in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/12642563311b4821b5ec93562a3a22c4 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4c2abd21a0c54b8aba3e71973390848e : SUCCESS in 10m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c8b4d69dc38249a9add3035c5bf85301 : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/96e89b1970304f048bca2ce33151a3ed : SUCCESS in 5m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7ec023fc7d154ed49d216ee86c3fb465 : SUCCESS in 9m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/00d1a5f500094caa9a9c7386765a105f : SUCCESS in 29m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8cc68ef9309342948bd28734e3af7c76 : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/187ab20659e44f2cb5c5f098941f3ade : SUCCESS in 13m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/be5a7e4b8b6249079988db374d5cd756 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2da3a4a77c4475e9f0ee8a66fa06f7a : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ff89b16013841e3ad4ff49b716113f6 : SUCCESS in 22m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0ab030468faa4c45b82061400bf6822f : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1bab760bc30a49e0a3ea161f92bf220a : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bf2952e0b3a40e0a8b3aa4bdca8db20 : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/eda2e7cb64e64824b21231d85b9097c8 : SUCCESS in 20m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d1523e383b2348a69b02ee4b50a9d642 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99a5a25f80734890b1a5e8e8b41b3fad : SUCCESS in 1h 02m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/267c02923f7845908d9aee362851f292 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3fa8ef222ade58e37aec7cd62f053d3cd031c8cc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-09 07:52:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a5a1a3547b15d973a57cf636e17a01d73fbf51a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-12 20:27:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"277ed0fcb1d88267ca91d3b480290dd52e0d8551","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:00:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3909f0efffa4f854de14a54191a9b99928b4d7cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:01:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47a414bad6428c69fed0a95812b55bba643138ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:08:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492d54d535614b1498c11e5582bc951b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824c0699e90f4b00922464a23b5f8e1e : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86144297311245a5aabfbc7d11490470 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68e45c6fb1a246a6befe598ea7ef6683 : SUCCESS in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04c6b92b5647412f95e9f91d9b66f4c0 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/316d37b3aa8b463898a7b7e8ddfa8b4e : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a45411f96914327a064a5e85667691c : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e0ea06431444480f8b63d03a7f33f924 : SUCCESS in 2m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f694d29d594e407dbae5192dee449133 : SUCCESS in 10m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b656f8b7831d4518a7b657b9eacb4641 : SUCCESS in 10m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/913413f5248a4dc79ee8d7113f4ceda9 : SUCCESS in 7m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/23208e33e0dd4ebaba97fd47362850fe : SUCCESS in 2m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d8576654f6b54db88a56193207b6ecbc : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2fb184bc3abf40f0b3f7aac931a70c2d : SUCCESS in 7m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2cccbcb4c77041eebf63e12ea7a63e8d : SUCCESS in 17m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94982919e8c24d0dad36a03055d30e25 : SUCCESS in 3m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cc5da3abaea74928a17c2b4fe6d11ef5 : SUCCESS in 9m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9b1db4ad69704bd28a17bf94cc21f80c : SUCCESS in 12m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0edc851785d84a1db2af78adde42a054 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2cacef2fe9a3449b96c9d8da574fa09d : SUCCESS in 9m 55s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3656b74a3b9147d6a3ab3e0bd6b07a1c : SUCCESS in 23m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e949dce29b8c44a1b1b98b7ef8e00280 : SUCCESS in 8m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f5722b4508bb4b389d29e34c406520a6 : SUCCESS in 16m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/709446bd48504d9aaef693cd765727bb : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/10954d80577f4c9ca49b36a63eb68402 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e559dcbbfe943158605f38bef167c8c : SUCCESS in 26m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9e6230203d474c62b344122489968b84 : SUCCESS in 5m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/37d88a3004804d18a46d1a6c0fac0c13 : SUCCESS in 7m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0631c9af70cf466788e3a2ea8e78cf1e : SUCCESS in 7m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3d13ca755144743a5099d125838eb64 : SUCCESS in 16m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/82c7f435ed4349928405aa053afb0f69 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c3d4aad22594cb598e21a5e827c9a95 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c0c5aba0ab44b5b8945f95b6954e463 : SUCCESS in 2h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"61614481f75f90d5ffc15ec5e6a18a8a1070ed0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:08:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a3689c2eaec3c65d490ea1ddb4faf1bb8c30695","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62f55945be9d4e25b892527131207b7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0006a2160f4e70a9d3b3444459d9a2 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84c5cd191b9273e150e83df541fca9da56355b30","revisions":{"84c5cd191b9273e150e83df541fca9da56355b30":{"kind":"REWORK","_number":1,"created":"2026-01-08 13:56:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/972693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/972693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/972693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/972693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/972693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/972693/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:46:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:46:29.000000000","tz":0},"subject":"update constraint for os-ken to new release 4.1.0","message":"update constraint for os-ken to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ida51b1d52a70f7cb16342e42d54a2677f6574cbb\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I962e310c97206fa88009cba1c9df6338360597e1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84c5cd191b9273e150e83df541fca9da56355b30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84c5cd191b9273e150e83df541fca9da56355b30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
