)]}'
{"id":"openstack%2Frequirements~972697","triplet_id":"openstack%2Frequirements~master~Ia7719163068da557c228775b9696b8e9b036fc36","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia7719163068da557c228775b9696b8e9b036fc36","subject":"update constraint for oslo.policy to new release 4.8.0","status":"MERGED","created":"2026-01-08 13:57:35.000000000","updated":"2026-01-16 14:22:03.000000000","submitted":"2026-01-16 14:22:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972697-new-release","meta_rev_id":"6807460681a5113ee8f57ea2bcce154f20b7a5da","_number":972697,"virtual_id_number":972697,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 16:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 09:21:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-16 12:10:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"dad8b518e70c7725f7070031c727abd9109a8d9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-08 13:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16e0bcd2419a9e8b61dcb017d379dfdbbafa4049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:00:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42e084cd871e414da6bcf9eb3e7ce124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7a5a989dd347f8ac52cb9551460e66 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd720a86e934ce7806c87f6236e8d8e : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53a8a15f8c184861aeb2a909024d2c7f : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6eb562b9e94c46e9b478c1a91375e4de : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0bdd88e099e436bb989e8e41c156c3a : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd638f880c9548b49d5cfce949330137 : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20c900160e424e879559f77ae15e0981 : SUCCESS in 4m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/59c759beea574be18cfd41f016ce062d : SUCCESS in 6m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/08218c6544884541936ed424001b33df : SUCCESS in 7m 00s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/79a08eb5fcd44bd5bd5f0af27c48d921 : SUCCESS in 8m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/dfad9d28ff7d4b48be879eee1f5690f3 : SUCCESS in 5m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4c2d68efcc1747d5b73a1f7e0b6c3e6c : SUCCESS in 5m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b6e5960685d44e028dd04a0c65380aca : SUCCESS in 3m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e2efeff6c83b4ea888b91d604aaf7431 : SUCCESS in 11m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b1ac4c6c9ac74556ae44be52e13dfc1d : SUCCESS in 9m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4939747bb9fc426f8a3842389f3eefe4 : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a682d3da0e29492bb2628ff8ee639195 : SUCCESS in 7m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/80dcfaf7cd104b14b54a2dab55a74062 : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac534fd5eff44248b49e17fbaf3274d3 : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a620e12d52dd4ae0a8adeabbbf507408 : SUCCESS in 7m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2b324299b4f24e03ba2b7bc29d922e7b : SUCCESS in 10m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/11804f6bcb0b44a1aa39d6889b733a5c : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/520d94f3de37491aba53303fb5b4cf41 : SUCCESS in 8m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cef2eee62d3c4ad28890f52e9d03977e : SUCCESS in 5m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e69372459ab34af0b5808205ff80e701 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ddcc6c2071754d58ba10dc61b983e230 : TIMED_OUT in 40m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/765d6121a94743a4a77386b29022110e : TIMED_OUT in 1h 00m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5bf89e0353b147d083d944e1dd7d9a91 : SUCCESS in 14m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/034f63fed9c5430b8436a9a6d97d01f0 : TIMED_OUT in 40m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0bfeca99c4994deea0ef92b74a7cf571 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ed8273f848d4940852c2ebef7331874 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72ee0aa95edb45d898657846ac1c54c0 : SUCCESS in 21m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/464a303de99e4c6d89fa409a3d5ab9a9 : SUCCESS in 9m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d5ca42c07b3f4c3da48ea27f9b91c4fe : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1126d22e10fc40269e893d57f0364e2b : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e6f2146e3d774cdfafbe64838e63dee1 : SUCCESS in 25m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/63201020e9fa42e88977c225e3a59140 : SUCCESS in 5m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ebc4a0be6ca46bb94742294113b9ff3 : SUCCESS in 49m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d74197ce946e4bde9ae0cf17e9f7732e : POST_FAILURE in 1h 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c202d01f289bd9231eec62af5170162ec20e80f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:21:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffea11ab7df5b93384c859c693cd15ccb621ac50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 10:43:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c57b5bc866b4e46b503d350479ce4b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c07fce0072f49bd9beef3884035d44c : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68e7a8832ef43c1872da7be582127a4 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f95de9a69dbc40b58c990ceb33e8a39b : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef69de8ec4f6436187d9f4fd7399fdf0 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/816fbeb4c4b04fd88129e55feae8436f : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8df14fac9ad74e7a92d7d3d272e905d7 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a1e7ddbb616645abb8c8a679b9fb42f4 : SUCCESS in 4m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9176e8a4bb104335931a92bfb1a84874 : SUCCESS in 6m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c3c53d5c424c43729ccc75c809e11ba4 : SUCCESS in 6m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/25431bcc97d34b7fa5d098fa67e63687 : SUCCESS in 9m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0459c3ca2e9841a794f114cdde00d975 : FAILURE in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0d93823bdf3a49d19336340b8391a7ee : SUCCESS in 5m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6b5e70f6c2634ad5ac2f222f6d25418a : SUCCESS in 3m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b8aae2c045c54027b86807caeee89b80 : SUCCESS in 13m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/36876cae055540dd9a9a96e49569ba49 : SUCCESS in 4m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/646524e39a314a7e90d16ae8c1f0b5f4 : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ce6d88b2a14a4967a97bded53ae8d126 : SUCCESS in 8m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6bf58d5187494a41b8a224c459047b25 : SUCCESS in 19m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59e4de62940b4e71b99cc614a2012369 : SUCCESS in 6m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5ef4e0761fd444459cb5704267f43eef : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0bef11d3b6fd4ac7ac6f7b7a34648d9b : SUCCESS in 11m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a8ad20b570474060ba20efec8b03c795 : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b3ccdc0fb2774339bf1a6c9d7823a3bf : SUCCESS in 9m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/218806f65dcf4785954f545670d923fe : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3a3d5c6f71364401ac5c049bdb5c103e : SUCCESS in 3m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/65031ea7725b49da9127abc34221a17b : SUCCESS in 10m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2ad8f18264dd420cbff33f6aefd95a3c : SUCCESS in 26m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/03fb5be3cb484989a257c0bf7a8ec05b : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1930f49248ff460292c3f79508cf984a : SUCCESS in 15m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c38769889b4d42c0b400d1d00933901f : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e8b187e38a64ec09726fc1a7f331f0c : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37bc76c46fed4dcdb04ee7e81fa36ede : SUCCESS in 25m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b38994032bf9441c9cbf22677238431b : SUCCESS in 7m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cdd104c6931f4b889c23d20056e4dd9e : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd9b061151434446a7997c59ebc14b69 : SUCCESS in 7m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6d923499d18d4c619c20f4660010fe09 : SUCCESS in 19m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/257c3b0bf457493cb32a2b9ca8693f51 : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78c55945a0f44775822577fa897c6992 : SUCCESS in 1h 05m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef20d3941d5540eca0e943651ed47632 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":1},{"id":"23450fac97501c77537685ac12377c197d393c23","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-15 14:04:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fba4d02a3678ae19e8b6159df0a10af702583c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 15:38:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f1ec8bb4fd431aa493317246cc124b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff7aa04713f47b2998989ddc0f4b55a : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bec7bb3cdf42d58fce0087f4071266 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c86d45f7c85c4a038fce5a992911a0dc : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7d8951c2e904652b5318d41a4542579 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dde397da27d3404ba98b0f9460574477 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e327aa2d035a4a3ea188d16985264984 : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0dc9db5176884bbb83a87e1cebd55f7b : SUCCESS in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9f080b2ca7cb41778b2be9a7978e2bb8 : SUCCESS in 10m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/16c41e46879a40e3970ce21da759e488 : SUCCESS in 12m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d249c4a024bd46da8363e5ff46f962fd : SUCCESS in 13m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0c1295d207704e178a4eda1042e50f55 : SUCCESS in 2m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dc5b3831c3a64146bf44cf166c02ad8f : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fa3f220696c24ad590a70f0687e37392 : SUCCESS in 5m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/def1e93e6a4b4405859f74d9fc4f89a2 : SUCCESS in 7m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fd49d99e03254f139561ea82ff191470 : SUCCESS in 6m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/77934369bdf74f8b992c36fe2c93c0e9 : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4a084fe9ffaa4cc5b14ab50501c01a73 : SUCCESS in 3m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8fa3e5f9793141378ec56e643c059b32 : SUCCESS in 13m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3301bbf95f44210abdc7cb0a4780d00 : SUCCESS in 8m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f90f77e8b65b48509bef36e32bb5aa30 : SUCCESS in 10m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/01d27e64878f4ffaa3fc958fa3e17fb5 : SUCCESS in 12m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1cc7178dd0814e35a283cb3361c69ffd : SUCCESS in 3m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac3b2dd5052b4b56b1e2edd11315b694 : SUCCESS in 5m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5177f875ca7a444da963297b7b704dbc : SUCCESS in 5m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/25e28758944b4f71b704fd32286f0d4e : SUCCESS in 6m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d2cb18b781074e5083f79015f9302455 : SUCCESS in 12m 33s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/caaa0913d8584e189d5933e5c8129404 : SUCCESS in 28m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b8404c15f94494f869ad09e45a5f597 : SUCCESS in 4m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3b0a347a9a6847539f0d37e6f3363dcd : SUCCESS in 12m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/57503028d29a44a581464a1f7e5aebd0 : SUCCESS in 3m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b4adb178bf934205a196049515aa40db : SUCCESS in 5m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14c2fdc8c5754f62ad9aebdcdcaf520e : SUCCESS in 24m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/81646b634e444b98b90c3d4d24b6c38c : SUCCESS in 7m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/35600c1c809f45658591345879d4916f : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e8debdafdaf48e9bfeaee0487fc32b9 : SUCCESS in 5m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d58598f530904d64b363698a6e16892d : SUCCESS in 15m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4ba8e0f35b5c4525a9dded436d06446d : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54711afeef234734a6cd350d81c7f7ca : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b518b60fbd8481e8e97ae092031df90 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"626073868bf2b7108a101902674742243e689470","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:10:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f87bf5ae8531d1cecda07bbfb186b96f52ec6cb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bae278c22de662384071c989ba6472103e6b8e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:22:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dbaa80304854a319acdeaf5bcb9a32e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d496c43e4ec64647bd7a8fa13cf84ec4 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d3738aa9ef4070850d9046ff79b252 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b37c28b988a4b7aab6f24f1c0e96461 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e415586ffa964e7fbaca45f38f6da6cc : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00bf3b84303341a6bb93e7ce50e5a21d : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/27c3f39081ac43c391987222d557331c : SUCCESS in 3m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5cc9de9d95224737b2ff12b36a125ecd : SUCCESS in 4m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/43e343c929ae42bd926126f604031cb7 : SUCCESS in 6m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/774377eb905148919ea7f4293a7e001f : SUCCESS in 4m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a847bdee167244a58e515e34cd6611a0 : SUCCESS in 10m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c94c7fe078384e31a88ee9527a1d0a16 : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f40ac6b1fe94c5092dcabb703fbc68b : SUCCESS in 5m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/37391ec5e9da4cee938862a8c7933f67 : SUCCESS in 8m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6fbe991ab75649b1a56f69774befb338 : SUCCESS in 19m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b4b824c4a8c43e583a60eb2a75abe99 : SUCCESS in 7m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5a13390ec16d40d4bc29e25fbae1a93f : SUCCESS in 9m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/34611c30d98846abb92316d01cbf2ada : SUCCESS in 11m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6d370823f9f84ec386c6622613a64783 : SUCCESS in 3m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/01d38bda1981493d84b276bc26226e4a : SUCCESS in 11m 31s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/53431541f8a04fd0aeb53e22fe3e614a : SUCCESS in 26m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b1c894cf91f2486781ba187df1f843d7 : SUCCESS in 9m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d74294246a254794a8788a4f48266698 : SUCCESS in 14m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d2104d5f5e9c403284d3179eb653f8da : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/075ab2278b3d4e60818d0ed6cfd14658 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4603edd1dda400095a620b9decb54ad : SUCCESS in 24m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9cf0028167474b7e960af2235a1bbb83 : SUCCESS in 6m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/df6cdadcc5ed4cf086a48937c84e7460 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6d7af2ddc0043939b10830657d50634 : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/85971d3850dd45aa9a4fd27d02d0b4ae : SUCCESS in 20m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/95d9f3fe0bad40dd8b6409f4a2f14185 : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39176dd6856a42188d15fd4a88de58b9 : SUCCESS in 1h 05m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1a761a37bd540b3a2fea0af9d1d5d94 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6807460681a5113ee8f57ea2bcce154f20b7a5da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:22:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f91349fac731bc11b510796d5f906216f1d47b59","revisions":{"f91349fac731bc11b510796d5f906216f1d47b59":{"kind":"REWORK","_number":1,"created":"2026-01-08 13:57:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/972697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/972697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/972697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/972697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/972697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/972697/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:47:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:47:34.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.8.0","message":"update constraint for oslo.policy to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I475e899a7f3039f45807e4e480987ce9ec7a537d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia7719163068da557c228775b9696b8e9b036fc36\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91349fac731bc11b510796d5f906216f1d47b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f91349fac731bc11b510796d5f906216f1d47b59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
