)]}'
{"id":"openstack%2Frequirements~972704","triplet_id":"openstack%2Frequirements~master~I8236c992cbd86781a998cc0948c4b7517ce4d895","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8236c992cbd86781a998cc0948c4b7517ce4d895","subject":"update constraint for python-manilaclient to new release 5.7.1","status":"MERGED","created":"2026-01-08 14:00:42.000000000","updated":"2026-01-16 14:09:14.000000000","submitted":"2026-01-16 14:09:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972704-new-release","meta_rev_id":"77f92d19adb05d95847f96d653ab9818773693fe","_number":972704,"virtual_id_number":972704,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:09:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:07:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 16:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 09:21:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-01-16 12:07:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"62ce2138e28d5f8a0248dacb0812740e9afbbffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-08 14:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1db0998ee5eee083e2ff9ef308456d170094447c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 16:03:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f1285c5272a454b9d219275714d584e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69aff7bf9f5449f4b7f4be3e42ba3c63 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad06879ac64d4f4ca0eb132d9dc71611 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fddda804f0f4c3fbc3d3be0a1a0a5e8 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7aef44b9ab94793a3b6c196b701e62f : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea76d5ec841f4914bbb1ab38d8fddbb8 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4edcbdf2152b489ebab3d209b0833e49 : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/50b895effa2a4b129b89f800b093a648 : SUCCESS in 3m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a156102722954d73a93a10c87aee0ff0 : SUCCESS in 6m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c03b828055fc4bd691a48b03e45d4113 : SUCCESS in 7m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a49be0e9b50640058254a8b15dee629f : SUCCESS in 10m 40s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bfba6153fe2a420c947efae6a7213755 : SUCCESS in 5m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4e6417a3554a4894bba023b0e0d0ee06 : SUCCESS in 5m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e2682dd3eda642559e5737b3577f1ebd : SUCCESS in 2m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7dc83c06e4154940a3793d0030ad4218 : SUCCESS in 11m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b8bcc1b081474f8bb0b375445379ef96 : SUCCESS in 4m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ad4fe31be03948ba9bd6ec1ae12c2c0b : SUCCESS in 3m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b316470b3f2a4e3888ea0d7988fb8b36 : SUCCESS in 4m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1b7ccfd9349347a7bb6fc41f973923d8 : SUCCESS in 16m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a894b0def2844728b4dc50cd342a17ff : SUCCESS in 3m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5eb387aa70c1420f9d32b176103037cc : SUCCESS in 8m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d3e39d2cfdca41e68212ad1e1e9bfd7a : SUCCESS in 7m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cce8b9f37ca348019c979506b54af918 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/09ca99dc4ff246d08956c58569b2a357 : SUCCESS in 9m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bb302d2f9b194b908d70d23d64f7d010 : SUCCESS in 2m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ea825b6b0d7482fbc67f1281d25db34 : SUCCESS in 6m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/44266269e6a34e05af6c44f656d53c97 : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4662f77562774c6797cf4afb199b8e64 : SUCCESS in 23m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7dd4efdfc190416792b1a7f7398e0644 : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/26f7b691cff343ac92278c54b6aafc35 : SUCCESS in 13m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9262a6f635f14be3947002fdeb2afeaa : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bd9efe428ed47e0973dde87c1ae450b : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09095e1bc33a44e5b4c2ef86dd53367c : TIMED_OUT in 40m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/370750e0a2fc4982ad7414ac0901ebeb : SUCCESS in 4m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d708f607e3c04ed1a88a6dc30750bdaa : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0fa31dac32d4481ac45dffa366e7441 : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ab167ba528244b2d8deb60a0f848a024 : SUCCESS in 19m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79846b2796c34521856cf441d1dcbdbd : SUCCESS in 4m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbd6cada8e68492f91da7996e149cf21 : SUCCESS in 55m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8703ba94fa084725a645d537278c3ee0 : SUCCESS in 1h 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"470d56c018c881eed1488b6a5608b0fd28f7f205","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:21:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcbe34e8e2f7bf99101376d9b296bc3aec437a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:19:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6088bf42a6f416f9ca9f746155ba3ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da39fbff8194cd298ea8bc60d97f811 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb387bab3cc54fb08d4612e5adfe7060 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10ad3fe54a6e4d4d825d81d51f88e901 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9cf958b6c8a418cbee489d557005bc3 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac473c8d806f410593dd3245bf343e3b : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79f22b45e67a40d88f40742604674495 : SUCCESS in 2m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1d3cb992dc4649d585c472205e50c433 : SUCCESS in 2m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc4b52f6891f4c5197acc448c0d177a5 : SUCCESS in 11m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/78c2689ab1194e28ada808c0c6ae1ec3 : SUCCESS in 6m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1c4d5c9e7bad4becaa303559f521db95 : SUCCESS in 15m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0466f23e2d6f4d5688f006f791c15727 : SUCCESS in 5m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/783fcfa6197c4bd88e4026a75bcf05ff : SUCCESS in 5m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3b6870f5d45042f3b44530b36b138dd6 : SUCCESS in 3m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5593eb837af249718c5f7536062222f4 : SUCCESS in 11m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0f08c428d10a489791e5ae97c6026733 : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8e875288879d4f19b4502b9b0dad6f6e : SUCCESS in 6m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/80b0df706b0943dda523e2feab0cfbc1 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/51feb4ca98324abc9114eb79c975f8c5 : SUCCESS in 15m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6661481ca07a4b13886a752a965409fd : SUCCESS in 4m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6c5111e34e3e4b25b9a60bea7aea354e : SUCCESS in 7m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0c453232e7c44eaca885f3746d556fb3 : SUCCESS in 14m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2834fc122a1c4221935df5c705ea2dca : SUCCESS in 2m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e676cd1a5e9043448fe7e51df1a90913 : SUCCESS in 9m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/025958ef7263473c9fdb81f3857ec65f : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/99591d7efda64415ab35a2e3c3448942 : SUCCESS in 5m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/faf926a437614db3aa3b2de24a015a40 : SUCCESS in 9m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/da6f5aaa0c024f6ea96c1e32a3433389 : SUCCESS in 30m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0324e71a5d04713879500c5bced4cb5 : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7bbf1d1768984e3fa3a08987773bdd6e : SUCCESS in 13m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b720b77a38ca4b5fa8768b2b7ae832dc : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f3841dae7cf4e82b078211f03347948 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0e6441bf97f4bde94dd6ff61513600f : SUCCESS in 24m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/beab350e36394f6394c7776b1995ab4a : SUCCESS in 6m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/be36d62e7b534f0a8abe9aa648f24c4d : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acff41b6418d45e1be24fffbe127eb5c : SUCCESS in 7m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3965b29f03a045ea937bf1d2cfa4da7e : SUCCESS in 15m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/477e0a8d69f842ce90ec24681a744288 : SUCCESS in 7m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e413a3225911411281e0d4441fefba2c : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/028de67b0e1044f7b29dc6aea2d83b7c : SUCCESS in 1h 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"52c7cb679062200efe2e5e4a41e55f208c49562a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:07:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98913df7a9f3cf5660fe4e1b2ad4ae4675e45f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:08:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b85efe5c54f0197621ed470947831bdc6e09166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:09:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c76a6d78a224aa293db722cdf3beff5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3dd78e34e6c4423b1b5423f10c79f75 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cf496db3314384b081ab34944d54e3 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba291a6ce8b04a9b9c3c23e604c28e81 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/738a705edc944c80bf10b8ee14b7d6a8 : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53f6009e3ea1452881ea4c8aa79b5a8a : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e69bc63d310647adb2914192899183e6 : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a15671bc01c49febfa19a358111f7c8 : SUCCESS in 4m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/57243d9e187041c9835200f3e774134b : SUCCESS in 5m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2d89fe062b3f485c8eaa58838c57fac0 : SUCCESS in 7m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/45690421aeae4bf1860886574eff0477 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4dbdab7f41fe4e86b212c91590aaca25 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/941c44d7f4624c95b23f6a2eab2fecee : SUCCESS in 7m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/21546a61a1b74054ad709bcf49601dea : SUCCESS in 7m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c0752d2d2e81492abb686c013037faa9 : SUCCESS in 16m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f61cb176408480ebb3fccaddaaf9a7b : SUCCESS in 4m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/46af68df4e594a0fb0801e8fe3d038fa : SUCCESS in 9m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f173e42f532f4d9f8e79c63d4be4ac4f : SUCCESS in 11m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f58733106be94bd6ab549be227c53cfc : SUCCESS in 3m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b2aa08a774624873a95f00c99c76fede : SUCCESS in 10m 17s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4ea8e48a8850440fab63456f0292daa5 : SUCCESS in 25m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1fab457509374a9c906db20ea795b9e3 : SUCCESS in 8m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4ba1d1eb67d64bbcb5eead0d1cf802d1 : SUCCESS in 15m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/54ab876ad3bd489a882efa7883f85b10 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33943f1d18a8429b8fcda6c35986f501 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53d8d8c93a5646959dcdbb0b3e8c0eaf : SUCCESS in 27m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c386a72c48374c90a6b0b1499d12bdf3 : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5a4c2b67a9bf4c449357e38cd715a349 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa8c9dd128e84ec79da57401c891795e : SUCCESS in 8m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/cf3ea6ce3d0941de8d680fa8e8bbdcf9 : SUCCESS in 18m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c60e955aaed843c9b9f1e567eba7a416 : SUCCESS in 6m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/656f6977765e47169abab6acdecbeef7 : SUCCESS in 34m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6cf09a73c9846d0bb24353ad38c0c67 : SUCCESS in 1h 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"77f92d19adb05d95847f96d653ab9818773693fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:09:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0dbcb4f8e0c3f53b99b692869ef76eefd393cf07","revisions":{"0dbcb4f8e0c3f53b99b692869ef76eefd393cf07":{"kind":"REWORK","_number":1,"created":"2026-01-08 14:00:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/972704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/972704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/972704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/972704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/972704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/972704/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:50:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-08 13:50:36.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 5.7.1","message":"update constraint for python-manilaclient to new release 5.7.1\n\nmeta: version: 5.7.1\nmeta: team: manila\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0f43d91f7c8c723698fab68636486682cd6838ef\nmeta: release:Code-Review+1: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8236c992cbd86781a998cc0948c4b7517ce4d895\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dbcb4f8e0c3f53b99b692869ef76eefd393cf07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dbcb4f8e0c3f53b99b692869ef76eefd393cf07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
