)]}'
{"id":"openstack%2Frequirements~973201","triplet_id":"openstack%2Frequirements~master~I3d451bc96d2ea6d355bebed5412f44511fc3b0c1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3d451bc96d2ea6d355bebed5412f44511fc3b0c1","subject":"update constraint for openstack-release-test to new release 8.3.0","status":"MERGED","created":"2026-01-13 08:05:04.000000000","updated":"2026-01-16 14:09:09.000000000","submitted":"2026-01-16 14:09:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973201-new-release","meta_rev_id":"5dc5e8b4895377e6f537d359a59e833c97f8845c","_number":973201,"virtual_id_number":973201,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 09:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 12:07:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"674cccd4115fff3b2d74f1a958dfae259619f1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-13 08:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31efacc8351f9485aaeda9ad9e0492d3e00f324b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 09:12:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e368d892eb4b29a0154fd4540442ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1492fb669484644b4c603cee97b66f5 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe5446c8b2343518bdb48bbc5e90f31 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87f15aa880d44945b0ebe052612db2bb : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3ed0dfd4d5b434ebed7d42a2a030767 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a604d96d58b84a69beaf89c30ba59de8 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/240650e9b5314b4e87d7ee34ce1b5646 : SUCCESS in 2m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/327b3b9a362e49c49b1dc902868ab026 : SUCCESS in 3m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/236594b5fe8940608f092294b61dd8d9 : SUCCESS in 4m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d9cafd7b216c4d80aeb0176e6cd2916b : SUCCESS in 4m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6c20a947c6274b7f857414f70d1037ca : SUCCESS in 6m 43s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8a4e16d722d647838d6b95acf3cba604 : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/61ce8c67a73f49c4a84ba3c86163b5b1 : SUCCESS in 4m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1488370dd1364101b94f1e8f0f8dd3a5 : SUCCESS in 4m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8cba2a43171a4275806a4732a7369128 : SUCCESS in 7m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8127378f0c3d42f484d9fdf1e11030c7 : SUCCESS in 2m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7ecf6384a6714cb588f83541064a0900 : SUCCESS in 7m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2e26dde46db44743b841a7c731ad94f1 : SUCCESS in 5m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/53abc03a2beb437593fbc568c9579357 : SUCCESS in 17m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e03b7ac36fa481ab36fc805ecd5f768 : SUCCESS in 5m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b2de50deb02c4318abc7d215b2411490 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dc5b6e4bbd54407997a5e8d5c83deec0 : SUCCESS in 8m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7cb96ba4155448fd8379a2c2ec26cf9b : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0cad4b01ed8b4dc9b42cf06cc4b27ea9 : SUCCESS in 6m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0e6af7e9cec84652a5bc20b4b5f9ca6e : SUCCESS in 6m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ca6078165f74a77990283bf51ea2734 : SUCCESS in 3m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/255adf2374dc4485bc89d1c033d831b0 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/470c05652a1a4cfa9a38c78f635e1c11 : SUCCESS in 27m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8785ea1334d649f498896579bbe24ae6 : SUCCESS in 4m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d497d364317f43ab8f1c2dd91f88c399 : SUCCESS in 8m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/267ea99011584a0faafb5fc9cfafe48a : SUCCESS in 3m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d0e724569d04c0eadd2966be9afaabe : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a926a172479454da7ef86176f897c4c : SUCCESS in 20m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9799fd64b46843d5bbf268aa32dabed7 : SUCCESS in 3m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1e783af4fb7f4215a9ff9d587a9a7219 : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e54fb50e298e4513bd11fc7188b4f05f : SUCCESS in 7m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/043263ae7d0f4c6b82d6fd36ab1bf06b : SUCCESS in 18m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c8939c2eb6214bf89f22f7ad5534e072 : SUCCESS in 6m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787de21178a447be8ff45829c01984e0 : SUCCESS in 1h 05m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad24f9ecf5fc4c9eb108048c0c5ee3ed : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"82aa76461677d3ee6671c5d7d53ee56745b21dd7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:07:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce805c076b64cf044ed39b65ba6cb7f40f3f64d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:08:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ada1763aca127dd2c9ce0e391f74c59e6210d1c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:09:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e93e4d7b5ed4bc6b8648e70aeb6ba50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea65232b9ee441d84aad37f381d62d9 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603321b446f645a09d20c5e3587f9aad : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6494c7f123954220978afb458151f780 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da4270f3c1a9421196fe7fb7032517c0 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/064156e643394d93a7a8a49530b9102a : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4946dede641d4a3e96e0cf8ac5cd5a03 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a35e4cd53fda4c9797fef6530a84d258 : SUCCESS in 4m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7be23d63746d4151a2fb38069a6f8964 : SUCCESS in 6m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/593e6e10bddd4ee9882c76caa65b3fa4 : SUCCESS in 7m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b168cce3917141b3842cb5de5eac7810 : SUCCESS in 10m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ffd87904b8ec43df8b76b473805c4341 : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/578a80f68cfb4842b9e248e320c51191 : SUCCESS in 6m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5b932fe88d1a47caa98b92768b423360 : SUCCESS in 7m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1c69d62fedfe436eac119af473780e1c : SUCCESS in 18m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e978daf030aa495fa3be195f4d411cde : SUCCESS in 4m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fbed7c6985d5417c8a2ac4cd6a150b83 : SUCCESS in 8m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/89d62a73e3e44a3cbc474fb1324eb0fe : SUCCESS in 7m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/28272771daa3470caf72757e10496123 : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9971186a2a354bc0ada6058e6f6ec01b : SUCCESS in 10m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b9ffddb686cf47709023828d03a17ace : SUCCESS in 28m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7387e84bf2b14fb68352fc6748c2fabb : SUCCESS in 7m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3b588d6e0ed14f609c50f35b8ff610a5 : SUCCESS in 14m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/26509148f3684abbae8c8473a4656c5a : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da999679119d40e685cb889be894c4c9 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79343132703541b0bf85200b7cd272c3 : SUCCESS in 24m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bb23ae23dc964670acdff82c36c63281 : SUCCESS in 7m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b49af0595689464fa7c208c4113ac931 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e2ae2afda224577ad7a2202350d8e01 : SUCCESS in 6m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8672cce054434586805407b7b5bed634 : SUCCESS in 21m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ba917f316ea6405ca139228ee70cf9a6 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/653390ec6e5742df97afe82b5641fdea : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdcc715449304ec298e2a47e51f2f390 : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5dc5e8b4895377e6f537d359a59e833c97f8845c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:09:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bf3f13a85e75faa979b9c9472a48a8d5f0112b1","revisions":{"2bf3f13a85e75faa979b9c9472a48a8d5f0112b1":{"kind":"REWORK","_number":1,"created":"2026-01-13 08:05:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/973201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/973201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/973201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/973201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/973201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/973201/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-13 07:55:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-13 07:55:03.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 8.3.0","message":"update constraint for openstack-release-test to new release 8.3.0\n\nmeta: version: 8.3.0\nmeta: team: Release Management\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibcb025168eb36e7cbe88915a1a6a8d060a55d6c9\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I3d451bc96d2ea6d355bebed5412f44511fc3b0c1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bf3f13a85e75faa979b9c9472a48a8d5f0112b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bf3f13a85e75faa979b9c9472a48a8d5f0112b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
