)]}'
{"id":"openstack%2Frequirements~973210","triplet_id":"openstack%2Frequirements~master~I5acd7fca3719cc8efc201fd8d8e9597772fde634","project":"openstack/requirements","branch":"master","topic":"pyroute2_0.9.6","attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-06-03 13:32:46.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-03 13:32:46.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-01 18:36:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I5acd7fca3719cc8efc201fd8d8e9597772fde634","subject":"DNM: Bump pyroute2 to version 0.9.6","status":"NEW","created":"2026-01-13 09:25:07.000000000","updated":"2026-06-03 13:32:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"85fed8ac513d973a410f8a9ac401eb589546d86c","_number":973210,"virtual_id_number":973210,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-03 11:29:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 11:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-14 07:39:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-01-14 11:00:37.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-02-23 10:39:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-06-01 09:51:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"d972d15d5feee71de1130fe73f543b24a90f924e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8859a0af6643ec2a70ef9334021d78ba60736c39","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:25:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b90aea6ac9021ec3801837d81a12a5c3e69196da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:28:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cfb47f27665573bf095f8d002091117a93b7718","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-13 09:29:00.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"308e5bf7d8727585267bece6e4e0b531434e299f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:14:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b150188643d435e8813c9d8be2ea184\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f653b1d94e94f1dba40385dda51ee2c : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a3eb1f453843b1a5ed6f3e132712e6 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbce145d24ce4ca69536b1413e18d724 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03f858f88bbf440b9866e494b0479f8b : SUCCESS in 5m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c3eb7cd281894c2884407d1895b2d44a : SUCCESS in 1h 39m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e54327cbaa94a4b8ecf9c5472934c97 : SUCCESS in 1h 37m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ce1e162a50c13bc51b1f7edd8d6cac231c173170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 07:39:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"196ada51c330bd5c8d2c41c87c2bdc0afcf855f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 08:41:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd244cfe5e8e4b0cb49021113c18be5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d94d5a293c476f80c2f406ac0c18aa : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa0395274b248d89533d4ed1001b3a7 : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8348fdef0d7b482986f17d647ef60b2a : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5bfb1669c354fb18e4f1f8a30ddfdbd : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71ab14c2a57943a590b618fb49cebbc4 : SUCCESS in 2m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4b6a0d01b5340c58f976f89159aa1d0 : SUCCESS in 3m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/beb0befcf94145c4b21916326741f883 : SUCCESS in 7m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c79f494bd1d94646930eeb3cb2d46784 : SUCCESS in 10m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/47d7270bdc614861ae17f48b250837d0 : SUCCESS in 6m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c90d9873057549e6808b03575ce0ca0b : SUCCESS in 15m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/34757019f9a84c6ebaf0a135eaaf16d6 : SUCCESS in 3m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f121f77bbb67484cb079944a5a8a5eb6 : SUCCESS in 6m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9648da116c134596b619b91f1f07a084 : SUCCESS in 5m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3c82312b65d548418ee78aee55864dc1 : SUCCESS in 12m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/af5f61153928411bbe946430c2f82ede : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/846eb4bb58374a17ac106b2d7742034a : SUCCESS in 6m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9c155783ad1c4a84a8c2cc952407c001 : SUCCESS in 8m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c31b4f1d74a849ad81f5fe817aa32f02 : SUCCESS in 16m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1fb09ae58124c108449eba1dae747dc : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d2b8afb795bc42f388622bcec5f3c7d1 : SUCCESS in 5m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cc1c2a8b26944a4ebff1382d1406f847 : SUCCESS in 9m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1f0f67e377a148779e1a0d62a83cd976 : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b590dca2d10e4bfcb0b959f5720abf69 : SUCCESS in 9m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/da250234e8244f94a146fca63e0f20a5 : SUCCESS in 3m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2fb55464daa443edb932bae8d766e978 : SUCCESS in 4m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ae7b196929334e42aa9089f95cd144f5 : SUCCESS in 10m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5507e294d66745929b09229862192a0a : SUCCESS in 27m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/887b225d5a494ec9981e8cb68bd20b37 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2415bab1ecf4445a8b8e544029d96ac2 : SUCCESS in 13m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0cc3ef741b414e769cd987c19879cd81 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f257f735811649aea2bd6f12f662f690 : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12cc0e213a554f65bc58a4579a26a0df : SUCCESS in 23m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d0f50639d0a844ffa97eeff90aa2ffa1 : SUCCESS in 4m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0aaba193fb1d4a858b6871429de8c182 : SUCCESS in 3m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4eb7f82bb4d9480db5e3360231341d2b : SUCCESS in 5m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bdd13098299844fb94ec96eee0bb7d17 : SUCCESS in 11m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1b67601c47ff4819b516b7f2c7a7bc85 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d681667f31f74f71b0c3517798996eb5 : SUCCESS in 51m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97e89483f8b140338190c69ee2726ebb : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9a84d2c55793382f1d1ae84df22089b25bd30079","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-01-14 11:00:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c32fd81a806bac3d583908edba50bf873b29962d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-14 11:01:16.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"3d31cc92bffc899cf656c966f3eccd9d6adc87b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-23 10:39:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d46d74da9f7113baf39a2fbd95fdd7249d20aa4c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-23 10:39:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91d290e093601ea577b2bf7ab43bd6c06f7ccd7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 12:44:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976f7b75a5b44d9dacdb11249194ecfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9a7514a99d418e964e77f769566aa7 : SUCCESS in 3m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7b040d2090e9409095fb0ce8f5da50c2 : SUCCESS in 1h 36m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37e359b8563b464f8c4565107e554baa : SUCCESS in 1h 43m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/c3d220a94471432584e7d33874ff2cb2 : SUCCESS in 2h 03m 36s","accounts_in_message":[],"_revision_number":4},{"id":"cfd1e88f0fdc70fbfa10e55f782dcec4435f2277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-12 11:12:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"cffe0413d1fbabab0f7333748ffe9c1778e17bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:51:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e632b13f2e446f78bd971d388d648bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74290ad054854efbb971fb97eb40e957 : SUCCESS in 2m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e4d847807204e118fe91d214c6d51b9 : SUCCESS in 25m 28s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/1f8401d6df424e02b3f3cc1f9a219aad : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5db8107aa6014851981eccde9fbd5bef : SUCCESS in 1h 05m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c4fc31de5884bc1a67d4cfa1474e463 : SUCCESS in 2h 08m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fa463083cdd049b0944c961cb2f84aaf : SUCCESS in 2h 12m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92943ad245a6474fbaa94165bdac518c : SUCCESS in 2h 01m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/8412d6e2460a42f39f84acde168604a6 : SUCCESS in 2h 32m 41s","accounts_in_message":[],"_revision_number":5},{"id":"78765a1b7021cfd7ac03a29346427ddb4023f7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-14 12:14:25.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"47b150b4413c3f2fb1254f3fc239722d62ce16b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 14:54:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57ad56c5084b494981654d85a8296047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a84e4103f954bb6abdeeb71f4f6edbd : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d262ceaf418c455aabff041947773f0a : SUCCESS in 25m 42s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/79421e16dd5144f29c1a224dcc354375 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2229ee495fd498ca598f5e15099af79 : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94eb74d1ffd949fca5f4e6af32b13316 : SUCCESS in 53m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a2e3493991614cbe9c5214e20081c0a5 : SUCCESS in 1h 50m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa2dbd868415415794d2f54273dfab99 : SUCCESS in 1h 47m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/be3e39bd249e4506a75675aef7bd0251 : SUCCESS in 2h 11m 50s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8033da896c38403da6ed9a6e47d55d57 : POST_FAILURE in 1h 48m 06s","accounts_in_message":[],"_revision_number":6},{"id":"2ecfda4f5f68cef98c54c6daee39142f9ad3416d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-16 11:08:50.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"85732e6cf997962b10cb91c63f077ae90221e7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 13:51:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cd6b9d9fc51457fbd785aae62d84f49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f67b9bfbbd74a38a9f7426ec5c5c64d : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/633fb4b2c2e5488bb0107bd63aa67f6b : SUCCESS in 29m 46s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2f0dd209ce0a4b62941acaa021dd0c5b : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae9907839e284a2ca17e0969c4c1b605 : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ec038ef407742bea4ec2d2d310fceff : TIMED_OUT in 2h 35m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/470cc282d94b4a47862b03721d06c195 : SUCCESS in 2h 06m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9be3a49f66d4480783f5496d1fae7ea0 : SUCCESS in 2h 01m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/145246b242a24bc4b7523cadfc3f3dff : SUCCESS in 2h 26m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/91a004a6a2aa49dbbfdc39f27e8a3f78 : SUCCESS in 1h 35m 39s","accounts_in_message":[],"_revision_number":7},{"id":"5f37fe820e6551812e4cbebd16e45d798abe92e9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-01 09:51:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f5e605a14d3b0c3ae0417b2c0ae2f733e684f6cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 08:54:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"35e09671e766d29a835842962cee38eb45aee877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 11:29:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ac0d91afb4642a88b44fe4a9aa40a48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4921d1db6ffa4e6190229568631480a7 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d20ea521a7c544cf8f771d1952aeb78a : SUCCESS in 25m 04s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/f506d777e85f4219bdde69f41ac25ea2 : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a6d0980874f4216ae0d371fcb7e0f9d : SUCCESS in 1h 03m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa588b15a9a7481186f2f16ff297a4cb : SUCCESS in 1h 40m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0d8b3cba4336426091afdda1f8ab1367 : SUCCESS in 1h 10m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/81c76d17ef8c4e89ad52bb7f2d2da34e : SUCCESS in 1h 43m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/06fadc6053c74e04a984cf83d5042344 : SUCCESS in 2h 30m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/08d2b6ffd372421890e15de7479ae93a : SUCCESS in 1h 36m 24s\n- nova-alt-configurations-os-vif https://zuul.opendev.org/t/openstack/build/5adef9f7baa44dfb88809e9b7c81d49d : POST_FAILURE in 1h 06m 35s","accounts_in_message":[],"_revision_number":8},{"id":"5af39c5e317ccc518e4596179956c5ba80f0f143","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-06-03 12:26:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"85fed8ac513d973a410f8a9ac401eb589546d86c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 13:32:46.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"027bc10d8aac6ade3ae9e3568d75b542894a8bbf","revisions":{"7af3b2ccce9a912f922d8ad75cf88eb66c4de88e":{"kind":"REWORK","_number":1,"created":"2026-01-13 09:25:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/973210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/1"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:37.000000000","tz":60},"subject":"Bump pyroute2 to version 0.9.5","message":"Bump pyroute2 to version 0.9.5\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af3b2ccce9a912f922d8ad75cf88eb66c4de88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af3b2ccce9a912f922d8ad75cf88eb66c4de88e"}]},"branch":"refs/heads/master"},"96b9a2b6b046e5c148f243a2e530215d219254b5":{"kind":"REWORK","_number":2,"created":"2026-01-13 09:28:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/973210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/2"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:28:41.000000000","tz":60},"subject":"WIP \u003d\u003d Bump pyroute2 to version 0.9.5","message":"WIP \u003d\u003d Bump pyroute2 to version 0.9.5\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96b9a2b6b046e5c148f243a2e530215d219254b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96b9a2b6b046e5c148f243a2e530215d219254b5"}]},"branch":"refs/heads/master"},"5648c3c063600379d0a37c3bb63fd3ed30260290":{"kind":"REWORK","_number":3,"created":"2026-01-14 07:39:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/973210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/3"}}},"commit":{"parents":[{"commit":"dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83","subject":"Bump diskimage-builder to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dfcad1d7fe24ad7a6f664eef2fe787a8c990fc83"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-14 07:38:16.000000000","tz":60},"subject":"Bump pyroute2 to version 0.9.5","message":"Bump pyroute2 to version 0.9.5\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5648c3c063600379d0a37c3bb63fd3ed30260290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5648c3c063600379d0a37c3bb63fd3ed30260290"}]},"branch":"refs/heads/master"},"f760b6be184ff90b80b9fe12a2a17dc888fb29a9":{"kind":"REWORK","_number":4,"created":"2026-02-23 10:39:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/10/973210/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/4"}}},"commit":{"parents":[{"commit":"d179584aeb59a023f065cbf763204525586f56ef","subject":"Merge \"update constraint for oslo.utils to new release 10.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d179584aeb59a023f065cbf763204525586f56ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-23 10:39:03.000000000","tz":60},"subject":"WIP \u003d\u003d Bump pyroute2 to version 0.9.5","message":"WIP \u003d\u003d Bump pyroute2 to version 0.9.5\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f760b6be184ff90b80b9fe12a2a17dc888fb29a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f760b6be184ff90b80b9fe12a2a17dc888fb29a9"}]},"branch":"refs/heads/master"},"acbc94d51d072cef1de0bdc47aaee38a64c0285a":{"kind":"REWORK","_number":5,"created":"2026-05-12 11:12:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/10/973210/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/5"}}},"commit":{"parents":[{"commit":"23cdc1c30ea90acc80827f8f2684b97d27fb179f","subject":"Drop support for python 3.9 and 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23cdc1c30ea90acc80827f8f2684b97d27fb179f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-12 11:12:01.000000000","tz":120},"subject":"DNM: Bump pyroute2 to version 0.9.6","message":"DNM: Bump pyroute2 to version 0.9.6\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acbc94d51d072cef1de0bdc47aaee38a64c0285a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acbc94d51d072cef1de0bdc47aaee38a64c0285a"}]},"branch":"refs/heads/master"},"316c2ff5622a30081185ebc304a19b72a419a717":{"kind":"REWORK","_number":6,"created":"2026-05-14 12:14:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/10/973210/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/6"}}},"commit":{"parents":[{"commit":"856de3be051e9abb0e342ff2da20fae51b6404ba","subject":"Merge \"update constraint for python-zaqarclient to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856de3be051e9abb0e342ff2da20fae51b6404ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-14 12:13:50.000000000","tz":120},"subject":"DNM: Bump pyroute2 to version 0.9.6","message":"DNM: Bump pyroute2 to version 0.9.6\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/988599\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316c2ff5622a30081185ebc304a19b72a419a717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/316c2ff5622a30081185ebc304a19b72a419a717"}]},"branch":"refs/heads/master"},"1c6ee97eaec53cd1aff769560fffa2dd703cd1ca":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-05-16 11:08:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/10/973210/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/7"}}},"commit":{"parents":[{"commit":"856de3be051e9abb0e342ff2da20fae51b6404ba","subject":"Merge \"update constraint for python-zaqarclient to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856de3be051e9abb0e342ff2da20fae51b6404ba"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-13 09:24:09.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-05-16 11:08:49.000000000","tz":0},"subject":"DNM: Bump pyroute2 to version 0.9.6","message":"DNM: Bump pyroute2 to version 0.9.6\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c6ee97eaec53cd1aff769560fffa2dd703cd1ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c6ee97eaec53cd1aff769560fffa2dd703cd1ca"}]},"branch":"refs/heads/master","description":"Edit commit message"},"027bc10d8aac6ade3ae9e3568d75b542894a8bbf":{"kind":"REWORK","_number":8,"created":"2026-06-03 08:54:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/973210/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/973210/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/973210/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/973210/8"}}},"commit":{"parents":[{"commit":"6f76e6442a25143e43bbd71cea2332f2e61e2187","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f76e6442a25143e43bbd71cea2332f2e61e2187"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-03 08:53:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-03 08:53:35.000000000","tz":120},"subject":"DNM: Bump pyroute2 to version 0.9.6","message":"DNM: Bump pyroute2 to version 0.9.6\n\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/991293\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5acd7fca3719cc8efc201fd8d8e9597772fde634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/027bc10d8aac6ade3ae9e3568d75b542894a8bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/027bc10d8aac6ade3ae9e3568d75b542894a8bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
