)]}'
{"id":"openstack%2Frequirements~973435","triplet_id":"openstack%2Frequirements~master~I918f6351a2c3638f56e00488306f51bd54019430","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I918f6351a2c3638f56e00488306f51bd54019430","subject":"update constraint for etcd3gw to new release 2.5.0","status":"MERGED","created":"2026-01-14 20:10:32.000000000","updated":"2026-01-16 14:21:48.000000000","submitted":"2026-01-16 14:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973435-new-release","meta_rev_id":"1740bd1092ec2622a39a66281285314cf65a8593","_number":973435,"virtual_id_number":973435,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 22:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 12:08:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"83679adc7c9ff8585c2cd4d7d11f07e602531acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-14 20:10:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3160000cf273e3cf6b183d45765bca4df9991994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 22:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e45add905c9475e8a06600484cfcc1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa649be1ffb94eaaae058a1c6e55bd40 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299244c61cb0459c8bd471b0e92d2436 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/846993e786cd40a49576bb6d5265464b : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31cf2279a0874d8cb47545adf20e16d8 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f611d0fce7594e79b1bef2efd9518916 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d70d72760e24c70bc1ebe9dcd20df9f : SUCCESS in 1m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d5c6bbb07b14b3e9d081dff38abf93b : SUCCESS in 5m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/33fffc20c7d1493e911519e63523c833 : SUCCESS in 3m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d4a5e96742594cf0a269851735540d15 : SUCCESS in 11m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f9073642e5bf493b90ef353f751051a9 : SUCCESS in 5m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8550679a27f24ae18c51c6a9e8a094fa : SUCCESS in 4m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/54ab6798c1bc470c8330e45f3787f39c : SUCCESS in 5m 35s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3b6aeb15820e4a168504330934de105d : SUCCESS in 2m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/18507dbfa16a41569b0767f1c02431d2 : SUCCESS in 8m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/52d5838bd826479c8d9409cdcb8a7b96 : SUCCESS in 3m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fd1862be94554c1aa267443299a044c0 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9ac1de78278b4a7cbf480b38b5ef10dd : SUCCESS in 6m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9bd38ba485ab40b2863eafc8673c0929 : SUCCESS in 18m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62c0feca1fbd4c1589ca4636d69485b5 : SUCCESS in 3m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fe882ad1af89482692a67a646e0fad12 : SUCCESS in 8m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8db55e3592294051a4d2a41d278643bd : SUCCESS in 9m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f0fb390740ba46a08ca3bf5f056735e0 : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5eb3fa29804f4d72bb4feecb172f8675 : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7884eedb5f4a48ff9722cb94c2f5c526 : SUCCESS in 6m 00s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f6c3f89657d34571ba732104674659e1 : SUCCESS in 2m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d90af5b4d29b4d83b9f42ef1ee225731 : SUCCESS in 10m 03s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/87e68d8178504459abb25aec4a316f3e : SUCCESS in 29m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/82a253688ac74215b19e3fcee857fc23 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a78893ab6a474f23990c544215136860 : SUCCESS in 15m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2ce9bfe5bb734ac39ea15b86e4e8c2af : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6cd7f9e400c4f3ea1dcef353ee8ef2a : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f63027ee10bb4f63927b3eca242f5c7c : SUCCESS in 29m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9a482385a2ba49cb9ec05f86af358cc8 : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bb5b99cda8cc46cc9fff2037a58df8ca : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7da218756a044d3acc8c1952e90f079 : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fb9f607a6b6c42cf99b023441fa4c2a7 : SUCCESS in 12m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cc390c39955c4f56bd48be14d02d4710 : SUCCESS in 8m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0758a5345e7c4c0d84e6c9d5d74788d0 : SUCCESS in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b01169286744ffcb08adb8164a967d0 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bd426ff7b905bd617c5c30f34f28dd776b7dd0bb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:08:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b78b961ba5e3770daa3ebd7963d7593f937107e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:09:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"124b12fdc221e0d1a3a502ee57bc36a7e26e8d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/719f157d7a5140b0aff7ae24e11a311e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39938b92eb3345ceac8a735530e2079f : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da283f615f8147baa469e4d2a77246db : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a25a9a72cfa4588b73942b3ada4f59b : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd7931a92d1042b0bbb4556b68d5e18b : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/504435b386f84c44b0c19e1f90858c68 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4729130277b40e08c67e52c77aef5aa : SUCCESS in 3m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1a0dab0459e490da0e5039e8d7f7636 : SUCCESS in 5m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7575eed2562845f6a53112b50ea1e9d0 : SUCCESS in 7m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3a831cc8576049bfb29e14d9518657ca : SUCCESS in 4m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/92cde07c060548f1a77dc21fca145242 : SUCCESS in 9m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6484ccd727ff4db18589f54634d006a2 : SUCCESS in 3m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/37e06520c39b40f1bb691989b28bc034 : SUCCESS in 3m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/27cffd29eee144649d8ba6f5ef0e3248 : SUCCESS in 8m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d0be16076b78453ba1a68650e348625e : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07962c1184134f3286a7c1b4971d06d4 : SUCCESS in 6m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8611947763a144febd9e4d0a684af46e : SUCCESS in 8m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/80992bc75f0f4f378bf80a1af7f41485 : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cd08d037e0b6436ead5196b5ad993a99 : SUCCESS in 3m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ee94f088f6f248a49cf09f2bcb7eed85 : SUCCESS in 10m 07s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3ac82fc7c9f74556b558cf10ec7398d3 : SUCCESS in 19m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/22b84026a7ef4dcdbb348f9d76f77884 : SUCCESS in 7m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0a4f67de84a0413788cd1ca73c21d614 : SUCCESS in 11m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/064c1c12cac547fda1419882e931aa1c : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/964706e7e70644e6a31f2db68ac87733 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3d7a4dfc9b24488a5b2b9d61d3ed106 : SUCCESS in 27m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0a9105f84e0840138c979b1b81e15b5c : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9521994f1baf44a3ad3f83ca731bb12b : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94c89035b47f4383b1d8bcf127decc8a : SUCCESS in 6m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/088c08b02db04d47b7cd16337b611f17 : SUCCESS in 13m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1e42f4eb53c6477baabb92fadedf6be5 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c941ae15a41a4049a94c32619f61cc53 : SUCCESS in 1h 01m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f0aaf80501d4db4812f221b7543f24c : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1740bd1092ec2622a39a66281285314cf65a8593","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c071304c0520157dd2a753f8f7ec2123eab590a","revisions":{"2c071304c0520157dd2a753f8f7ec2123eab590a":{"kind":"REWORK","_number":1,"created":"2026-01-14 20:10:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/973435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/973435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/973435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/973435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/973435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/973435/1"}}},"commit":{"parents":[{"commit":"400050263e9f21cbbb034954d4972c8513adee56","subject":"Merge \"Update cotyledon to 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/400050263e9f21cbbb034954d4972c8513adee56"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-14 20:00:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-14 20:00:31.000000000","tz":0},"subject":"update constraint for etcd3gw to new release 2.5.0","message":"update constraint for etcd3gw to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ic250d5d6b98e5918d80314933d5424a62c985aa4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I918f6351a2c3638f56e00488306f51bd54019430\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c071304c0520157dd2a753f8f7ec2123eab590a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c071304c0520157dd2a753f8f7ec2123eab590a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
