)]}'
{"id":"openstack%2Frequirements~973482","triplet_id":"openstack%2Frequirements~master~I493a775fabffbc248c8f89362fdfd4ebe7652c30","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I493a775fabffbc248c8f89362fdfd4ebe7652c30","subject":"update constraint for oslo.privsep to new release 3.9.0","status":"MERGED","created":"2026-01-15 14:05:51.000000000","updated":"2026-01-16 14:21:58.000000000","submitted":"2026-01-16 14:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973482-new-release","meta_rev_id":"5df7a42b0924eae81c8f2fb6661dadff0bf414a9","_number":973482,"virtual_id_number":973482,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-16 12:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-16 12:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 15:28:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 12:09:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"53894af487ffe1f4a566b2857a27c36a6c26fc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-15 14:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bbccaafd0bb529cf06dfec4f9a20143256df829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 15:28:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e099984fa3cd418bb6767f66df83a58e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215840dc220e409082084a9acd67608a : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52128105035044eab0a86b717231ecd7 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9834d444d1b4826a6de91ac5911fea4 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d969f1c073704ea697b9e0e28da43373 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0848dd5d8fd4b0d9b4a8f5f0f5b98ee : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dbc51f09ebe546cea93f28b2f5464251 : SUCCESS in 2m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bb365c894c98444785e78a7e77bf7a24 : SUCCESS in 4m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c5b0f4b4ee7c40e39287b1f2578d5c5e : SUCCESS in 6m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b1eed373bb5f42108130d9b149064a2d : SUCCESS in 5m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8b4e5a4b7f1c4f5f80c8012aa479ad9d : SUCCESS in 11m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e2a0afe5bd7c4e28a60fc9ae3d73e594 : SUCCESS in 4m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/abec6058cc5a4dd984b6dfe2e56bfd5d : SUCCESS in 4m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3cced22226dc488d87503c007a1a1620 : SUCCESS in 4m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/374d046491c442ddb0b32a14a0a47c59 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cadeda0d5316419fa5385de33cc9f31e : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/76cd514e70bf4b60bad19473c4eb48b1 : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/194d7843a48047c5b8825d7b8f23a4ea : SUCCESS in 6m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e7f27876c3134addb3a13db0d2f63efb : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43dffabf7264416e835aa20d1de828a1 : SUCCESS in 6m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a5b9fb07814b4bbe88ff25d78b7329e6 : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7eed62240df74c75a983be93d1bb5b7f : SUCCESS in 12m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7c8d1d99e8004c91a689330dfd83408a : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5d99ade654e94fcdad0cb1d1e403ff5e : SUCCESS in 9m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9aaf32e7159f4a76ad3e298d30571690 : SUCCESS in 4m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cb8284b024fb4e4da32bba7ab52204b3 : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0ebd2b66e96c4ad186a4eed76ec0993d : SUCCESS in 10m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/924dee4888274079b5d2709fa297f111 : SUCCESS in 21m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c762652c5b1c4d869651713dc78d6237 : SUCCESS in 7m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/417d159fd5a44fdcb8784b8799eb24fe : SUCCESS in 10m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/148fcce7a2954b5eabdf62bb7ad98f21 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/076cc7e0211847a8957915781087066d : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/349c20a5d1a54dcdbcc3c9d2db425cb5 : SUCCESS in 24m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d2f71ee90738475a904f92119a455a5b : SUCCESS in 5m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c33ac2db9ac24d85bfda4eba1a334d52 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/343c1e3e0aff489580949fc21d71d03d : SUCCESS in 6m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/591ae5c0c8ce4bb8bcfbed256286c5e9 : SUCCESS in 7m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/63d0113d7d4e4d35ae9bafcc9d2ade1e : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/286f53f822834b33998f33b76555eb62 : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7a15a6a5ede4b7a96a2a3b09299b3e6 : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c94ff577bcd0c33429cf0a30cfedd3c113af9bb1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-16 12:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d235fa32d8c7e5b4a081180704ea0a9a85756420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:10:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e560187a947d4d4064587475dc0d0632cf134a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:21:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffc7dd7458d454a9f5dd287ff13a7ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3147947993d84ead802c89cf88873307 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3472ea04c2840348f0c580aab6eef19 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b5c22c01aed404da2c999e765cbe4e3 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa3abf8ceb1c4c9db77e012e52e8af01 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83f552d529bc4d9f8a2206c161721189 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63bcb11ba6e146e0b58c229f5a94020c : SUCCESS in 3m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c2be60b046a40739d31a9dbd6713132 : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8aa8945464064c3fbbf348cfe6819901 : SUCCESS in 3m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/805630920e9d49a485bfeea9e11b910e : SUCCESS in 6m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3a45227886b44191babca310b0077548 : SUCCESS in 7m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ae33157e4e6e43698748b9ddce430410 : SUCCESS in 5m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ba1530c3e4a04b54a3f7da2b180ffe6a : SUCCESS in 4m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c796d4da69b64e05bdc8e7929dff1e4b : SUCCESS in 7m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1198cf6fc8d745ce989431022a93c8e1 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9d41bab98e448c18659fa67e6797627 : SUCCESS in 5m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7e80b96e66884cde8bd8ec254c36db2a : SUCCESS in 7m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0f91f93139b442183289beee514fe3d : SUCCESS in 9m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8cce1f1fe82c44e5a66d7bbe4f1394bc : SUCCESS in 3m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6aa223cd2f254cd8b733a1af5cfc7934 : SUCCESS in 7m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/185f26df3fa249fd9411aae1aa5d6997 : SUCCESS in 23m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3c85fc76abea4b089a295abb24453502 : SUCCESS in 5m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/39e4b94af78e444aa18771418259e296 : SUCCESS in 8m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1c6bdf0beb7f4ba995d484dd089ab7f8 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a83244d56ce41048ecae3a15a9c348e : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38004fc7b7584386977af9d782a1918c : SUCCESS in 28m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/212de8cfd5b443858642f28399da8bc6 : SUCCESS in 5m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8838e973be3b49869b7a254e70f7c7e8 : SUCCESS in 4m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5667d3b5a544c6b940f8323a23aee68 : SUCCESS in 5m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/74ee806a5f6646998bf7fa00200f18bc : SUCCESS in 14m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/057f778e72cf467caa8e0386feb12694 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33858cfe0204cafa22a7a9c74603bb9 : SUCCESS in 46m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87d78d8ca1034fc9bed90adbc173d8b8 : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5df7a42b0924eae81c8f2fb6661dadff0bf414a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:21:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4bf3b9c839fa2e64e6b2aed7badae2c22622ecd","revisions":{"a4bf3b9c839fa2e64e6b2aed7badae2c22622ecd":{"kind":"REWORK","_number":1,"created":"2026-01-15 14:05:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/973482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/973482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/973482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/973482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/973482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/973482/1"}}},"commit":{"parents":[{"commit":"400050263e9f21cbbb034954d4972c8513adee56","subject":"Merge \"Update cotyledon to 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/400050263e9f21cbbb034954d4972c8513adee56"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-15 13:55:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-15 13:55:49.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.9.0","message":"update constraint for oslo.privsep to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I31dbd41c05f91505e3f93ad9ab3845cf8ea6f39c\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I493a775fabffbc248c8f89362fdfd4ebe7652c30\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4bf3b9c839fa2e64e6b2aed7badae2c22622ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4bf3b9c839fa2e64e6b2aed7badae2c22622ecd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
