)]}'
{"id":"openstack%2Frequirements~973775","triplet_id":"openstack%2Frequirements~master~Id4b72f2c6df917ce2e0db2aef1399846447dd822","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id4b72f2c6df917ce2e0db2aef1399846447dd822","subject":"update constraint for keystonemiddleware to new release 11.0.0","status":"MERGED","created":"2026-01-19 10:01:38.000000000","updated":"2026-01-21 14:50:41.000000000","submitted":"2026-01-21 14:49:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973775-new-release","meta_rev_id":"c0aeef94dded4ab40e46438d5dc58ab3dcb0987b","_number":973775,"virtual_id_number":973775,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 11:09:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:03:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:01:43.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"5c5a1d714d9b7b263d3c3370182ad29b6b9d5ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 10:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c04e7a29cc946be1f20324658966b031ec5abd78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:09:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/482680be41074ee69d7574a7855bd923\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b99c35ba4249f49a8a115e6eb4a73e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8c2a785f6340cf8b66a6454d3a939d : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd06bc13446e45939c9694872b95c522 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d636d82c40f6435f8fc03e8d0f93df1d : RETRY_LIMIT in 1m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff31eae2ad144f778375cff67abc684c : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97fc55c0d464435394a39041273c810a : SUCCESS in 1m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0499b5a4829f43fcb853169c99f40257 : SUCCESS in 4m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/552453a1fa024ff785cfbda77644505f : SUCCESS in 6m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ab63bb422a3746a1a30823b46678e209 : SUCCESS in 7m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/571af80d2ff54e1bb305525b2d04e28d : RETRY_LIMIT in 1m 36s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9e27204b3df84f65a86fd5824cdfff1c : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ee68cc7a98f14029b4529382d1c1a966 : SUCCESS in 5m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/48cb04a27c06412683211435678d2373 : SUCCESS in 2m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/29488e054d75411eaf55e2e1db2a76cb : SUCCESS in 10m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/47db6fb54f0c400a85001fb32e3e5968 : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e6e1af13fb124c5681767871162163f2 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d3d1850f25784d618e4ecb18e6ed04b8 : SUCCESS in 3m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/65d23047c11245b19ad400f31d63cbce : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee41a23299a34665b4ff0b5e185b5c46 : SUCCESS in 3m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/71227d9c109e41038fcbcf56a8d386a2 : SUCCESS in 7m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d55fe48131174b25a547bb88c614402c : SUCCESS in 12m 26s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8f92ca5ac5b44905a78714e1908a3417 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5dfebf12a687458c93f7a1d70cce03e9 : SUCCESS in 9m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d64f1e74e0204539b7b33f16f7148365 : SUCCESS in 3m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/748ddacc8fe842b5b13b417b4900ddf6 : SUCCESS in 2m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/af872a1fb9644104859041c270e8ab4b : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/20408eda2a0d47f78ef72c925f615bfc : SUCCESS in 20m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9cf6dc84934b402182543bdfbb34c95e : SUCCESS in 9m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/882a6fd3a14c4c9285a53b10d6bac0ba : SUCCESS in 14m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fc39fcbb8f1e4ab68f19cce9ad1e9f16 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9eacc24ef86843c9ae6379a8a9a6c740 : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3321a1436bd84b3f957c53f59f7a847d : SUCCESS in 21m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/71a84e1357764157abbb003e741ed764 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/86b4fdce095243a8a30d37d0c8e6bb2f : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f83ce939191d44fa92948a082712901f : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/25404a73d3ed44989ec33b8cc29098a5 : FAILURE in 8m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7fedff0f63e14d298f1bdb4727407aaf : SUCCESS in 6m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12eafab1e11b4d419dd5d01afff2c0f4 : SUCCESS in 52m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5defe6fc8e44cdea9cc2cfb3319da7e : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b9426d32c31654f8d1f6b7c768e62b0395300810","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba19ca84b6e073c76cb7ff769b910178ff2f89d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:55:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42688f2f7f6c4c11b63ab3186455f4de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749bfd1362714f3ea8f851f9c7079b63 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc84f839219429bb8505f18627ee37b : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adc1e8ffa0d640ae8d4518846b003e58 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f671633a1e464e6fbed85816e05f7958 : SUCCESS in 2m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1cf81ee4e0342cca13e04a6bf4b214d : SUCCESS in 2m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0aeeed67d92a46939230671084e4cdf5 : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8f457362e53a4851bbc7e21e84f1512d : SUCCESS in 4m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/808147a751e14f6aab91fc08950726eb : SUCCESS in 11m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1f8ba23eb0614862a051a1bb8b9dce9b : SUCCESS in 3m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0a31a9112d6e495d947ccc12979fffbe : SUCCESS in 6m 15s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4e14333655d54ea6b0d68f4fd1a458ee : SUCCESS in 4m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/66c7db185c3c432bae916691173363a7 : SUCCESS in 5m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/992d7bcbc81b42fcbb4b54d2997ec19e : SUCCESS in 4m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6a218d0e3ed14a439f0f1c0a79f72393 : SUCCESS in 8m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ef39f1350fa440a7887531b6adb792b7 : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ea4864c939234085a8906f24f31af162 : SUCCESS in 5m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fe1d9af62f5547b3b005005ac8cf06b6 : SUCCESS in 4m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c112d416973249f5a0c0d52967f46a02 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7046f3071e77463bb2c940a267428fe6 : SUCCESS in 9m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/33b903a96e3c441088124a67f9733442 : SUCCESS in 9m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7f2cf85705104505ad75eb09dcdfbf7f : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/002c24a6681b4a6d879566c51e7bb3e0 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/51c038b23f4b4a62a1e570c44bf43305 : SUCCESS in 10m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/91bd4ab3435144928973cb113255533d : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/de440073aa014e9aafec6061ee728ac1 : SUCCESS in 2m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4614e8627b344690acad1417d8387836 : SUCCESS in 10m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0f3e1b1816cf41328e51559b00fd540a : SUCCESS in 25m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/151f55e860d24b029f95a042164c6432 : SUCCESS in 9m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/983ac05b2eab4586997811f4db115b2e : SUCCESS in 14m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5fc2c41fbdd04fc3be1f7add59bd9875 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f1383bd8eb8486faaaf1378a03558e5 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af188d28211a4034a790a05be85dd6c5 : SUCCESS in 24m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0f18eadd955e42b4a767d4519b90caca : SUCCESS in 7m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cc80d7a965ad48edbef9898c47f8ad40 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dbee601b14e4b9f99fc96667dbcc0c5 : SUCCESS in 7m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3af51d489fab411780bcfac75e007f7b : SUCCESS in 26m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e38a40b2302840f7b31561a0ae6a8201 : SUCCESS in 4m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aedad5334bde4f4b993a8d92be36a934 : SUCCESS in 53m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e1efdcf8dbc47fcafc1d1efdc7ded23 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"86ccffe66cdd00bd08ccd2027da0d35a1378b36b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:01:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e436ccbbc66bf73334822c6190a214702b495b1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:02:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5640c08f5e5fc3f4a9bce67dc5698e9ac9d0285d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:49:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c767ed24ff4b9ab80426e89d0541f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f60ac5cdfa44080a668f502975ec972 : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d30c95749be94fff9714c18e76ebbdcc : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cecbe0fef9f4af0af68dbf5eeb480ec : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b6a1bf702af4fae98f14ef3de94c864 : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d8ff1e4a52441fd92308e628cc65ed7 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/711d07e0cfa34af6add0ea1ecf98d33f : SUCCESS in 2m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce042349480e4c6c9bff651e382af00a : SUCCESS in 4m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/da3c2ef3c71a4ab38d1ca274a26c2330 : SUCCESS in 6m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/59a584def50e493cbf96cea7f0c31494 : SUCCESS in 6m 09s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9815758ef7f04c478da8fa60e5709dc7 : SUCCESS in 9m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5a9d503cae3740ea8e0165ae3957c2ea : SUCCESS in 4m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eb3a07b63eb24e6fb0a1a3a977d30d0b : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d3de91f0f3f14821a621bddd699515a4 : SUCCESS in 5m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fe8c50cc7c014ea39a60edc9fdc8f0e7 : SUCCESS in 15m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8883e231a5b4ddab0b7675552ad6d31 : SUCCESS in 10m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4d54e3eb7e0145c2b6b4388732affe21 : SUCCESS in 10m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/69dec0f18d9544f5a9fc89d59d121b72 : SUCCESS in 7m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f11d97a993dc487d9c91bedd03e94e89 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2766668c8e39458eb84377fb80dcf458 : SUCCESS in 13m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fd2d76e6366e4fe6a82d06b1011e204c : SUCCESS in 30m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b7d388de98e4a24b7323bd91a7f6c17 : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/394d1ea891ff4cb28c2830854a43fe8c : SUCCESS in 9m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/777bdc7ca6a84808a9281f37685f74db : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a48506f01115410384289f911c0477a6 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8866e578f194487c94d99a0f2b46094c : SUCCESS in 21m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fd342c6495f7493091079bd876e07fba : SUCCESS in 8m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6af9e2c4ca4a4e109d9c02469bd4e24d : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bc70deb0cef40e190308b28b75228a9 : SUCCESS in 6m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5226baa27bea407581d9590f68822953 : SUCCESS in 19m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/629967ff445b4f10ac7884bbea37e978 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81d9e12fc6644d0f8181796c4c4aa1ad : SUCCESS in 55m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d6f38fb1beb46f68b4c71821140aeda : SUCCESS in 1h 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"35c58f1561177b64c728f035df61aa6accba46b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:49:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0aeef94dded4ab40e46438d5dc58ab3dcb0987b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:50:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dbb33ce38445a5a2a25454d3617f35\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3412337d31e48689b504d202d41ecd7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5d3ae2ff18516336824c824a540a3430151ca06","revisions":{"f5d3ae2ff18516336824c824a540a3430151ca06":{"kind":"REWORK","_number":1,"created":"2026-01-19 10:01:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/973775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/973775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/973775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/973775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/973775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/973775/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 09:51:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 09:51:38.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 11.0.0","message":"update constraint for keystonemiddleware to new release 11.0.0\n\nmeta: version: 11.0.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I997968beb272ca2459c9a35bd7baf743d9a32fef\nmeta: release:Code-Review+1: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: Id4b72f2c6df917ce2e0db2aef1399846447dd822\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5d3ae2ff18516336824c824a540a3430151ca06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5d3ae2ff18516336824c824a540a3430151ca06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
