)]}'
{"id":"openstack%2Frequirements~973788","triplet_id":"openstack%2Frequirements~master~I3fed20dfa7943555231e995e7956c0aea8b81f47","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3fed20dfa7943555231e995e7956c0aea8b81f47","subject":"update constraint for oslo.reports to new release 3.7.0","status":"MERGED","created":"2026-01-19 10:56:54.000000000","updated":"2026-01-21 15:08:11.000000000","submitted":"2026-01-21 15:05:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973788-new-release","meta_rev_id":"e84155becb0240bd9e66a70a2dc0db8061fe8f34","_number":973788,"virtual_id_number":973788,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 15:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 13:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 13:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 11:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:03:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 13:24:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2e89d45b75cc682fc96271642fd84ce6bcdfc5f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 10:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f2606c1da1d0fe2b6c5c8a96a42d665087ec21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 11:57:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a481affbf21e44d6884981eb1dfad249\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c4c19c70454dcb8ead7bf86ff1a7c5 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449c8472f41148e39a405d125e888b1a : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9ac4de46bb24f8aaa26fab7f19dc288 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37a156e48eb04eeca99b2eb7e8363dae : RETRY_LIMIT in 1m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa5b00a678bb4f30a288a82b37d37802 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc85cbdd9a104378a9a431a4a8aacd0a : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3bad34c0b486421cb14292d18a0167af : SUCCESS in 3m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/278c36a6dcf04a84bfcfc02a9f2aae4f : SUCCESS in 5m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/93e5a497d81e4ded98b7fca5199cb593 : SUCCESS in 10m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/46ed93ec2b024520aa353fb3b5fd9a62 : RETRY_LIMIT in 1m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d8ac1119dd394d39baf6f9a6d1624e3a : SUCCESS in 4m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/98e06329c5964b47920c70a062e02d61 : SUCCESS in 4m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/76bebfb3eef6496cae6f254a5f101a5b : SUCCESS in 7m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3386483e033545b689250d6559e1e1e2 : SUCCESS in 12m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/30390077d5e44cf885ab43d013666303 : SUCCESS in 7m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e7d110eacb0a4586b37861feb43d7110 : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/316beb71c1f44501a4cd4be912b64d68 : SUCCESS in 7m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3846868b6a57419a912fd3eb09decba2 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35c8743d8e244062b3e9aba1acc6b068 : SUCCESS in 6m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dcb024b1e3d246f3ba11bbad57a372b2 : SUCCESS in 10m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc45de8cf85b4c729427532a9aa585a9 : SUCCESS in 13m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a3887102e96d4a80b7abe30bc4838dd0 : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/554d824c0f244400bc99b2e0d967cbfc : SUCCESS in 5m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c5013d7df43c4a37b15dba978afc44b7 : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6b5b661f054143c59bc95eaa8ae6dda1 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9f96b4f83fff4017baca2e92ff44f07e : SUCCESS in 10m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b3e49c52a3ac429d9d00bbea7560658a : SUCCESS in 26m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3537518a1c7f4b22b468407d7125362d : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/25ddc3a800df4131affc1560e0b6d49d : SUCCESS in 12m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/df324c04e8f84df1bfca1b2bbd3e369d : SUCCESS in 3m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f2f52a7f31f44d45bcb1245280b3cfee : SUCCESS in 6m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0f3ac3f72074a2592b71eec94d13a74 : SUCCESS in 29m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b31ecbff96c84575a40e1fac1833a10a : SUCCESS in 4m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/885d5c953f574f9eb8180d357fc0cd69 : SUCCESS in 7m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55537a313fbf40ce9b0ea71854c881a1 : SUCCESS in 5m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c0c6405f4143485b917ae5c73bf44ddf : FAILURE in 20m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3b8cf13e318849bc891069c937bd704c : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5b68ec401114601b8845492260cc8f3 : SUCCESS in 50m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6407aa20df64ba791bb23d60a0393ba : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":1},{"id":"367a8ed439a635fa1ac6d54bd6c3eaf28a564c2d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c131a63b77bd43a17f4ab86e25f83490c38d0409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655aa2af33aa46e3a4d91043316596e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6013e2cda34a4e9e0d132adc59b824 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fe9b2afc6843c3b36a61ec51803f51 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e9c6fe9dc574850b1e001dc74fdce31 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac9fa06196a74bfba587fc1a4d92f4e9 : SUCCESS in 2m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81c793a741fa43598b40cae1e10c4c96 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce776600fc654f568c92f6e64ba42cbd : SUCCESS in 3m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f546b559bb2341509532df5fabead99b : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b7435ff13e4340adb932edc49a714194 : SUCCESS in 3m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8522e3351182432b8d69c365ea9adae5 : SUCCESS in 7m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/850731bcdc43416bb0443b0a2d5d5c42 : SUCCESS in 11m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/60242da904d4400dba8cbb36bf63afde : SUCCESS in 3m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6ddb50e46c0c4231b05bddaf53bb2631 : SUCCESS in 3m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d3a79ac4b23b42758184764a7bb46339 : SUCCESS in 5m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5cb05f07613047a99407ccf25618c9f5 : SUCCESS in 7m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7892bb61e7834c24b2167258eb7979fc : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ecf9cfe338e14024be30deb1e6be44ec : SUCCESS in 6m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2383e98705a9471794c3db32310c3c31 : SUCCESS in 7m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f4406026d18843f6b51e1cab732fa20f : SUCCESS in 15m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a69168f1835f49b3a89ef28ba5b5bd56 : SUCCESS in 4m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ce3e116714944955bd77bca4ca3f0a95 : SUCCESS in 7m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/33507595b51e444a85dec78341c641b0 : SUCCESS in 8m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/679eabd785a642699d70c185b194c07f : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8a7b79c64b2b46d9bcc00cc62df67399 : SUCCESS in 7m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9e6883f7c0844e3e92c2358365e6ebf8 : SUCCESS in 5m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/baf0a8e4e8c34dcbaf7e2471f98e2994 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e1ffd4e8df6a41c7b585b8ed02ff3447 : SUCCESS in 11m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6a3a6fd36a4340bd80f98147efc3ac59 : SUCCESS in 27m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eb95bdd025954a328351252295a7b269 : SUCCESS in 7m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fe1c99b8d3824da4b72b6003a66adb91 : SUCCESS in 12m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f45737029e5949bcbdcf412417e44598 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9500e76a767452394fff6e695c7de88 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee0fd1b87f704f48b431c676afe778bb : SUCCESS in 27m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5dff5983a8224ade97bc996d174281f6 : SUCCESS in 7m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0a8d7845a2674ac3b794e8328edf15b2 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ab7d8e5d7bc4e9898cad354d936a3aa : SUCCESS in 7m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ea18d5e772aa4385b09d6831bd6076ed : SUCCESS in 18m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/376c735b1fd44d43a5608375866b22bb : SUCCESS in 6m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9156f79da32245c8aff9ef0bc0c567af : SUCCESS in 1h 00m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99dda566d0b74395b7a6ed34ca0ef522 : SUCCESS in 2h 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"385d952b2e10ad7d023ecb5cfd523099eff62dea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 13:24:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9089f9df336b9d9ca73fc634992a01663d94ff52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 13:25:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c680490963bca207ccb6dce8ac653d7ef0cefd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:05:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1267dae3eaea45bf93e4b967484f6de9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae9c76b369a44faa5c1a19de02e9bd5 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0324ce9fcf294193815b419e9e0d2766 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be513d4e3c8b4707af42012dd88afd1b : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/634cef34eeb34ceda2483091654f6b2e : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfd89e04dada4ddcae4aa392cc1ac26e : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80b05d2e015d4e6cb897948b2ac9e374 : SUCCESS in 2m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/78eda128b3374311b4050ce53400e435 : SUCCESS in 6m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/749755347fea451f90ba5e1df0a0d9da : SUCCESS in 6m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8f13c1d886a245b5829c423ba7fe7b2e : SUCCESS in 3m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/99c916641ff3422c8a6e9e64e07d3341 : SUCCESS in 10m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2362ae372a8e4616813c7062b22f3ab7 : SUCCESS in 3m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d62a7fe205a64910862e84e8b8a4974e : SUCCESS in 5m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1c4b68b6ebd9458982cfb6c8a6fb4b72 : SUCCESS in 5m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d6407358ce8f4f159edfd666d4e64c1d : SUCCESS in 17m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3181dde5c38a42ef89f58c558aa240f2 : SUCCESS in 7m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/19d253b873614c91bb1a026e0003b9b7 : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/847cf6f57e3f4a05ab6aa4bf401083bf : SUCCESS in 13m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/92ab30008a63477ab033afb198f0d1ef : SUCCESS in 2m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac0134105034487694d5a58a81c77759 : SUCCESS in 10m 51s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/03e62d0a679a4e26920cf2d6fd33d0f9 : SUCCESS in 19m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/60c19d8137084449b63940b6990f0142 : SUCCESS in 8m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/86609ba153bc4f85895b76ee60d021d8 : SUCCESS in 11m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/71e637991bec404999a73208d5eba6bf : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d82bedcce0b444fd938a211c2deca903 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38b036aed01a48c3aa3b024463029370 : SUCCESS in 24m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/98e2113aaad946878b021bdff05a677d : SUCCESS in 6m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/36fd63102144430f8b7eaa0b8e94b72f : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ade39fee83a9417e95879eab31d711d5 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/68e47247157b4c819ec45c164fe2e2c1 : SUCCESS in 20m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/088af795d8754fafb93662918c5de503 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c74c4444d4a4715958a8a56b0788d26 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/679a4bd70abd409a815f9f15f4e31cc3 : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f273955386eed12787913e73eb93fca4f5445bca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:05:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e84155becb0240bd9e66a70a2dc0db8061fe8f34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:08:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb17c2122df4664b96ead87f5702f5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03cb5467144d4e83b0e285854b7f74aa : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68d8adef9cb3d747f32057c4b9b2072c84f7417f","revisions":{"68d8adef9cb3d747f32057c4b9b2072c84f7417f":{"kind":"REWORK","_number":1,"created":"2026-01-19 10:56:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/973788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/973788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/973788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/973788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/973788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/973788/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:46:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:46:52.000000000","tz":0},"subject":"update constraint for oslo.reports to new release 3.7.0","message":"update constraint for oslo.reports to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Icadfa627f58f6efba8ef46c6e4be49eda77868e4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3fed20dfa7943555231e995e7956c0aea8b81f47\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68d8adef9cb3d747f32057c4b9b2072c84f7417f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68d8adef9cb3d747f32057c4b9b2072c84f7417f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
