)]}'
{"id":"openstack%2Frequirements~973790","triplet_id":"openstack%2Frequirements~master~Ib6df746a7f38ca2aa503a3679e1cf4b632458417","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib6df746a7f38ca2aa503a3679e1cf4b632458417","subject":"update constraint for oslo.upgradecheck to new release 2.7.0","status":"MERGED","created":"2026-01-19 10:59:48.000000000","updated":"2026-01-21 14:53:35.000000000","submitted":"2026-01-21 14:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973790-new-release","meta_rev_id":"8441e08ed35ec1711626a57b6081e12e9219244f","_number":973790,"virtual_id_number":973790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:04:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 13:09:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:05:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:04:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"781c2913d1a4aafd270e60b15c22d8e20e27fa99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 10:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbb13979bda591c6dd05c420047817375c8aed41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:09:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b732327778694fe1a11a0eb0ba4fc406\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72491f9d93ef4625b497cdac8e96b49f : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3d124d69af4986916d765454a0f8e7 : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4b834c96cea427fa24b60401fcc0025 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73fbe5952b694ef9ab52ce4f6ec56531 : RETRY_LIMIT in 1m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6dbdb57e73743e799b56771d130d36b : SUCCESS in 2m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84744fa60467478da30a130a01052070 : SUCCESS in 2m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/07b496396e73416690cbb92149758cd4 : SUCCESS in 3m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/efede19f76e74440ac1dc4778210dd79 : SUCCESS in 4m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2af420653c7341cfba5239d75700fde9 : SUCCESS in 4m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a01cc9bc3d984043a9926852b6b6c531 : RETRY_LIMIT in 1m 56s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5df65543b9af4567936a97c5195abad4 : SUCCESS in 6m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dbf67d9b36394b8c92704e2c67461ceb : SUCCESS in 6m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eb97db750c824d7ab078ff9b9d601f99 : SUCCESS in 6m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e0383ba9184a4a9d948fc3f94518775a : SUCCESS in 12m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b4bf6015c9174398a6f03fd693529612 : SUCCESS in 6m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/20a006ed62fb4dd5a13a551216d51eb4 : SUCCESS in 3m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7d55ec4b65484f7c8cab7415fe9bd14c : SUCCESS in 8m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4b6e23d11b204401b05565191d065426 : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b12963664aa4e0e8cb556e9a24ec943 : SUCCESS in 5m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c4c1addc5db04f22b17cc569c938615d : SUCCESS in 9m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dbe1e931cc2f45008a7490e2726c772f : SUCCESS in 11m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/37461cbf79ea4397ae9a7291e1aa4a93 : SUCCESS in 3m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/86739095c8d644d685deaf6b81eb6112 : SUCCESS in 9m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/806b47d297854c16ae57d2c5c803092b : SUCCESS in 5m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3f61f6c45a4e409daab436e8e9e88e8c : POST_FAILURE in 3m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/31dfa73f656e424c9c3f12f2b425ddce : SUCCESS in 11m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/acbd6bbda91743e48f290d97c896d31e : SUCCESS in 23m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a717f00bef6450ea23348c62fa92ae2 : SUCCESS in 5m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/00c5a1b1ea9641be83c02aa570a32e02 : SUCCESS in 15m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0c8aa927fdb0478ca5830d4bb5df3b93 : SUCCESS in 3m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bcaa4e2e4724cd48cff6642500d131b : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/897eac53e7f74919a5b4bcc6f4ce1a42 : SUCCESS in 25m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bb11519b5ee74f41aa16b83cae11cb68 : SUCCESS in 4m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ddde33c52e084919bf7e160fab56a196 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec7f40f210fc44bf8da0d190c043d2a3 : SUCCESS in 3m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b74bf1fe0deb42d58f6e65148950cf8e : FAILURE in 8m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d5244f4b9eb64f78b45d72f9fb80aebb : SUCCESS in 5m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc8160c76fc4db4883bb1b2fef321e9 : SUCCESS in 1h 05m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1801b5620f384f9dbbb950f559ef510c : SUCCESS in 2h 04m 10s","accounts_in_message":[],"_revision_number":1},{"id":"629042d4275a2e1797ef27190e55599025823f40","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:05:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b73d492bc058170a957317c784f9e3b9d79ae6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:56:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cfe64612734522b902fd3abeb98420\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4817b90f6e94557b2c53a4c63445747 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2766fbdc539642c5a53b1b0bbc1a8557 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a7e463af98943b997606b5ac9b7d6a4 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6ff71ccd510488087c4e290f3a5670f : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7107bc7b65c94a1f8c8471a90a61c4ea : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4fd4ead09de41e5a981f55f130dd6d5 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2f2dfc1b22a74e9d9752c54c9d1b2121 : SUCCESS in 3m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d0d22a50fb2542939704554e2e682cf4 : SUCCESS in 7m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1e72632fa867467299487a13740cd715 : SUCCESS in 4m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0022185e3b4d4809812d893962861ff1 : SUCCESS in 6m 39s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f4d2a2d73dd544bda4487743385a81b0 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5b1a7cb9cbd74e5ab7bae6c139059184 : SUCCESS in 7m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6e6f90c934234da1b75477c266d37495 : SUCCESS in 4m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cb8781d4589547ba8153f80e4b63bae1 : SUCCESS in 8m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ae981f15078b427fa549848779dbd849 : SUCCESS in 4m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f5a3965bdb5b49648e107efe810229c9 : SUCCESS in 7m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9e3251d128384ebb8db3236658faaa3d : SUCCESS in 4m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/df22a09d8e8c464cad4c1f82db264297 : SUCCESS in 17m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6361b6d6a0664216bd4d028a55f1aaa4 : SUCCESS in 7m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9f7fc42a18c24036ab342d2156e85267 : SUCCESS in 8m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5d0da8c989824f35960666aac7af2bd9 : SUCCESS in 9m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3da2063757264e10b229f1b0dbfd158c : SUCCESS in 2m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0ab5ab2807144afe95bc9739709b7a18 : SUCCESS in 11m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/22c2d71cfe39444eb43dca17875748e8 : SUCCESS in 4m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b3529bbb8fd04f0fabca93807e0398a1 : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8bbdca28b6674700ba0654e6e532cb4c : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/59ccae1e7aac49999344952ff00ce7bc : SUCCESS in 21m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9e5c99d9f7e3405284bfe0f2698adf93 : SUCCESS in 8m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/67d81a85f99f4f9fa4ae63ece9a700b8 : SUCCESS in 10m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b34fd76526ab4746a5e179a2555b4c39 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f78d16a075a483fb360ff27039573e7 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebaf33333ee14c639d1f2281eecbcca5 : SUCCESS in 24m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b27e41799221437a9c3f7aee4360e70d : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a9f429a4c4a745efa4916287911d285e : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efd1b6984fab45f8859af0e7491850c9 : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a07d6d1395d4463aad23f8b461ebae86 : SUCCESS in 19m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6cf891dcd0c843809ec0482f00b58dde : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f39ca440392f46568ceecc5634e79f27 : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8d2b7c7eb964b87962d8bf5faa22a29 : SUCCESS in 1h 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dd240660acce42d949aa08ed4361023e1c5b7fbf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:04:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf50b4b87d84149ccf7dd105b0865f6eb3edb936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:05:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8082b64530580d0d882c22bf14b9c40b968cf2b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1b5eac8aec14b83a7610a7d1f490eba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7da2d428444711b72cfc0a6f246bcb : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92f93a8025447648c359506705fbf9d : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44d6ef3205714a109d13c4cfc1163eaa : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ffa7c84f91c4e0baea401312f3ae5a1 : SUCCESS in 2m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f5691b13fd948099a495fb8f4470fed : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ceab568ddf1647e7b8b0fe00d9f9fe4f : SUCCESS in 2m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9e58a16194d44302b6cda63fc7e0122a : SUCCESS in 4m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/303072f6a33243fb931228216418c743 : SUCCESS in 5m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f1fda4cdd6794fe1bc4db83662aeea85 : SUCCESS in 3m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b136507e5d97483d9686df8b614709ad : SUCCESS in 11m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b05ed308538d4a81b79bc8a9dfeef1c3 : SUCCESS in 5m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/655b37ed3d204025abb3abb661e4ce95 : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/983719c7b3ef4740932a503142edc91f : SUCCESS in 8m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/98b386ace5fe449988ca1161034b2579 : SUCCESS in 8m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4dd5e0ef26c84ddf8ee571e36d6a187e : SUCCESS in 6m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3268e536f9a24efe87353f23139d51be : SUCCESS in 7m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a265cab1c2ce490a9703aeab7123f90a : SUCCESS in 9m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aa00e68c2fdf411b8eddebbc2327f71a : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fc85733045a94c1a93ac654e44f2fe8a : SUCCESS in 9m 54s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ab1d20f8d86a4b58bb7b328bcf665298 : SUCCESS in 29m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/116d655876a24b30a8614a254f336c93 : SUCCESS in 4m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1241e98a748f400d91df66c2c22ad1a5 : SUCCESS in 14m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a4acaa6fb87f40ed80e20dd832649d34 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ddeb6756d9849af893990f531fd9126 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f664b9d53c55423499a14915457a2bff : SUCCESS in 21m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1b1f0a45133a450ca728d65d349e63d4 : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2f18db73872a424988a19a448f0c4b35 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e3bc65c9bd13408e8ef7fd1225b90846 : SUCCESS in 5m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c0c500b58b9c47f6b5ca80d89b6ea074 : SUCCESS in 12m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aa2b72a3cc7e4753860c6b811c1fb534 : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/227ba617140143c7be990030d38408cc : SUCCESS in 55m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ecf26e445fa4a0d949ce1b2dc35ee7f : SUCCESS in 1h 35m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8441e08ed35ec1711626a57b6081e12e9219244f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de4185b322eb82db355632955730927c1d7a500f","revisions":{"de4185b322eb82db355632955730927c1d7a500f":{"kind":"REWORK","_number":1,"created":"2026-01-19 10:59:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/973790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/973790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/973790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/973790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/973790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/973790/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:49:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:49:47.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.7.0","message":"update constraint for oslo.upgradecheck to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1c808b036a8d213f48a67cc3daf3e1638f7aab6d\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib6df746a7f38ca2aa503a3679e1cf4b632458417\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de4185b322eb82db355632955730927c1d7a500f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de4185b322eb82db355632955730927c1d7a500f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
