)]}'
{"id":"openstack%2Frequirements~973791","triplet_id":"openstack%2Frequirements~master~I3756463ea74340b690627a9ac4fec3668e2b8685","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3756463ea74340b690627a9ac4fec3668e2b8685","subject":"update constraint for oslo.vmware to new release 4.8.0","status":"MERGED","created":"2026-01-19 11:00:15.000000000","updated":"2026-01-21 14:55:05.000000000","submitted":"2026-01-21 14:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973791-new-release","meta_rev_id":"6c7c2b066c14f05c49ce80196a350d51f61c734f","_number":973791,"virtual_id_number":973791,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:04:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:04:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 13:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:05:43.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:04:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e854b9dacf8606a9184a497f71dc45b9682f2706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:00:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c419d3e841209a304b930f72516f59ca8fe0f727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:16:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4755e1252414910b7ce0c7c89f09192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c6a583ef4f4922a7306038e88da799 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd69e959f2fc4804a6c03e4910fac23b : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8664920a948247738cf6d75a8eaedb1c : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0880f627656f41ad99957d2c9ca5c295 : RETRY_LIMIT in 1m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd0d0522aea347b4b3aaa89752cccd39 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8623bcfbbbae4300bb1af936db15b18f : SUCCESS in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3e6f3a8779ed4eeb9d1da44e8489119f : SUCCESS in 4m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c516a9dd5fc44e989b0937116abbbbb3 : SUCCESS in 7m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b531c94dfd9d435595f9126d76f80858 : SUCCESS in 7m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/472a7d954c4746fab30ac695ea47f13c : RETRY_LIMIT in 1m 50s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c74f6315921240f5899bf328b528a667 : FAILURE in 5m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2c7b5086d4e24dcba74ab0aaaaa51a38 : SUCCESS in 2m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8c1d2094ad1a4e51996cb8ae89da08c8 : SUCCESS in 4m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0189580f839143d99193ec95d27c008b : SUCCESS in 9m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0b02915fc55949acbe20df1fb4e04862 : SUCCESS in 5m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a061aabb388f409bb8a52b954ff1e830 : SUCCESS in 3m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3ee65d1ec13d454d84022b051ea74d8a : SUCCESS in 7m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a0d720c38aea4d5abf17004734fac1db : SUCCESS in 17m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63c1f8bec4704222b70e95a0cfa00135 : SUCCESS in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c6a649ab746b405fbdec4693e78f7fb1 : SUCCESS in 9m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d2eeecda4e47467584ab5c672a96324f : SUCCESS in 7m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a0e078e52dc24c9484a52bc70fb3201f : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/58d1ff0a76094462a4a947d5731bdac9 : SUCCESS in 7m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8e0204db4d224a198a1d9d789fc26c0a : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/aafa8a8bf6764e589e5236f0ce111437 : SUCCESS in 4m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9689637c07b348e68a22678c0a09a730 : SUCCESS in 10m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b2f940616d7f42cabde622e4e5315833 : SUCCESS in 29m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f6d619a2f9c4355a1b26c092bc1a280 : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7a8e5b46c3984fbeb408c756bcf5ef93 : SUCCESS in 8m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/510aef379d264905b9f2d548cab44f42 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b18de0518e14cebbb0f668a35213c4d : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fd3aec3e24d445e93abb991a02ca344 : SUCCESS in 24m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/72ffc9a798d847458647b58c59fc6a77 : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/14875714e9c04a0984265517721bd2aa : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3b5af01cfe04586a0b0f61f9c2891de : SUCCESS in 6m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/89813c701f53422ca2507ea4513eab5d : FAILURE in 13m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/224adf7918f64c179f229d80244bbba0 : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4c484b744674586b043a66083ab6382 : SUCCESS in 1h 01m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f589e7e7d0c8440d880b1f36c1cfc80c : SUCCESS in 2h 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dd68b3f4ab2a503c760110cd133dfe99a9ce4ed3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:05:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1c141bf9f772557e295497f11e7d871b2e5c11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2aed89f1b44671be5a2d864405eb53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8deedf1e2245778e073de2845580e6 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d5a8be8a2a45d19fb8ed074b1b32e4 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2639c05763474766866cf31626d00bbf : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7669ad47055343dfb044ffadb6e2c967 : SUCCESS in 2m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc37bfdef71142b5bf472517e4d97cee : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8a25bc72b184ea48d3f76d3b14bb7ff : SUCCESS in 2m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9ab07fd4905744028dc6c3a7625bb7cd : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5323ff7f39e24321b013b2942e87da7b : SUCCESS in 6m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ef0833aeb62f48d182991da791d8c435 : SUCCESS in 7m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/efbe26bab3df42f988ed14e379905dc9 : SUCCESS in 6m 10s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e7d5b7852c3948e0afb96a8e46919e44 : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8fa3479972464aed8c8de572cca72a77 : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/28620d9337654996a6b6e02abcc8df4d : SUCCESS in 6m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6036c96922764f5db66f2ae30009feb2 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1719856c015146b8a128d8966661a27e : SUCCESS in 6m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d800225215c47b09da94988b830c28c : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/23c7ba686bb14ced908b9674acdc5376 : SUCCESS in 4m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5946e352cde745789a0923f27005b960 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52af3d05f1d847c9a33fcbea49a82be2 : SUCCESS in 5m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5a84adccc8dc4bd28f7c19ab1102e99e : SUCCESS in 10m 01s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cf40508ae20e4557bd7cb3da9fbc290d : SUCCESS in 9m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/50798962b3424678a35f4019d53a0e91 : SUCCESS in 2m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d7e5dff534ea4740b9fa8ebb4b0a6ce0 : SUCCESS in 8m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6fd3fe8c56cd4983818bde9e81ab5893 : SUCCESS in 4m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0569d63a1d734056b01bade7dfdff718 : SUCCESS in 5m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5fd0452ddf284ce49fa32f61ea846225 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/dd60e5083fb54d4388744b7e7823ea29 : SUCCESS in 24m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0e970720da94560b69d168794ef9af2 : SUCCESS in 8m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/05e0d6054b9f4c64aa69056313b3d64e : SUCCESS in 13m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b2b8afb61d6048ebb39146a2701c853f : SUCCESS in 7m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/370114058fbf4fb9945dba1a00eb84dd : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2c2a21dd9f54ceca90809d4c3744b03 : SUCCESS in 27m 07s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/00644ea7a1a647d783ae64e20732939d : SUCCESS in 3m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d4ecf4bf8ee647bda3960cb054f50750 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45d81cf0b6d14a64b27375863810ad82 : SUCCESS in 5m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/13a19a4d47d94a1cb4bead9514182497 : SUCCESS in 8m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/07bc57461c8d45dc83d0cb419d0c86f1 : SUCCESS in 6m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65ee72145457405ba15d50da49e57edb : SUCCESS in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2e991d17f23446bbf260a3f228e3787 : SUCCESS in 1h 48m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d6d64543f0eeb21b4bcffc821edc1f1601b49eeb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:04:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0753aa32de951fa75b9b3928d3c510376f10593b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:05:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68eb3860b06f825209361a79bcdcf818345b10e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66096b3f9c6b4f01b9e88e7372ee46cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837c1aae6b254dad8d835acb299bf36e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c983c236b14f50bde9010baf31a155 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c7dff70826c49d2a8e4985845a867a2 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7553bee9255f4e7893f042583f30135e : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d6b8e3ddbfa4183af668bf21198e50a : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f6c28df4dbb4adf97d5e9e73201672d : SUCCESS in 4m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/98a164b33e0746359ce3e27f70ea79ee : SUCCESS in 4m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/21ed0ed9e95344ec8ab005da02e4b9cd : SUCCESS in 11m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f07e7ded327e418b96b71a28c6fd246e : SUCCESS in 6m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b287ad915d2c4b199100a5d37ffbb07b : SUCCESS in 11m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b4634e692313470ca892554b55a7fbf5 : SUCCESS in 5m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90b3188486b54bc9942fd9b73cd611fb : SUCCESS in 4m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/85a6c297bbc24d5ebb8314013133f8cb : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8795723b3a834104a5815a14ed477816 : SUCCESS in 15m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5bc49e13a9e4d5e8b68a1a1cd575244 : SUCCESS in 8m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9b8a0e06e82d47bcb14f272ead68458a : SUCCESS in 8m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc41ec908af04995893da8be5532c9f5 : SUCCESS in 12m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/75123b9309cd47bd9832583ccdaf3b9d : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ed627dc743e443d8b4e2608942b2958 : SUCCESS in 10m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b296f4355ae748ac8439b2de619dde6d : SUCCESS in 26m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6c937ea00f674d1b8643201abfffd96e : SUCCESS in 8m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3bd421ec632c4049b153aa67c0948af7 : SUCCESS in 12m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/312f246eac6147399e38471a7d072681 : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32a903f240a34b4798804356f3dae0a5 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52f442b0d31a4cdda078f4264ccebd5c : SUCCESS in 27m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/334359b64fe542809fd93b1d02f43d04 : SUCCESS in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4f68599c7c0b4bbabb75a51c5917e999 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d29e88ed8e74e3989df45b4d6c28818 : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/150c302636f54a0396bafd9687452179 : SUCCESS in 13m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/33874c69924047c2b5dc5db0cd0c006e : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ebd40105e3144c0861b53b956efdc7c : SUCCESS in 1h 07m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/950cd81e7b1d498398b3e4a693ad7ab4 : SUCCESS in 1h 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fa868f1ada11e7711d85ff0297ebcefccb14c280","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c7c2b066c14f05c49ce80196a350d51f61c734f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:55:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83ab86d11ea473bbe39cfc6de72b9dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78ccc2adddd4b18817945e52a2b2679 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4db8b437a382c04315ceee29eb1f9b30c49fe75c","revisions":{"4db8b437a382c04315ceee29eb1f9b30c49fe75c":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:00:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/973791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/973791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/973791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/973791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/973791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/973791/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:50:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:50:14.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 4.8.0","message":"update constraint for oslo.vmware to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I28fb101442298043694392b3a984d8cf61f36336\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3756463ea74340b690627a9ac4fec3668e2b8685\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4db8b437a382c04315ceee29eb1f9b30c49fe75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4db8b437a382c04315ceee29eb1f9b30c49fe75c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
