)]}'
{"id":"openstack%2Frequirements~973792","triplet_id":"openstack%2Frequirements~master~I2f17a9664b5cb632c5082f0c944dc593825d792d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2f17a9664b5cb632c5082f0c944dc593825d792d","subject":"update constraint for openstacksdk to new release 4.9.0","status":"MERGED","created":"2026-01-19 11:02:18.000000000","updated":"2026-01-21 14:54:15.000000000","submitted":"2026-01-21 14:53:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973792-new-release","meta_rev_id":"48fd30d0d2da9fdf5b345538be8ef693bd214616","_number":973792,"virtual_id_number":973792,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:03:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:04:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:03:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0e13a48afdb05c78a50ed06b9664b3d70ee1e98c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:02:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c43dd4b0b2c7da90a869c8ea29bf8de5e9f17b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:44:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/061ed08447664aaa9abea0654d4a90c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6aa10f0bd4547949873e1964b398850 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78dff79bce7e4e29b38a4fa3add4d14f : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d423bdcfa4d4b36b6ba0966c8269571 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed139c9312b840589a17c7aa5b9c4620 : RETRY_LIMIT in 2m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/442c48a190a343ee8f532fab8230d0e2 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e94e8109c6e4d4b95ff230e9eb45552 : SUCCESS in 4m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed46382365024e9bb9ce5c6a8622b375 : SUCCESS in 3m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/005d54a980a04dac92d081bb81e4ccca : SUCCESS in 3m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8cf064c65f1c41f488c4805597997de0 : SUCCESS in 6m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ef87ed6c3ee242cdbbce3312e5eeb4b9 : RETRY_LIMIT in 1m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/72396bcd278145f39b64c87bf8b35de9 : SUCCESS in 6m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6368f067844344579b60ca1ce228f72a : SUCCESS in 5m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/55b97985e2af451d943495bc52e00fe6 : SUCCESS in 4m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/633cba2c0b0744c888691edb0c56e9a3 : SUCCESS in 12m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/40cdcc9a0cbd4e7faddba4851445376c : SUCCESS in 4m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ce5ffa3d48204c2d983d51ec9c6b0559 : SUCCESS in 3m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/871fa89310a0446fb256af637e16f54e : SUCCESS in 6m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/badfc5c71aba44e0bcb52a4bd1e109be : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e1f2b3066564c6790ec59cb3bc604bd : SUCCESS in 10m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/21547a4876cf423c8e5ea0773697e079 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9abf0767f4354ae2b3c1d8cbdf460e65 : SUCCESS in 12m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/852547f3221e4d27889b27c407542c06 : SUCCESS in 1m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aff47aaed9c741cab6be590739d21734 : SUCCESS in 5m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/91cdf5e231a8426197a194c81e7ac0b7 : SUCCESS in 6m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b22af79df4c241e69663f081409dfc38 : SUCCESS in 5m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d69d467184694fe5bafb2f4781cb18d2 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/da3a395a55da4f5287b8147752cafb96 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/945dbac3c4324b0a9078991c3c316f78 : SUCCESS in 4m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b95e2ef27e50490495972de410079113 : SUCCESS in 13m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c63ccd0c2bde4debbbf5ceaf162e8227 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34e3a9938e2e4e9aa81ecb4585c585fe : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55944a306cd84cd7a46a28b5e8f45829 : SUCCESS in 27m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/672808c26e1d4a278e03918e46d89245 : SUCCESS in 6m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f1567b741a4f45fd98447f757cfcf4ef : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1968e808c48746a4b34f74a5483e6d95 : SUCCESS in 6m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1b59561826c3421c8b760c1e48ae9701 : FAILURE in 19m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9563d21d37ac4337a0ad645237e47f8e : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e4b75e8a764493585d9cd1fbb317db3 : SUCCESS in 54m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c659f76ed6974579899b069b5997f9f8 : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0a91e2992a6fbc3f12d8b69564f851098f4c042a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:04:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7c28bc30efa7a6894c11cfd06827dbf0a73e692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:57:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aae880af52740de956dee3e40e7edac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2e2dd8acc44c2db66b8bcd09e5ed04 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f434ac69f3c40f6ae68c45737a4ac06 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22fef290bc6d478491599abdf6dc410c : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c81b5e11325c431a888b7bbc63367b47 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2520a9ec5b6e47b097cdbd858889a60b : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7446dfab50c4451cb61939b16b0c170b : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/77db137bacc34aaa977379ef17817d69 : SUCCESS in 4m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4d2a23f1100b4934bf1dbb64fbc60564 : SUCCESS in 6m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2b075acd256a42178b7b33172ad35b22 : SUCCESS in 6m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/01c2dd1fa8514f5da9a1cf6c039a1f4d : SUCCESS in 5m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0f630c15062f4c3b8f69294837d3dbbc : SUCCESS in 5m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fac9d03782fc4749ba89b0072c32c419 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b664ec2272b44c0dae72c59e356d1ef3 : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ced8834907a444148e277f5292b15926 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9b7a2d6fe05e4eb8b6080708ec135cab : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ddb924e16ad7481fb851e0382eb552b0 : SUCCESS in 6m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2f19220150234ebc82c3ad60efc29c8f : SUCCESS in 4m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/67f17231f430461298d04b58b5499bed : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cc95c1f130a427dbc2037b6e3e1e3bf : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cd9570b88b5c41cd9f0caaaffd48c453 : SUCCESS in 8m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36719b78458742a7b3c784ae294a2fe3 : SUCCESS in 14m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ccfdfad956c549bdbca872bbd9471898 : SUCCESS in 3m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5a1eb0396b2e4e49baa07d919a2d7828 : SUCCESS in 10m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8ca8087f322f46c9a330c7e0d5b0febb : SUCCESS in 2m 54s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5f9d2a98c8df4d9d8d707ed0eb479fbe : SUCCESS in 3m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8475b9126d46404488ba8893d3abb08f : SUCCESS in 11m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7b2cb16902b8485d901dea95ee3eeb3a : SUCCESS in 25m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b7eeb045981c4d0087fdc9fe7bd565b0 : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a0e57f2f6a5944998161c47c69dbde64 : SUCCESS in 12m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/144d88fd02f44429b5e05784b2b12c68 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23530a1f655245668844ba0c62aa66cf : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10e7d39425474f4ba1b9851416f013b5 : SUCCESS in 25m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d516defd09a241fb97a1379ab75e54e7 : SUCCESS in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aea61e8e8189462ca8218af1f5ae987b : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44e4b6855ea34f6d923f614a32eef170 : SUCCESS in 5m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ed964df484e945079ae186000e9d4041 : SUCCESS in 12m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ed627282f18c4681acf8b7afaecf90f9 : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/730f2919012947469c47e51116ad3527 : SUCCESS in 55m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e367868923f4f238e0180a3f3b20ac0 : SUCCESS in 1h 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9fd10aeb3eea1f45b2bf67011c523af37801b733","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:03:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4ebb5ce8745ad9a812833ba877b3cf85d717807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:04:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36d0a2b3c92473acbd0fa71a4e66fb3f800279d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c8aa180b7247a196767be2512e75ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e692fdec74456890e51ffcef9625ac : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8ae247a08044749c0a5d15c5136ad2 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11921b93675140d9828197d185b9f5d5 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e95ae481ce3d48e3adbaa10f76df8dbe : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fd6b9b42c5c4ae5b26a5f626559f8cb : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df4dc1ed81a94c989490fc86fee5a217 : SUCCESS in 1m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ec03d7bf6fbe44868849e0272f48ee68 : SUCCESS in 4m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/44a70ec1cbff4a778f274de5f3a991cb : SUCCESS in 6m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2ea3af71e0c74f53a97ed39da1126b04 : SUCCESS in 6m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/037082eddb1c46beb9a4c53f8b1138f8 : SUCCESS in 7m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9d4f0a70b0a8444bbdd2695d1a38a033 : SUCCESS in 3m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8d7bb6afd0c84655b5d19e007347c743 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b36d0665b34d43618ac632948643551f : SUCCESS in 4m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5dff78ff0cb847a498be1a914154681c : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0c74014319e420bafa5849c5ef76a5e : SUCCESS in 6m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/70ec55b070104af78543f397cf0fadc2 : SUCCESS in 10m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e45aa14d34bf4edc9adea5ff17829149 : SUCCESS in 12m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fa8b182a926d4beba0983757afaa5da7 : SUCCESS in 3m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ebfe45f3cadd4e40918780f4a83a1019 : SUCCESS in 9m 53s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ce15656dc4374266a1a8b1d896f42952 : SUCCESS in 29m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9cab81bc80bc412b9b1242c893647589 : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9109768bb3f64733b9bbf7a1ccb8103e : SUCCESS in 14m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/52142558bb5547859abcc1b90dfc7e7a : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d1a987aa26e47a49821ddb375fa94ae : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e213b292549a457182abfe3b05265ec7 : SUCCESS in 27m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dec8b7d7b28a4fcc939a2e34118fd75c : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2486649aa7474d15ad98bc1eb3990975 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb2f7fc7b5d048f0b7298d3f93bc88f7 : SUCCESS in 4m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/834d4092a6e44783bd8c79895762a341 : SUCCESS in 19m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/810b37d4e39141f38c78bdf00c64b713 : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de9dfda5e31c4c61a572b48c5113a78c : SUCCESS in 47m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbdf5a8bad19495da5b0823fec05124e : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0772b0f6a3a1e147315868fb4fb94fb17135c82d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"48fd30d0d2da9fdf5b345538be8ef693bd214616","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c836ea9518964fd688de8b88f270579b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/506d3367a46340d48a75a5e9da293e58 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8f4ddb0355ea5caf8c579d210574dd50122a657","revisions":{"c8f4ddb0355ea5caf8c579d210574dd50122a657":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:02:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/973792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/973792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/973792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/973792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/973792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/973792/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:52:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:52:17.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.9.0","message":"update constraint for openstacksdk to new release 4.9.0\n\nmeta: version: 4.9.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I29c4169e268c6b793eaab02acd531755d9cc7b0a\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I2f17a9664b5cb632c5082f0c944dc593825d792d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8f4ddb0355ea5caf8c579d210574dd50122a657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8f4ddb0355ea5caf8c579d210574dd50122a657"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
