)]}'
{"id":"openstack%2Frequirements~973794","triplet_id":"openstack%2Frequirements~master~I24890fd0d881a18775de689cae2118d6980aa808","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I24890fd0d881a18775de689cae2118d6980aa808","subject":"update constraint for keystoneauth1 to new release 5.13.0","status":"MERGED","created":"2026-01-19 11:03:57.000000000","updated":"2026-01-21 14:53:30.000000000","submitted":"2026-01-21 14:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973794-new-release","meta_rev_id":"49af705956a4cfa2f996350f07dbad5093089217","_number":973794,"virtual_id_number":973794,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:03:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:04:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:03:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ae7c19f5aca364d9fefc16442efddf0a7b54fde4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:03:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dd60f4ae1f42d942ed85a35125b3ed536d80c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:52:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57f86e466d0b488fb8edad0e885e3f88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e394ceff702f407b81af11a9f7dbd1e5 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0e486940fc4056844477e690a67d68 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba0200b0715d4ca7aa0ad760d9ae951f : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6dfb6f5890e4b738b59a1e2b1886dca : RETRY_LIMIT in 2m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f6c304ea6554ac2bd9d5700aa42541b : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5175f3ac3c3465d906fa704fa5e4ae0 : SUCCESS in 2m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e3126983149c4bc2a9ba78e90e885ea4 : SUCCESS in 2m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66113a21c1c941719a27d619c2994ee7 : SUCCESS in 7m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/af08a8fc7c974151a65c6ab6b8dfdb94 : SUCCESS in 9m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/62547a7398b8458eb1255a40b731dcf9 : RETRY_LIMIT in 1m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f5a676ce76f14dc59acbf13f7d382abb : SUCCESS in 5m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1b6e7c950ac042b6a7101326a3c8c83b : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3812ff038f4942a2bdb8bafdc6073617 : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c9d79e68263648b0a0f5eb0a013fcee8 : SUCCESS in 8m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/07ed0ad056124aa69a92e273afe39276 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/090a237805da42009a389e8ccbe6064f : SUCCESS in 5m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dc16e9fe947d43a3b2c16408a5df0058 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e495d6c528b147bc8ba504a8fd349a8b : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9839b38f527343da8bd57de1d3ef4642 : SUCCESS in 6m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/23428bbad2274a1b8e174289237b510d : SUCCESS in 8m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/84dbbfc689ce4a778dbaa1a98fffd051 : SUCCESS in 11m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ac93096c9cea4890866ce9046fd65a27 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c75e06430ca24f41849672d04c6df6bc : SUCCESS in 5m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d90a0e0cbc004fc6baad9e3e1d145e77 : SUCCESS in 4m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3fe6e0987c8e4420985d019ad8e52d1e : SUCCESS in 5m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a8c86f98df524c22a059a00d5427ceb4 : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d7ec5f7231d0438da808e5eec23c7926 : SUCCESS in 29m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8762fa6a7bfb4aad936ec98453d64eea : SUCCESS in 8m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8efc3ff8eec44043bddd6d47df56e956 : SUCCESS in 8m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d00b2692ccc2413ca72d4d529e5022c7 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08de45ddc53c4aa899640c44ab4d4b00 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00e5f920bbff4b499418cffc1ecbedf4 : SUCCESS in 27m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c642df4614784c428d010d47a4c4706a : SUCCESS in 7m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5a5ae17fa56c4843b08891297b637544 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e0526a5cbf3491ab316f850b8f89bdb : SUCCESS in 6m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b4dc5012aeb640a8ba1c43de4495bb6b : FAILURE in 19m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/232b04823c504b30bc5209e049cc09e9 : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c552c90c26c490aa88ed721bf97ebd5 : SUCCESS in 1h 02m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13ad4cd924b14a5c8683b7f1c16c4330 : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d13daa0a6fdb316e95af185b50ce70384d2efc58","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:04:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcc57aa7a85be997772d11cf57e7b88d563b6254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d65e5a48d24d8db4380ec25439ff23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f0927755dd40f1b547e6716aa88b68 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb711ee709c42ff9be8e467fe70afb7 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eb5ea78e57e4dec858641f748654c4c : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b9923e4ed344c72b7ba9d6e157c59ab : SUCCESS in 3m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44b7bc71aa664ba591e60a988d732120 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9489447bfabc4342bca908c7df37adbc : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/727d65ed4830497391b3acd39716b695 : SUCCESS in 4m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d2baf26713b4dacbb4d4d47c7592ec1 : SUCCESS in 7m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0af83d7865094e26937e188993a193c6 : SUCCESS in 8m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/003cc10f09384622a45cdf301f915c5c : SUCCESS in 10m 57s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/055629f6df9b4e4d9478676ee7754dfd : SUCCESS in 3m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8afbf920679c4a4cb5e932e053b3fd5f : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e024dfa5ad3e405abe3b5f9cdd2998d4 : SUCCESS in 3m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b376b84c964e4205af804cdb38a54834 : SUCCESS in 11m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/81dccba1146d4ade9512e5261e98276f : SUCCESS in 6m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4f16369f6ff542dd914124c3497d707e : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c8924efdef154f8ebbdedec236f1ecee : SUCCESS in 7m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/20364b59fe434526a6b9f010b0e92ea8 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e17b941016a4599877dacef865db5f6 : SUCCESS in 6m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/23c3b517aeb74355a348ca6ea6c312e7 : SUCCESS in 9m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/15aba3e1d1c6459c8dc8f6275400a947 : SUCCESS in 12m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8843249113c44014978a160d2d2f6e4a : SUCCESS in 3m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c2ca60b13b854718879d97276b305948 : SUCCESS in 8m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e8e5e5c0be9144e08eaa8224e314b888 : SUCCESS in 2m 50s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/920c96d23f47402597307a8e761ba2e8 : SUCCESS in 3m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/83d6723b8a5649c78690100dc044fd8d : SUCCESS in 11m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/43eb05980bec416fae673621853987a5 : SUCCESS in 28m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9f5693bacb15418d92e5185cef9d9064 : SUCCESS in 7m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c12462c6c21a41a2a906c3a8dd9a2c4e : SUCCESS in 13m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ceeb185265014555931ee66ad7cc0b17 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/475e2e75bed44de8b5c57421e140f305 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1def07ab35f2461fbb26331af800fd11 : SUCCESS in 27m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d2e764dd1c0a4dc0a6052a455a63ccfd : SUCCESS in 7m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/663f10eb1bca43e3a985fb58cc21d804 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31bc96dd6c2a44208e8de572435ab0d0 : SUCCESS in 6m 42s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6120ded09728435384fccbac8b4a8ea1 : SUCCESS in 9m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/60a5364e9ae9439dbcebb5b6d36ed550 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99bf0c92de964281a3af13aade41f818 : SUCCESS in 1h 01m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b870cf24ddde48d7a2d9717ca422989a : SUCCESS in 1h 43m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e0bdd5edc928f80b2123935dbf5c1f40b0a0b73c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:03:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ef96efda13442ba61bbe331dc86dc6f8276eb23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:04:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"335e2df7dceff7740fd8381f4c900cecada0f98b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f2a6a0d02a4812b774dd41648cad04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb309bbe9804b0fa655f56df236ad5c : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1eebd027cd418fb3f721ec9df81b04 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd39d029add24bbfb3c22e6303548442 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96e47878c5bf43e081e92033f4e57a8c : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0286e892677944a390c8465c9e1ea7a1 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f73b11d337f4556943730c47e11c053 : SUCCESS in 4m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/36f358b8bd4a4fe6917af6e5333282a0 : SUCCESS in 4m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45e3528b1cad47d58221ba54e73931ab : SUCCESS in 6m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/53722e093df441dd8a0983bf7b751795 : SUCCESS in 7m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d9dfd4a9479d47c2b7dc0c3043692c97 : SUCCESS in 11m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ac33b1c0e41547dbb1ff50c7237f81d5 : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a2533e810b643aca49c9162e37d263d : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/592fcb83598a415495b553ca6b17983e : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/34f605410bdc4448b0bba8eb3ac5e327 : SUCCESS in 14m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e449f75eb9c94de88fc3a0ae6c3b27c0 : SUCCESS in 7m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b33b04d29fdb4e259541990ffbcaf8d1 : SUCCESS in 8m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f446eaf5b58f4233b5320302efa84866 : SUCCESS in 14m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/595f220b36f64ef5a083b8c922787e7d : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bc116587cb32431984a0f445070e7940 : SUCCESS in 11m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e874a10b63c401cad831017f71a94c6 : SUCCESS in 27m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8dcbf5ee28e34969980216edb28f00f9 : SUCCESS in 7m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e9c4bab315d14613b30219962f09f380 : SUCCESS in 13m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/774d596cad774ed09bcf7c0786c1bdba : SUCCESS in 3m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7ca28094dc024a0bb5474548d3dd9fe5 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82048f62067d4892a1d92eabc0e77ef9 : SUCCESS in 27m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/70288304f91a4e1b859e01553fa5e340 : SUCCESS in 6m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c01e818975d24098b9e76170788756dc : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0a448b0dd0e49988ccc16ec7af6a788 : SUCCESS in 4m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/be5cd49f61c94a8b8d8a6e35904427ba : SUCCESS in 19m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b83efa360e0f43bab28d2d112315f26b : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fed96011c650471c833e6a3d62c764d2 : SUCCESS in 53m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c917318e93643dc95138282092701ca : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"49af705956a4cfa2f996350f07dbad5093089217","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:53:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acbb6228e8f970760a4d8f649a412d54ee563e6f","revisions":{"acbb6228e8f970760a4d8f649a412d54ee563e6f":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:03:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/973794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/973794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/973794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/973794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/973794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/973794/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:53:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:53:55.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 5.13.0","message":"update constraint for keystoneauth1 to new release 5.13.0\n\nmeta: version: 5.13.0\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5e689ee19bc148a6442ec549d114945baa697052\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I24890fd0d881a18775de689cae2118d6980aa808\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acbb6228e8f970760a4d8f649a412d54ee563e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acbb6228e8f970760a4d8f649a412d54ee563e6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
