)]}'
{"id":"openstack%2Frequirements~973796","triplet_id":"openstack%2Frequirements~master~Id1d1c2a6abb878d0ba9ffc112b4323240831ad3f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id1d1c2a6abb878d0ba9ffc112b4323240831ad3f","subject":"update constraint for python-heatclient to new release 5.0.0","status":"MERGED","created":"2026-01-19 11:08:24.000000000","updated":"2026-01-22 11:35:33.000000000","submitted":"2026-01-22 11:33:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973796-new-release","meta_rev_id":"e52ee22a090375137412a1091303562efdd8541a","_number":973796,"virtual_id_number":973796,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 11:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 11:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 11:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:03:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 11:30:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"479f027b40a0cdf320fbad969d6022292ce116ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45e9728473c3823543c14f19827807e840c82486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:20:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3458d3147bc541278fadfd17d2ef2350\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99ba1636bd44f1e89ceabfd01bbe4a7 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a654454ed862421b9dc9c3b7966cdaa7 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65715eeb0d3a447ba94da2886f46c4be : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdc095eb30954ce0aa9fd85cbaf4bd7e : RETRY_LIMIT in 2m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac7882bbd417485099016f74e71af2d2 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/915f959a818447a282e723adb6daa889 : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6bcb01fe4d924bae8cbb5624eca46a27 : SUCCESS in 4m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/18f07db9bc3748159c5b3c56492facf1 : SUCCESS in 11m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6087904d46f840ba970b2fb56525b080 : SUCCESS in 6m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/436d3545b51c48e890dd5dc649f4021e : RETRY_LIMIT in 2m 47s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bc1dd5e49b1a40068ecbf774a631bcec : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b4b04e02cc6e49c88ac12f1e84d25cd2 : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/68c7f71c9c374b8e8aad7e1f77699595 : SUCCESS in 3m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/027f5026285042088d4c8c6635049fb2 : SUCCESS in 8m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e54220043764427da50461e3449a6fd0 : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7151c63e21c348b79453c28ecfefb975 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/00bed14f4f5a4ff59ffabd9235e900b7 : SUCCESS in 7m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c7651882a1fb47dc86f93fb3ac9f5356 : SUCCESS in 18m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5097019d1a564f798b6878579209a1ad : SUCCESS in 10m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/54fdbd9ec8b542758f76248151227b18 : SUCCESS in 7m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6fd1fbfc77a24249b35f74e8199d22e3 : SUCCESS in 10m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/83a812338e734b3ba029b0873709ffb2 : SUCCESS in 3m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e076bb8446044b42a00eef932dee929f : SUCCESS in 5m 49s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/79b94cda2f4a4913ae51d0a33db4e3bb : SUCCESS in 3m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/822c1c081cba4b06a3de203b7678acac : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/36a4255322d24aec960054b597484b65 : SUCCESS in 10m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5365ca2693f548b79af15b24e7fd6f8b : SUCCESS in 18m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f3790c25ef64f88bb7a66a110d44c12 : SUCCESS in 7m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7e8ae1a987164493830b793823198be5 : SUCCESS in 14m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a9c6933070314aa3a2ac60349e06c7d2 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3132ae0dd1bb43498b809075e70e3adf : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9888db4fb6e42e5aa1f616cbdc8e28b : SUCCESS in 28m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ea51feca78004f2483a19a964f394332 : SUCCESS in 4m 05s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/34ab6d1ac57f4c46b13b808b9e7196f0 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ddda730eb29444f9901065b988e0360 : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e7531e7c447c40eb835610f045fdb675 : FAILURE in 19m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/299e9c01a1864deab91d4ff643e828ca : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1a75202b8b46409b69bbfebc416d35 : FAILURE in 1h 05m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc66607db93641c392aa83274dcf4ade : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"717a47b19e2cfed9dc91c57f3365065c07f6ab4e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:03:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf5c653bb0524d3fd3220b1b363afdf1e3a891e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:05:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d173a91934e44b3b2417b6622a51fdb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22f60d8827542f9aac30356bb91d4f6 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67164cbc836a4fddb6a59c5820d1264f : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bb65d9a69964464b348f2d51fc2fc67 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5376eda72474b11bc912369286715fa : SUCCESS in 2m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5fb32f28153f4501b037500866c6f25f : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/508467243dd84d92a4245d1efd211352 : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/deef290480e349738b7b5159a63fbded : SUCCESS in 4m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f32f4328dba74c3bb2335ab5dac19d33 : SUCCESS in 6m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c4d1e826c97a4697b0c8d6a30b2c3035 : SUCCESS in 4m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d00a3cf306bb4f069a0a1ddd0f85a034 : SUCCESS in 5m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4a9a5795ce8841e98fcd2778b1411390 : SUCCESS in 6m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0b7c8b26612b4375abab0bd782dfae6a : SUCCESS in 4m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e75bf21e6b8e47309136fee84fd81bdf : SUCCESS in 4m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cb53de88d39f40309b94d83a470e0ea4 : SUCCESS in 9m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/368dccd0fb7e4c13baf53df64dcfffaa : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d72ab24e67044248b7270652b1cc7a19 : SUCCESS in 7m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5a85689e917443bf9802c0fcc168969b : SUCCESS in 8m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/edae76e7911341658086a0165f781924 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c87e6be01a764269a7e91a795c79810b : SUCCESS in 9m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9b113bcc000e4ccbbc773088623dc1e4 : SUCCESS in 7m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8b29f2f4f5cf41d68f34ec8b61a17d8d : SUCCESS in 11m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/368095feb1be4265a24c8979944d6b08 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a23d3031678245a38586c5e50cf1a0e9 : SUCCESS in 10m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/64c40dfe0c874aae8ba80354a4885b67 : SUCCESS in 3m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c4ad2b93b67a491f84d5b79b48fca92f : SUCCESS in 2m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5fd49d5b8b774cf78c0d6e3fbac55314 : SUCCESS in 8m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/85ef6fb0a314466697c334cdd6e66eb3 : SUCCESS in 19m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b8894e671fc45faa354ac8439caaf4b : SUCCESS in 8m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/85bb50db4c284dbe83d6d00055eb3dfa : SUCCESS in 9m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6fcba3169e834d1ca1a68ff83a489e8a : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3439dff271d47a0aa7f05ad0d2c192a : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/848f88b1b1154a8db88c06cdc88c05ff : SUCCESS in 24m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2578878e16a84191a2c2c2273a6763af : SUCCESS in 7m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5f2fab697aab45339ba201bc60afc170 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9af876d8c91c493db90a4923a0c34c74 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/94fa2bcfb8044c67a888aeb831e29a0e : SUCCESS in 14m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c06d85d9a5d34403b938644b637bb66b : SUCCESS in 7m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1cbaf71b7af45b5ba0479c1bfb0c6d7 : SUCCESS in 46m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28d850b5c1bc43e5aab447544a7fbc40 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"79988782cd4c24d59439be46275e8d44d2a524c2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 11:30:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be347671e0a5819d1af135da2600943b9522d385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:31:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0014a640d95366ebb9625d237f0875ddc94199e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:43:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9783ba306ed042aa8ed06884564d3de3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9493402edc47358ee6a6919f09647e : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03daff1472b4ba583c3ca9895798585 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9cd3a445df44d4c98a6d6a93c1369cf : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a17e680099da4610b891b9b7bdb54537 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a31b6f8f390444719a9d012fbb2ee424 : SUCCESS in 2m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16d06774e70943aaa18479464e9fdbb3 : SUCCESS in 4m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ccb97b511dd949799cc0203b0422f4dc : SUCCESS in 3m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98dde439787644c1a77e521465db3610 : SUCCESS in 6m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/230db329642c42809c8637d82e175c22 : SUCCESS in 12m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3f9d315964744fffac14fab9caf59a81 : SUCCESS in 7m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5204169d34a145f5b63acd4abebc1d16 : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/73c88386e8c341a89ceee9d7254be8c8 : SUCCESS in 5m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bbebe0cfc45444cfaddac5c2a8747b33 : SUCCESS in 8m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b3bf91beaf084841b2862f98f734670b : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc3aff9a8399430888220f0a72923448 : SUCCESS in 7m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4bb6c44aec7449a2a0b781f7b450c186 : SUCCESS in 10m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a35c9dedee76413ebf2c187b57373752 : SUCCESS in 7m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b79042344ec742c59f301f45a69b3b72 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1fc49cbf6c7f4a18ac0826f320e18ff1 : SUCCESS in 5m 34s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c39d3645f5fc4b4f9cc13c5fd83d5032 : SUCCESS in 19m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e7a258d7441412a9b30f96be6e89b82 : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d779f280000944f28e98836057ce96dd : SUCCESS in 15m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/00d1a4f61edf4bbaa1c1507fa7a9fc9f : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6973fe3633824a5b9955074ab6f567f9 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/746a90700a45414f909f3a42a5cedeb2 : SUCCESS in 26m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d52e2e15c8264eff9ed83772ec0adc6d : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cea37101519c42d780a07f1bea71fd95 : SUCCESS in 3m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fc2c14bdce544969c9469a4506c5cd7 : SUCCESS in 7m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/53cefd8a5e574eb69d266f8e0a94b3f9 : SUCCESS in 7m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cb06aa849e15486ba39baa741832afd1 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e66214e03064a15be4b0bd27965cfcf : SUCCESS in 54m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a44ea98910c4a779677fdd0413f1b20 : FAILURE in 1h 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0cf1339f186bb1c8fda9f11d2dc2e2a282f43d05","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-22 07:16:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b30fe3e219cc39339ca883f363967a86863f498c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:58:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6556902c2b45ab95f97ebd8c243b1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be400e0f3d849a4b98d8533b6a53ff3 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067eb2535eaf416abc3dbf2a4cb03682 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c67d8d5f3ca8438fafc96ac4e8b2d96e : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0591dd1d71024b38bec20be3116e2282 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ebcbaee74134370815376c91d428aba : SUCCESS in 1m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0208780519344c76ae0707d64aea9861 : SUCCESS in 2m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a866c5732a847d1bbc23fd227e2cfb2 : SUCCESS in 7m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f7c7956f912f4c238f65fe5556e6e264 : SUCCESS in 6m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4fc51d952d9146929f0f5fc8ef2c3f2a : SUCCESS in 3m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/56679926ae1541a795ac595697ef14bf : SUCCESS in 10m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/72766b8a00b547afaf4b20bc94e2e068 : SUCCESS in 6m 21s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4b8c7e144b4e408196597d4d2585b932 : SUCCESS in 6m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e4e88192ddd64c01adf854235e9cfa1e : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/66f299bcd3f9450bb1315871c034e866 : SUCCESS in 7m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1b43032541454c5893f55735076b6cf7 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/42f70acfdd0d45a99ba0a6f0dce47116 : SUCCESS in 6m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/072d6420312e454e9a118d5ecd58b062 : SUCCESS in 7m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c227c8cf24684c93ac6fcd4b95154790 : SUCCESS in 17m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0972b8de0af84194bff3176e50ea0bd6 : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bf88d3b22b094b81b38abb4d7a3f421f : SUCCESS in 8m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/165ec2aef26644b29401f7a3bdf2d97e : SUCCESS in 11m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/851a16da410f403b82d998533c788d16 : SUCCESS in 2m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6a976f3ab84b407e82879fb57ad322f5 : SUCCESS in 6m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c362f8e5e4e244d789675330aab950e6 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/92c66cbf456f48fcaca6f72a9eb08384 : SUCCESS in 2m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b55155939fbe44c0b2c9b0b3cbaaab67 : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1ac05a39262d4e34826e0bcec3b6807d : SUCCESS in 22m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3590bf1428d84e66966171d168848efc : SUCCESS in 4m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/aa3878d2e4464a07957861e063b9bb29 : SUCCESS in 12m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f2fbdc7ac5824732af74d27590caf354 : SUCCESS in 2m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80dcee5feca741bc894e106f04cdc98c : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59dc8cfb711e4cfa8ed3dc392f5a4ad3 : SUCCESS in 27m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fcb3131fadcc4a6db0aabde782458838 : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b35eee6b8ec44f4c9893fed891966b1f : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/942495783b414224895244f5fc6962dd : SUCCESS in 7m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9e86e775905849999c7a69745c4cbed4 : SUCCESS in 18m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/226e699440bb4faa8a01c2c1c9e57385 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a67267bd1404885a22f3bbacf395c84 : SUCCESS in 35m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc412ffbbdb643d09eb559e47b7e74bf : SUCCESS in 1h 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"379ff6f394711ca6c77841969fb47d1156b398da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:59:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62c4646ee38050895f9d0602c43e19f16c709957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:33:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f14016bdefe5468fb141e7a37560117e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a14c085abb4de08adb5aa4e3fdf405 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc3f3be10684be0a311baa28c3cbdec : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6b3c590943b4bf79c9d1379f91a9027 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6749eb0d6ab14ba7a2928c392254d710 : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d67ba089747416a9bda95e8b4eb1941 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d78685aa7eda4635879462a4f2f7f8fd : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce4469a4c25f473e807f80342e4cf63d : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4d0a687a54714403b587ac66dcba2414 : SUCCESS in 6m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/77fbd719218449febd2eed914260b44f : SUCCESS in 6m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c250cde5b4d4482faf4a62c6be1056a4 : SUCCESS in 9m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/34f4c3cd15454588920a6c8c03354113 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/41b11938772c480fb517062c514925a3 : SUCCESS in 3m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/861137fd311a436485f916192ea7dd67 : SUCCESS in 6m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a50ce3b78b9242018594eb40ca50a97e : SUCCESS in 16m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4173048421a41349912c2dce611e3bb : SUCCESS in 7m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6de87625530a4ee1988e1f84639f784c : SUCCESS in 7m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5fc67a533ac842cf8a448e40e935c526 : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b5108bf1200d425fbab279b69a3fba0f : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac33868354fa460395744ca22790354f : SUCCESS in 12m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bab7f4e3411f4d35b14934c8fe7142fe : SUCCESS in 19m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5555fe4f5a044f358b3bedb83af636aa : SUCCESS in 7m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/46271a4134614718851d207ae5461d1d : SUCCESS in 8m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/61467072cff54ae9bdaf8cb339eb75df : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7946a144a8394b00a6c57cc984e84ec1 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a04afc862d3d4338aae0e926b8e00f6f : SUCCESS in 24m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/891a0bd5dee24a1ba9d93e8db6de70c8 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dc5cf2b004664652be6da6665c386a67 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/962bbfa849954f8ba896106561aaffaf : SUCCESS in 3m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/06e164fc5b214c1ca0eaa3ede539f087 : SUCCESS in 13m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a91fb9b6bdc24ad2ac4af81096b0163a : SUCCESS in 6m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97e2bad7c5cb4c12afe05e5576a5d60c : SUCCESS in 1h 05m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa5e49243fdd4056ada33a047d0197d8 : SUCCESS in 1h 52m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2cd931a222a61bf9287f6638a777a22bcf10004c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:33:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e52ee22a090375137412a1091303562efdd8541a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a5eb9a9bbe49c4925e0cb4406448e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a161912bb00432087cd84cd74a1695e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fb54f0733a248ec2f4c37eae5cd5193c2cf0017","revisions":{"4fb54f0733a248ec2f4c37eae5cd5193c2cf0017":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:08:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/973796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/973796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/973796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/973796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/973796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/973796/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:58:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:58:13.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 5.0.0","message":"update constraint for python-heatclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: team: heat\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I19af1e6a7af4cf868d125ef7bc7e6081e1e7fb1f\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id1d1c2a6abb878d0ba9ffc112b4323240831ad3f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fb54f0733a248ec2f4c37eae5cd5193c2cf0017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fb54f0733a248ec2f4c37eae5cd5193c2cf0017"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
