)]}'
{"id":"openstack%2Frequirements~973797","triplet_id":"openstack%2Frequirements~master~I10e984b2b567424008611b592dc3c82934df3eb2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I10e984b2b567424008611b592dc3c82934df3eb2","subject":"update constraint for aodhclient to new release 3.10.0","status":"MERGED","created":"2026-01-19 11:08:33.000000000","updated":"2026-01-22 11:34:32.000000000","submitted":"2026-01-22 11:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973797-new-release","meta_rev_id":"07109aa54402f64808d57ab87ab247a982c5de53","_number":973797,"virtual_id_number":973797,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 11:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 11:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 11:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 13:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:05:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 11:30:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1fbc1db647b04f7b9d8d8cb20e81e748ab0c857e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:08:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75a92390a418caf40a32104fd33e82760c4ce4b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 13:01:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/487c8dc5209c44508635161cd95713f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dcee3ee1b5c4598b3cd1d99131bf8b5 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936346f14490482bb86c6d995bd4aeb8 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4494d39efc5941c7a5883b380c968d35 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b72a6181a3244bed81a60f0eb2c0b707 : RETRY_LIMIT in 1m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c98f6261f96d4b7c87351c4dd55503bb : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5fbf00ee29c42c59e804bb18490a0f1 : SUCCESS in 2m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eddd573244c04608aa9973a80a4afc91 : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5089422f54ab4fac94befeec896cea69 : SUCCESS in 3m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5f9f3103718d4b84a5e95780fbfed4ad : SUCCESS in 7m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/856573ad453648c09081aeac839e9b44 : RETRY_LIMIT in 2m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7fd43bcf9fda443989c5d3870b8ea102 : SUCCESS in 5m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4ba96adce9fb4ae5963f06af9d851eea : SUCCESS in 3m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6b61f5cd1b934f82ba3ed2a9228804a1 : SUCCESS in 5m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5bc1ab7f5e3c4cfa99243b9b7d50aa8b : SUCCESS in 8m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/51f3184c42cc484cbc58c9a9a3301cfe : SUCCESS in 3m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f18c8d76f1b4160909370ca00096921 : SUCCESS in 5m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b888581ca2634be68c739d14dfa5fef0 : SUCCESS in 5m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/10d39bf404be45068dc2b68f902630c0 : SUCCESS in 19m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f62212d7fab641b28909c7ed3356d19b : SUCCESS in 7m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d6ec31a5bf949bd86b6ebd8fff6a7cf : SUCCESS in 6m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/016ff518c62149d88806e51cece4fa4a : SUCCESS in 11m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2a51064e90c64f678df3d534ae18d011 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/58b962e212334ad5a594e2a67a848fb8 : SUCCESS in 9m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e3c6e67a201b4407a45117cab03b3305 : SUCCESS in 2m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/722e1815c48c47b0bfd04083658c5364 : SUCCESS in 3m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d7688bb1b8244d978e78be3b966acbdb : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9a9bfd80109f4f0f96cd26cf671feca4 : SUCCESS in 22m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71935ca5128e4e2cbc2e251193829927 : SUCCESS in 8m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f1571681fdd84a45a60cb39e4829043b : SUCCESS in 12m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/838058c53d5d4004bbad781cf0d51b36 : SUCCESS in 3m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b959203d15f14b10b8088c701d8c7841 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c5d9d1e01404df5b1114811bebae30d : SUCCESS in 24m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0ba961a3fcc341ef99d72ec09610fc5f : SUCCESS in 6m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1f0d043a7690425cba64cefd95cd0e59 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d9fbf8a087043549c881e8b9230944f : SUCCESS in 3m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6da3a0ea0b84412b9cd040b67b0831ff : FAILURE in 17m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a894521aabd8470890aeb7b284529fc7 : SUCCESS in 4m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03d8f8fffae8406aad8530008e2e2e19 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09d00a0fda574084b9aa488af5c6c207 : SUCCESS in 1h 47m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1a3a26e04fb72b8006deff71875bef5b18de7186","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:05:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0427722110be863aa7c21e184a6ed1525d125274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:12:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74fdd88686bd4dcdb4293b46c63688c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873548a8ba7b46e9905a5a1f4d0940f9 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c71ee416144fc393cffff213ea9e8b : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba002324d3d247fa97fe7b34677df8c0 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/704a0c210532416782732b6326ed5491 : SUCCESS in 2m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf43807abbb04de19876cbc869431e4d : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4eca83e6c1a049739fd14c147f78e4e3 : SUCCESS in 3m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/165c453476fc433a8c0ac95f0547b371 : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d21485fc9bdd4944bb02c7f8150e8161 : SUCCESS in 7m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5212f69ee4154aba93d9323656b1209c : SUCCESS in 7m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7a9220adb1764688b634c8730cdc8ec8 : SUCCESS in 6m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5b3809f5bca84571bbd6777b6e784e27 : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d5adae2d7be044d886db0b3624e5bd0e : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/43c1d5158dfb46f5bad96843fec7353e : SUCCESS in 6m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/83d5082847ef42e7897aed9a90bd3165 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c8cb38d830dc44baaeab588fff8ec2d3 : SUCCESS in 6m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b594e893ed88480c94ae0641487dad36 : SUCCESS in 5m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/66f11f5f871b4604aa77fa1dd413eca9 : SUCCESS in 8m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4a7436e5d98244c496a122aae068d254 : SUCCESS in 18m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0258faf9d6de4a4e97176a4501f73d68 : SUCCESS in 7m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a35532ca5daf4458b60ce94eb7380a93 : SUCCESS in 8m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/01adb7be8eb54f619691143416e094d8 : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e292b942d34047cd97c2cb507cdba02d : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/880092d49b40420f9296b047465146d9 : SUCCESS in 10m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/19f839ad57b44592ac3cdaa509aedbe8 : SUCCESS in 6m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/37ab8e0cbb21449da673ff590290c93e : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/88bc0b1716b54c169be5160dc39c2213 : SUCCESS in 10m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ee30ee7e04364c55a633cdac818fbac5 : SUCCESS in 19m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/83cf69f2ddb64fc3bb83f8656a05e6ca : SUCCESS in 9m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/36587355429e439894de165aa8f113a5 : SUCCESS in 13m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f80d7143270747d391448eebf761d3d6 : SUCCESS in 6m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbce257c1aae46fca4441bddca127065 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67ffda8a3a094a9391c184b0d2e85601 : SUCCESS in 24m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1135025c05ef45e38fffd498c2eac2b6 : SUCCESS in 6m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/52d9d36111a64d658a78474d0c7fa18e : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af15a635e1b44635a99f91c31bcacc6a : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd85770274534b759cd5461981eb559f : SUCCESS in 18m 38s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/90adc24fefcf4b7fbc06b948a9369f1f : SUCCESS in 4m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30817e0c6ea242c6adcecbb2a0d1e0a1 : SUCCESS in 48m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6658e524ccfb4d0ca1c9e6ab89b132e1 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9aab91c6ac9f66963554dbe2d1cb0761a5a1009d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 11:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac6313824b7148f4d30d2ec07a36c2373e20b117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:31:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35cdb8a945abd5a1bd02ae5d43f5991063a85941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:57:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab00f9b9bcfb4bde8064db392472bf40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db41fc0ac56b432a9853ab4155e36cfa : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/847f5a4affa247bfab6f0f1717ae7bd6 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65570ea73b4a4943bb54cc89ccb8aea0 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eafe68391c24dcbab966d4a3bc15ca1 : SUCCESS in 2m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a0a4d079875472fb604aeeb1f3b6077 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81d49dd46fd54ec2a7534b1ac3c6cad7 : SUCCESS in 2m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/72dc9f1d56db46168915d12086368352 : SUCCESS in 4m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8871ba50c2194399835d266c1ffe1ab8 : SUCCESS in 6m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/70605feca6784d9c8f60ac24443b5bdd : SUCCESS in 6m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/21135449ef18488da0a99aec272570fb : SUCCESS in 12m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cea46a9e02184f5c9cb5362e73de58dd : SUCCESS in 3m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/011e29a0327d47a6a46fbe52b936b013 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fd1102e80918452fa44ba46c71d8c928 : SUCCESS in 7m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/03b73e95414d4776b112e4423aae1d7b : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b3aac2a1bc34eef9cb756c4bbb69d18 : SUCCESS in 6m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f98141a313554c7099e8c7eb5689d65a : SUCCESS in 7m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/66e22b2f8eb74a5e8d5cd4343d1d57e6 : SUCCESS in 11m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d1630df31a5f4636aff8ff427d0c852f : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ff98a0973d794709a28f6f80935d2f8e : SUCCESS in 6m 23s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e3057f0d45c4286aa377ab035333add : SUCCESS in 27m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f188372e885a482f8bac99e82cec4bba : SUCCESS in 10m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3b8d3acc6e7245c39d0a1b834b58ba46 : SUCCESS in 15m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/de8cb6c82a8d40c481df9ba21f3c1bb6 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfa07fe8922842939687efca936aeb1c : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/107238e557734fa4ba01080531c27e43 : SUCCESS in 20m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bb901eeeb43244328b891f4f6ed74807 : SUCCESS in 4m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8cd7a17b076b4e6c9dd299e05f96f438 : SUCCESS in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cfd8e018217842b5bfb45fdfefa2fc12 : SUCCESS in 3m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a04f68bcc4cb4fc6a403009553acccaa : SUCCESS in 7m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2cde58de01e6469f87c0be1ad69fd620 : SUCCESS in 6m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89c83f132221484d8d415b4587e8e6a8 : SUCCESS in 1h 02m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a64dcae41f254b89acc32d731e0b2200 : TIMED_OUT in 2h 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bc099c3b36d5f73f967c155ab971d82e1f14a7df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-22 07:19:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d111648fcac144e9dfddc8416788e102e13a207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:44:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cddf1192edd04f5b9928786e77b25afb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa24f25c85047268ee1180ce4a81967 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411e95409a1a4896b53df3275227e1b6 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f57f216c2794b6baab43dfbd8b13419 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f881215baf0944f990f23dee0fcfb432 : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/280a05a763ca4413b251500f69f0767d : SUCCESS in 1m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/061c7e0b84f2457d87903a5c6027f3d4 : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c3eb45f896a4b5e981306bd81a049b1 : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/52708044c6a14738910ae00c7ee31b46 : SUCCESS in 7m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/78a51710684b4274a9528ce349adb8b2 : SUCCESS in 3m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/35db5cc5cdac40868b3a8b18f4881aeb : SUCCESS in 2m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6544d86b950446df9dc978cf60410ca9 : SUCCESS in 2m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/757f514fbd54430da4048a8f6887d92d : SUCCESS in 5m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3c99e1d6682247a58f9c0ca9e452676a : SUCCESS in 4m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/042cd729551a4b3dbba00bd2f15e3f25 : SUCCESS in 11m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d5b7eb5936be4781b93e6275c739714e : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/23e74be17ab8446ba5809bd0f9e702a9 : SUCCESS in 2m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/80620a2f8c424f2e9167f271cbaf49ad : SUCCESS in 8m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/389880d74da64c5a878d15edc0ec383b : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9db076fa083f403b8d931edf396edc39 : SUCCESS in 7m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3dc84054f4604d1c853a7e8da207aa1e : SUCCESS in 9m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c3e80470df3f4d8d91ad4fd660fb67b8 : SUCCESS in 8m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/59edfa0b0452469a84774c1fe5c816a7 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a506ebfbcb424054aac54735f38d41f0 : SUCCESS in 8m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3d69dd6fe5774935bac7f91467a11803 : SUCCESS in 4m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fccaf087080b4bdb9801c059cb791d3b : SUCCESS in 4m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b35b89795d764042812aef01f8462b9c : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a78285f1d1f148dcb7c0e7320a6bd55d : SUCCESS in 19m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36c5477a7faf4a3d9801c5d727088e0c : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f0f3501c88b74dbc9f20f1e3030cc6f8 : SUCCESS in 15m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/da5945eeb2f5479e81f843a8571988ec : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78cacf6ee7ac421bb655e98e9299dbf1 : SUCCESS in 1m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e97910cd1e114e098c6713c6f1e9411c : SUCCESS in 28m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f2da3fe4c684653b06667fdc5f186f2 : SUCCESS in 7m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/38956da6b0d0497b99fef3b7bc534ea8 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61e2145c86c54792975c839d174f9f50 : SUCCESS in 3m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d25b1e6b34d245338e375bb36e0e1fc3 : SUCCESS in 7m 46s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/efa3b504be1a4936a770d778f79a2100 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b75775b8b364c8dafd452e62cbd4f7e : SUCCESS in 1h 00m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62c1d5607aab4ae296f8d387bc2c7b82 : SUCCESS in 1h 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"82779e7bd53b48fa2523e1c55d2af6e977267bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:44:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e35d42b2af2be6e8d5adf8553ddb1de3434ae118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/039be4fb27b94967b89a3bb282e2f01c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2ac33c964d460ba1e2d31ef475419b : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5117a87d39a64979b917f51daed96fab : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96dc165c22aa4f91b4278051e44e2540 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0c1d1340cdc454f8229d2915be08f69 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2d27deec6e24afbabba1ee9a9ae8be0 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ebdda9e10984197887fc1b01bde30b0 : SUCCESS in 2m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a66b1e0909974aa4ac89b817ce492b1f : SUCCESS in 4m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d854232e390146dfbf985fe4cf919941 : SUCCESS in 5m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4235bd381c814e63975dde472665977e : SUCCESS in 3m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6ff00b2f752b44f2b675267a49f42b9c : SUCCESS in 7m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2f09198968d147c182081f166a8f5472 : SUCCESS in 3m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dbc234108db245128db3595d98a79c7d : SUCCESS in 4m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/48b5982e39f54a1cbac7d4874a6c7dc6 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ec121f215a04a0f975a11be5c1806f6 : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e1d2ebaf84b480493a1714f1ca7b582 : SUCCESS in 6m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a52703674792405bae38e4029ddaad4e : SUCCESS in 7m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/332f595e9f3f4b7aa5a07f25590e8d8a : SUCCESS in 14m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/26fd981b3948464a8ad76954b3e962d9 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8490b21c02684976b091662993fec281 : SUCCESS in 5m 42s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f903f0d5175741ae8c2f203739d40f1e : SUCCESS in 25m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/09a35ee15327433cb1413cbe09dc198d : SUCCESS in 7m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/227f10e2ba0a4e6abc7c548cece2e6c0 : SUCCESS in 8m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/51bd64b7fd6c4d2c8cc3778d32e8301d : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed99fb77f7a247709c0c578c33a9263e : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3172aa5ca29942fa95d28ee9d1ec81f0 : SUCCESS in 27m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/210ade0a42b34f658f873ed44ea62815 : SUCCESS in 7m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7b0ab15d59b245caad50d7fb0e5f399d : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c2f2f15633a45329062062fa237e542 : SUCCESS in 4m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b08ad98115b1496e9472efa617436b83 : SUCCESS in 7m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0fb75ca57e7d42de8675d7e53b9beaea : SUCCESS in 6m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b03d0f54437a43869e89655ff357d398 : SUCCESS in 1h 06m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/846b85769d534f7ca7244517c77441ac : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e8ceb9141e06c851c6f13f46b4de9b8a5c416133","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:32:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"07109aa54402f64808d57ab87ab247a982c5de53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:34:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab466632bd13475b94c5a7f656495cb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9568958653a148cc9b5a6e5cdde7fc46 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ca7b8a19145f2193c18b5c1e0b926bc57c961f6","revisions":{"5ca7b8a19145f2193c18b5c1e0b926bc57c961f6":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:08:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/973797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/973797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/973797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/973797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/973797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/973797/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:58:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:58:31.000000000","tz":0},"subject":"update constraint for aodhclient to new release 3.10.0","message":"update constraint for aodhclient to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id48425fb8abf7fce27f8b44385f09cfdd39ecc05\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I10e984b2b567424008611b592dc3c82934df3eb2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ca7b8a19145f2193c18b5c1e0b926bc57c961f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ca7b8a19145f2193c18b5c1e0b926bc57c961f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
