)]}'
{"id":"openstack%2Frequirements~973798","triplet_id":"openstack%2Frequirements~master~If9421fb91f0c2cd4a3518a1421a527a192c660fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If9421fb91f0c2cd4a3518a1421a527a192c660fe","subject":"update constraint for taskflow to new release 6.1.0","status":"MERGED","created":"2026-01-19 11:09:40.000000000","updated":"2026-01-21 14:49:54.000000000","submitted":"2026-01-21 14:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973798-new-release","meta_rev_id":"cf3e215b7b1d40222f43dc13ea090e6a9a0f9de5","_number":973798,"virtual_id_number":973798,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 14:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 12:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 12:02:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:04:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 12:02:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c687201a662a7d2063711275bea17cecc9dfe34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0853e6f90bda6aa889de6047447354d22f6d9bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:34:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e485029f193485bb0760ff78dc92784\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698ac059c6dd4f26a65d2584dd7f7948 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429f9d16ea9a4f8f8037eb27f86e49da : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6ffe471388c4baf9b85a2226ecbb38b : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/816d433ea4b04ae5a71151aa461244c7 : RETRY_LIMIT in 1m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58bb1b6c59b24b508f6e4a14d303e9d7 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0468091f89e447d99628648369cc7d9 : SUCCESS in 2m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e303181285cd432d8b3427dca30541e3 : SUCCESS in 3m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/de8b14bc00804e258139c81597ed3132 : SUCCESS in 6m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/62106c10e71049188ad90aeec2cc6891 : SUCCESS in 7m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dce417b834844cfc9d9dfd6a09388a92 : RETRY_LIMIT in 1m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/773305692f43419b8bd3ac7aecb0c540 : SUCCESS in 5m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/57524efa4cbf42c2b5264510e0ad7561 : SUCCESS in 3m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6f8a73a95b554d058f37da5cd3e5d74e : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/549602828a694a28818bb97b46f8b8f6 : SUCCESS in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/577d172d711e4897be8e8b69b2b47a4b : SUCCESS in 6m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90feeac0aca44de487bd7f20a72a97a7 : SUCCESS in 5m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5c5f67a1ec1a44b4add0e88747901a66 : SUCCESS in 4m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dec3aba770224ef99a24c11782c03621 : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71a84bba76094063a96bff47634c4e5b : SUCCESS in 4m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f857b40775c14699b493addf4ff26a8f : SUCCESS in 9m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1ee4c282011f417e8af0978572beb41e : SUCCESS in 12m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0ed871958020404aaa9d3fb048a714ba : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d0ffabd69110480b8a6641dd6fda7c93 : SUCCESS in 8m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/18f527cfbac04e759a73cb245277390f : SUCCESS in 6m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d794ffbdf84e4e01ad3764ff061f1ec7 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9320468fe00d4205b488508b38ab95da : SUCCESS in 12m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4bb784fdd8894667af5feeffbbc48714 : SUCCESS in 24m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d5e638dc72a4f908755a50e1e9ed6f4 : SUCCESS in 5m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7a708e6e7d9e47c6a8f47e2cce9595fc : SUCCESS in 11m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dcdaa67a885142748c832b423a5ed34e : SUCCESS in 3m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff157c89d5ef426fa470ba86a7f58c43 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92d05d4ab0604b9ab35a4688ea19f51b : SUCCESS in 25m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1e25b02ed40f4d268f009b142eb04ed9 : SUCCESS in 7m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4261b495487a495cb3bf6abdd7c51b7d : SUCCESS in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b61363fbf124f38aeb5c65c6a16ed35 : SUCCESS in 6m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f63c05575fc341359e17901839fb2261 : FAILURE in 18m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a6e72748e43640188bd7f5033b13de48 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be6506af9c544556843716d8bbbd608b : SUCCESS in 1h 00m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7677eb3b083c4805b7e8f8fe874fc703 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5b4635bc41d21edb35f4775fd0a09a664c58ab83","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:04:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff4b195a41b14b9bc79b06e1bc0d82dde00ebc0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c529977da24420eaffc8ef268ee427b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cf039f81e349b0b2991f9062fc2ed9 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f736cef15224c38a68627c4f40737e9 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdd64f394d8d4844ade281d4fc2544fe : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85cbde7f3c6a49f796dd11476b6a97dc : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e40a553d3ab9466e8f6bf92f099da1ba : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5430d570a1b14b249bdae8a6591f0ab7 : SUCCESS in 4m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/98b70b162c3644eb8888878156ad40e7 : SUCCESS in 5m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/48cb8881c33e4cc89c33ba99c0fd6001 : SUCCESS in 7m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7676c7dcb6f44807915c5a60190313d2 : SUCCESS in 6m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c154f13b2fe647a89480dd486b8bac48 : SUCCESS in 6m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7c41dbf3e70a46edbda2a5edd32ffcb0 : SUCCESS in 6m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5c8fedc9ec7344daa26942dbf5ebe21d : SUCCESS in 7m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5a12ae5d2e3f488480e30f06a48d7a02 : SUCCESS in 2m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c4d70814ce2f47449adac47258fd2ca3 : SUCCESS in 12m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1dcc2124fb004b9a8f83641ec671627e : SUCCESS in 3m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bcc3babbe16a46c1859bb0626d8d581b : SUCCESS in 6m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1aec2cbfc3d24843a41ceac118b9d8e6 : SUCCESS in 9m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a45274a45dee45e1b1310d50623564fc : SUCCESS in 15m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bf3b8887815401f92c381f4b26d42d3 : SUCCESS in 7m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9956124a66ad4fa3b1425ece2900e8cb : SUCCESS in 8m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/710644d6f055469986e57773611cf501 : SUCCESS in 10m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5ba18f39ac9d4a04a9250eb06b4e3e08 : SUCCESS in 2m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b49a2490b857427ebf9035110abea69b : SUCCESS in 9m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cafe17c5ca5c4b4f9c0ce3bd4e398166 : SUCCESS in 3m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/424f5acb0d83441e84417bc126021479 : SUCCESS in 6m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8ce638cdaa8a4ba5ad4c6476b3055e3c : SUCCESS in 10m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e403afafb7fb42fc995faa63b6256eb0 : SUCCESS in 19m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3c990a2cc72248f393c962602116cdd4 : SUCCESS in 9m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8979552e194b4273a115a9f123593bbe : SUCCESS in 15m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4eb66e6c6618455993f457c786ab4679 : SUCCESS in 3m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/edc6765be0ca466089d9881e19c539e6 : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2448717a6bd64a84a3bcf473aedb924c : SUCCESS in 27m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c26661c335924e50ad3dd4cbea1fae63 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f469e54daee64f2e8f6f920055cfb2ea : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3047b6d2024a47af8424709096c8b018 : SUCCESS in 5m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b1266015526440da87e074bfc1ed1adb : SUCCESS in 8m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6daac5aa32004b0399b39eaa4fb34894 : SUCCESS in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2186f5b29dfb4fa7b8d9ed9a38ac5da7 : SUCCESS in 46m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/865ea043cd9f40e3b92e71107d11a3fb : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fd9df5f377374d219090347792d48d3a1a1d97e4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 12:02:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6793584cafb54697e4f0690f75ae9e18aba7e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:03:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ade5f791e4cee3cc58240cdad8eb7464d92f930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:49:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f707c069634b4afa86209b2349b54bc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ee75b664c04c0d89b687d31c13aa59 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc5ff1c9b1545b6ad3abd6d45959dcf : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/328c5ddd70d54bfb97df951010d90df3 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1f2677d62cd4972a8da6f3476d44552 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcb035ec17c04b69b13f7f41e993bdbb : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6bb4c2307af7474b88d9c0a0aabde6f4 : SUCCESS in 2m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed8dd48da7004e93b5c3a017f8a2591d : SUCCESS in 6m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/91d3749d93324e7c8d3985e2582278a6 : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/da47b148c47944a38884bca38b09a42d : SUCCESS in 7m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1089b44a713e44968f22e059c4d7b995 : SUCCESS in 13m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/52a97f63e3ef4d2e8dc01371a555de10 : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cbf80d645fe540a9aef20b88d5941618 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e3665659d741496ab5ee766b568e763c : SUCCESS in 5m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0c05513c3df745c59d9f46c800bd1239 : SUCCESS in 18m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a409334b383b49dd8db49c616245e973 : SUCCESS in 7m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b7c582944c654ac0a6854bb8ac3824ff : SUCCESS in 9m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/af42804c067b4dbc97779ad0b108bb19 : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/30455c47f324418f8d6eba6095e48805 : SUCCESS in 2m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/93e704b444cf44efb72f9a03c39b6e8a : SUCCESS in 10m 33s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9e7bbde0b2a644b3b1290519893e889c : SUCCESS in 29m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/143a77f28fca4fd8829a7cce904223b2 : SUCCESS in 7m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3bee10528dec449e8a75b07dd5993f43 : SUCCESS in 14m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d9fe4e9fcb124911b32aa4a01a8a72d3 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66d3e08a986e41aeb87a4a22db91bd3d : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e0fd52a4f414552a70b6ea3f68aad3e : SUCCESS in 29m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8b535f52629145298a94e529f01faafa : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4e0f8a35eb3744949887cd5824a291f5 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aa48631455e4137a8e5bf3ec6ca823e : SUCCESS in 6m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/954f397a433646ebb52476c724d88f0e : SUCCESS in 20m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6e42212eb04c43a98574c6df6c29901a : SUCCESS in 6m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627c9c7d45b342fb82e5e4dd88124154 : SUCCESS in 1h 01m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfbbf036885241b09cd7c00cd57f5b99 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cf3e215b7b1d40222f43dc13ea090e6a9a0f9de5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 14:49:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3828bd8395ea3bc843bc7047553ee4c247272f6","revisions":{"c3828bd8395ea3bc843bc7047553ee4c247272f6":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:09:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/973798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/973798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/973798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/973798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/973798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/973798/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:59:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 10:59:38.000000000","tz":0},"subject":"update constraint for taskflow to new release 6.1.0","message":"update constraint for taskflow to new release 6.1.0\n\nmeta: version: 6.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7daa1a16d499d08de6199bf4434051562f5931c8\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If9421fb91f0c2cd4a3518a1421a527a192c660fe\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3828bd8395ea3bc843bc7047553ee4c247272f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3828bd8395ea3bc843bc7047553ee4c247272f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
