)]}'
{"id":"openstack%2Frequirements~973799","triplet_id":"openstack%2Frequirements~master~I307eb38417352c50256a3e14418f01f76af6ed5f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I307eb38417352c50256a3e14418f01f76af6ed5f","subject":"update constraint for python-cloudkittyclient to new release 6.0.0","status":"MERGED","created":"2026-01-19 11:13:20.000000000","updated":"2026-01-21 15:06:44.000000000","submitted":"2026-01-21 15:05:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973799-new-release","meta_rev_id":"02ddebf644684807fea9f561df73998d15a444a6","_number":973799,"virtual_id_number":973799,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 15:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 13:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 13:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:45:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 10:04:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-21 13:24:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"6bd138825e889d9b1d72811a925f0de4da369c2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 11:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e50a6e6171fa3e5af31f53b62bf4d48f9cd33bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:45:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a01199d763e4425995eead90c6544870\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a269ec3a39124d9d85b2f65e8ceb22a8 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3387a49dc6e94addbf0cd01891f8b9d5 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a40ba49743ad4a36b4ec5dfff6d05dc8 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7393dabb1b9a4ab9b8da823c7bf77570 : RETRY_LIMIT in 1m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/094b6b8f7b514ff8969277abe13742d3 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ed295ef5c7f4d93b3c11c17754200e5 : SUCCESS in 4m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d8624c2077d64dfd99a17f207bc92bb4 : SUCCESS in 3m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c06af119639b485b9eb0db190f5551be : SUCCESS in 10m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/309b463828fe413b93aba608f6dbfd06 : SUCCESS in 7m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/80027c94328a451ab3301067f5a97984 : RETRY_LIMIT in 1m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/419dc93515b5497fb9efaca7ff2218b6 : SUCCESS in 4m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0f7226eb21d9448ba23e1b32a383756c : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/20788c0a44e441a3a88e6726c89d2ceb : SUCCESS in 4m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d783fda55b094678b10941f0a3f15c83 : SUCCESS in 11m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0c586379b9be452c89fc8eb224dbc024 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e50ba9c643bb48ffbaf506d3b862f089 : SUCCESS in 5m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b065d5bd4f174c4c9defe5d9851389b0 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/57fe442bdcb442f0a36fdeed0d652508 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46055e468ce64e2badee399db9672fb5 : SUCCESS in 6m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3dacb81ee2a547ae934aeff0cd591d43 : SUCCESS in 9m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1256430aeb86402cbce89983476f974d : SUCCESS in 12m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e702a06369194190be55b6e3ec1addb6 : SUCCESS in 2m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b433d840c3e84ddea3aa7452d0bb45b0 : SUCCESS in 11m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/59bf17d400d046c9aafd09be3844127f : SUCCESS in 4m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bf38e66772ff4968a9e42b15db4ea7a5 : SUCCESS in 5m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/29718ac8337a4c598cff318ef1cdd34c : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e8ce598411ae420ba401e563d86a4a98 : SUCCESS in 31m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b905b90bcf345cc869bf909dfabe3e7 : SUCCESS in 7m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c8523aa4bba54e4f8e710673302310fc : SUCCESS in 14m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1bd53fe81abe4c9ca3e9e24cc2cc7fee : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/263d40e1df414adc99036ff4585c3c60 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8d7877aa0f44ab4804354ae254ea8da : SUCCESS in 30m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fc29dc23f193420ab3007d62c18167b7 : SUCCESS in 6m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ff01095318bb48f0b6797a841b84984b : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55073cbb79f1431e910186ffdff3c7df : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd686c32dc3e45d28d4e38a50a6d55bf : FAILURE in 18m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5a5d908c766f4cb1b69250ca8337b0ea : SUCCESS in 7m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cce77d0d47f485eabdd16b9b9f2970b : SUCCESS in 1h 03m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bffcdfc5472461b978b5e3926c43a9a : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2a6c2e81ae99fbbc24c44fafbef64124b01812de","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 10:04:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaca6b50120b1f8bb6a12c24014f15a73dc9b10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:13:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/474aadbb65f641df84aa36255b5652fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c950cb0239c3415494378cb7b0b9a572 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b76a76d3d49430e8dece312676cd93b : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85b2c1ef197e46c8905acc1fc2e4c2aa : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0ec162a18284d8d8aa5d4a099475f33 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cf83d5215444d30896197d3eb9439a4 : SUCCESS in 2m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55ac5446f533413797d3e943ac71ce9f : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7e80f6eff5d44e40bd0d6a12e529f16c : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ddae1e629dbf42829c9624efdca4de37 : SUCCESS in 6m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/546d8dd63033439f97460bdb3e15f8ac : SUCCESS in 11m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6fb52611b3414af39b2f25ebaf3d547f : SUCCESS in 6m 00s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6468bbfac3d7470998035acbac2d0ab4 : SUCCESS in 5m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1f407b1d2fa54618a68477dfab04c0c7 : SUCCESS in 5m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d478cdfc8b65496093c9d62e33fd7029 : SUCCESS in 6m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/55590ea63f3e44d3b8cfe4e311f14fb5 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b27ece7d869b4e948368891187189fe6 : SUCCESS in 3m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c04ebca0f9534933b0e6fd8793f996f0 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0041d2b353234241a4dad5cfa4b2adca : SUCCESS in 6m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0b8ab1fda7ac4451aba652297e7f62cb : SUCCESS in 17m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b629285686b477dad927124b2d1225c : SUCCESS in 5m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ecef3b6c3b3045dd98b26c0f20bc76b9 : SUCCESS in 9m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fe7a3adfca7445b2941464b4b306234d : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8f5373a798c947aaac4ce888e95f71b0 : SUCCESS in 2m 43s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b66767872252422ab2199bffa97d5ae4 : SUCCESS in 6m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/62b97d87d9ed4565957ad6478cd4aa82 : SUCCESS in 2m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/182478ec5a4e49abb39f92b2b37eb077 : SUCCESS in 5m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/23f6191cf0a443c48d7d15bc45848719 : SUCCESS in 10m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c7503af679ab40f3b9fcf6d7b73f497e : SUCCESS in 23m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5814f0d5786442e9312db87efd8380b : SUCCESS in 9m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7ce137f8ad99419f93dd1c6b109d684b : SUCCESS in 12m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7a0a33de139949f0bebb6fd9420eeb38 : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7d492d09bf8d4dfa828518efbe619dc3 : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88153758143243bcb9b7ed30c6bca1db : SUCCESS in 26m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/234eca95a5184bff890276d51b18fafe : SUCCESS in 4m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dd4416a4dc524c9eb789eba059ef6e3c : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd1505939af34f208f6a35e62b2c7b2c : SUCCESS in 5m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a0147999ae3441c2add5ecc2fc8fd409 : SUCCESS in 15m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/916d66515a9f4f80ac982b799e3f2956 : SUCCESS in 6m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a61a9284a0474c69a1b9f7a188457385 : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c0a8ba833ef413ba77a3e44d58a619a : SUCCESS in 1h 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fb3ffde6c8f35f7de2baa418b53efc1aefdde046","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 13:24:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77ef3bfc332931cb9c1c3189d90e1ea2263651bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 13:25:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b11a673c8d48796306a4d40aabef77ba86139356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:05:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8b1344186f424ca734046ea04c2e6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff3d034fee847ea99910bc5c3629e89 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6182cf393374485eab2560200c01d814 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/062e4b38f8814e048469fad9a049105c : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/749df042465a44889867aa9e2b8629dd : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76df04c4af1a42878bc81a3b87391f61 : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2a763ac92624a9ca42f7f29a26948dd : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1fd21e164e2843288546f4d3c02b2abf : SUCCESS in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a14c1986518e43a5b16da85da4b0c3a7 : SUCCESS in 6m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d7da9c7bd154940a8e3c7108c8da577 : SUCCESS in 10m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/170a5f35bfec4596a3a2378471f81da3 : SUCCESS in 12m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b93c97b34e044c8db4dd61dc1ba46a33 : SUCCESS in 3m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5e62a5bbf05d42b4a4e0ad12da87e02e : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c42bc73ed22b480ab9a360eb5fb4d2c7 : SUCCESS in 7m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fa19c0c6c82c44df8e596bd93cca9bcb : SUCCESS in 16m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e28af5ea985442fa8f1fb4d09c7bc6b : SUCCESS in 6m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d4e89e1a6bfe41ec9e6022b6ce9dc9ef : SUCCESS in 9m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6f5b748f65c042a4bc58c967c59ecc89 : SUCCESS in 12m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ad1ae61ffbd74a63b0e664c5eb6bbbbc : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4a9d201b0a1647399125c9e1fe5f06c1 : SUCCESS in 6m 52s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/742235006da945aea7bceee129b0e961 : SUCCESS in 28m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1dc25a3961fe4f198b975dc9d5d850f1 : SUCCESS in 9m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/152017be85b04728b0af3bdfc3290712 : SUCCESS in 13m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7e52d599b7ff47ccb5433447d984d783 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56b556160f544e02b77150375fc83c3a : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6cbe76c969a43b8932b1238156f8eac : SUCCESS in 29m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b10c5b41863e496b93a878c50fc06ba4 : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0265650cdfb744778196798761a83b0e : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83c8702bdc834fc78bf0ee4de91e9a89 : SUCCESS in 7m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/37cad76e8c6144be932a8fdb2f35b30a : SUCCESS in 13m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e16c239f065944b0a6dc33d00c131182 : SUCCESS in 4m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c587d1e4e2194e07ab92aa28822fcfea : SUCCESS in 34m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35b3012c223a4e86926a4d1d600f990c : SUCCESS in 1h 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9499b3dddb9c32baf2522ba85bbc75a162350b48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:05:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02ddebf644684807fea9f561df73998d15a444a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:06:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ec68641fe14a02b8cd6b1ecdb934b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b09884c9cc4245bd798a4b6d605d38 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0ee37d38d8b13f5d4bdb9bb337fd21c3c0dd9f7","revisions":{"b0ee37d38d8b13f5d4bdb9bb337fd21c3c0dd9f7":{"kind":"REWORK","_number":1,"created":"2026-01-19 11:13:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/973799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/973799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/973799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/973799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/973799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/973799/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 11:03:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 11:03:18.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 6.0.0","message":"update constraint for python-cloudkittyclient to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: team: cloudkitty\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifddc48c7c38074570f65a2c7031286f454df04a4\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rafael Weingartner \u003crafael@apache.org\u003e\nChange-Id: I307eb38417352c50256a3e14418f01f76af6ed5f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0ee37d38d8b13f5d4bdb9bb337fd21c3c0dd9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0ee37d38d8b13f5d4bdb9bb337fd21c3c0dd9f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
