)]}'
{"id":"openstack%2Frequirements~973852","triplet_id":"openstack%2Frequirements~master~I848f462e1bce51c155c36513390a4bc9427a5f33","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-21 09:42:56.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I848f462e1bce51c155c36513390a4bc9427a5f33","subject":"update constraint for os-ken to new release 4.1.1","status":"MERGED","created":"2026-01-19 15:51:00.000000000","updated":"2026-01-21 11:35:57.000000000","submitted":"2026-01-21 11:34:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973852-new-release","meta_rev_id":"40c101eeceff79c7aae9022d8509d76fe27566c0","_number":973852,"virtual_id_number":973852,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 11:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-20 08:02:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-21 09:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-01-20 08:13:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-01-20 07:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-21 09:42:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 17:38:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 07:46:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-20 08:02:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-20 08:13:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-01-20 20:01:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-01-21 09:42:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"54153f3e4b0070c42098498014cbbbfe02444ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 15:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a80033597d705b974519ea55911cfab89dd7816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:38:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6967a0239b742ff87f69cd730be204d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6f09047be54716a9713e47ed7f544c : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bf4d847fd114dac9826b1cd4ef84a11 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e94fda3fdc4d17b1ce8e33bb6077c8 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cdfcf694ee34cd8b3793a7c601359b9 : RETRY_LIMIT in 1m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9535bab141be4c37bfb7a5af0edd84d1 : SUCCESS in 2m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aff5cca381e141ae9ca789fc0d5319c4 : SUCCESS in 2m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/60963558e4564f4288f32372366daaed : SUCCESS in 7m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2c3e39ef56f647909a10496b7fa9ca25 : SUCCESS in 11m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f0ebba230e6a4e33988764aa3d4a1bba : SUCCESS in 4m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b152e51ab1b443d9a0f5967aea1153ab : RETRY_LIMIT in 1m 37s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/93ccdb83354b419fa6356edfa4d9301f : SUCCESS in 5m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f6552a8eea5c421e969d3015a3feb7ac : SUCCESS in 6m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ca94e2fac74242a386981653833cf1be : SUCCESS in 3m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/373bf1304f2e4be3ad52087d50d1335c : SUCCESS in 12m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/828db97e9fd54363bab44ca0974d4c04 : SUCCESS in 6m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d3fda969fe05447f985e65115942f72f : SUCCESS in 3m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/506ea8d1c47948e88d9856d917a53012 : SUCCESS in 8m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1cbbe088565448eaaecd9ed4b14f5e53 : SUCCESS in 15m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f68cdf297ae45b68395d3e1b81e2e6d : SUCCESS in 6m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/77d12fa353444663adb4ffc98e394471 : SUCCESS in 6m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d7ca8fc8170b407ba1c16a901c62dd54 : SUCCESS in 10m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/36cb1b81a4a54d72ac8b048620dcbc47 : SUCCESS in 2m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2aca564b278d49b4ab115cb2a9c7e2b7 : SUCCESS in 10m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/94fb7db0605c4418b55a56b5d557e9aa : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/42e49ec6e1d84deba17d7e388e4f9822 : SUCCESS in 6m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5429588cdc48486d8f4d96687e9ada4f : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7f7f4de8cf1b420d87f3d12380867755 : SUCCESS in 19m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca406bd80f3c4cffa73f60912a7e5f52 : SUCCESS in 7m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/56f2666b9ba942e9851785e81993318f : SUCCESS in 16m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/28092b88e03e4d85826dcfd4fc96092b : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6dfd6ddabb542d7a494ab7a4e104164 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16b7233cb29046dba47aebfbd2326dfd : SUCCESS in 22m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/54fad18f2e084e658b0a119ad312ca2d : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8d68f11842664954964d3a469371683e : SUCCESS in 3m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07d60d0fd0774446b2ad983a2acd1509 : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/79187cc480234486bf4eda97adb3afc9 : FAILURE in 22m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8f23d767c3154f22b3c1add6b9d925e5 : SUCCESS in 3m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35e6dadecb5a48b58f467018ecdbf590 : SUCCESS in 38m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e827bf0dc8764bf59c30c7b6aa2da4df : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a3daa122a9d089a7546366e197114186b3e91a88","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-20 07:46:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e33df484adcd86ea8119d636c0ac454edee4747d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-20 07:48:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"078312751f6d519bfd7057311148f98812bea1ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-20 08:02:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0dfdd1de444008de3fda539d48622a2649d09191","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-20 08:13:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f4c6ab364340b4fc36ecb18bedd83e725e541993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:50:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bda8b30a4014337972d184a73985a26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/846b3a6b61cf4f6c9373420958439d27 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf86fadf4e34f338c818b83df22afa7 : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51a4b21c61ed424582b19eaadb342b99 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e7c2e4721c44c1ea9d28655e868a665 : SUCCESS in 8m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d80b1d21182245a7b24be218d66cbcdb : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56b3158ea9c64684b70700aea4e5f600 : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7189512306f245d5a4834d70351c92ea : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/540a09802fc54c498ccdb9ffef5543a9 : SUCCESS in 6m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/757525c83cd5428e8561f09b11de1086 : SUCCESS in 6m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d5cb3529c5944413b1bedde4833b93d1 : SUCCESS in 6m 38s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6d6f4a0c6c1446f39b3736fbad25dde2 : SUCCESS in 2m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4781103873f5481ebc28ea547b597e30 : SUCCESS in 4m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/92fbb9f050e24bef91da0f2872a55ebf : SUCCESS in 3m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/464d457df02143779fc0175ff65c0cfe : SUCCESS in 10m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a43673a2f0594d98a0946396dc76f4fe : SUCCESS in 4m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c2593b8817964bad807b7fcd83507322 : SUCCESS in 6m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3851e102e87c43a2b316570fdbcd7761 : SUCCESS in 3m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0a827d083b8b4c74a8be25d1ee554b5e : SUCCESS in 17m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6347591d85d440ceb5d5b89c1626903d : SUCCESS in 7m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/511b607d685249d1abf97c5501da8dc5 : SUCCESS in 7m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1cf099b2fb4b4bab8f89ae4173167e50 : SUCCESS in 12m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ad444ef01e8c4739bd15a6cb07193ae2 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/741523ef78a24a3992a35badab127f78 : SUCCESS in 5m 58s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/dbba66bc9dd84d82807f15f6e685babf : SUCCESS in 4m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/71218fb7ac4f429a920286b7e78ce99e : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b7d8878e78514aa192dd46ae3d5178ae : SUCCESS in 9m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5331fdc8b63f4fc499f4fcb214f3146f : SUCCESS in 18m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/241b8b0e9f0443a190dacea943564b74 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8ffacacd03884658b384f1fc90b0050f : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6e7c2a8544c3441da16158819a7113d5 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd08dd8ca2d54d439a38d692f632bfef : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/849062991bbb4bdbb53de8185c847675 : SUCCESS in 31m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/acc59d436df2487c8060b140775ebe48 : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/30d7b402bb9d4f498a5b5194e428d48b : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b625e76058e84d638729746ce737bab8 : SUCCESS in 6m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/77e1b74796b242a88c496ed8c0af3413 : FAILURE in 14m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/77a54480d0ae4401b031a84945c55764 : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f6079770e7846598dd0f1fbfd955596 : SUCCESS in 1h 50m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff4a53705ee544a39222f521bbef9aac : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fac43666f5497dc5fe3a24ae6a05f24977f37382","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-20 20:01:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"039b8b5c9990731fc85de34db069465c258e559b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 21:16:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bad2de31aaa74702a9c6b0efaa1d43e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7609ecf1f9408ba97664b42c3e68f4 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e26c62a0714f97975814dd02632208 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fe994724c6c4803a456d045e7d49c93 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23b33284c7564a49b7de0ac9f22e9f66 : SUCCESS in 2m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05a24399d77c4ec4a5fe7dae021d9b0d : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eade169ce544425ebcc095baa69241d1 : SUCCESS in 2m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2037492ed70643c2895581df654cc3e2 : SUCCESS in 5m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fffcb8699fb64142ba59b41136d73c95 : SUCCESS in 4m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4dcf2469948241bfafeebbd9ad920ca4 : SUCCESS in 8m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dc8e7c65796a497ab40adac4867f88c4 : SUCCESS in 3m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6972ce4ea51449aaa5773ccfdb897076 : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/79cc5c8dc1ba420ba9608dcf36b39975 : SUCCESS in 3m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/64d97feecd364103b74e5c27b9aa9e3e : SUCCESS in 2m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/77ca0446372441bbb9a2873d000e7755 : SUCCESS in 11m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3da909a5559143c2a884080635a0cb1f : SUCCESS in 4m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dc76b100a0a5442989d0f64d3b3da790 : SUCCESS in 3m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d4236241b2ea4058a79c0057f135451e : SUCCESS in 4m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9950a5c7c89a42a2b86201b102684d2b : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/854b4c9006af4f2abac4f827dbd5fe20 : SUCCESS in 3m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d4f21b8bc0a64b28bf7a21a422264985 : SUCCESS in 6m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3ec2e336403f4c54a1ee5cc756ef1f05 : SUCCESS in 8m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/73db8eb7e47f404c881723cf25fc7500 : SUCCESS in 1m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a0c368a95f72449ea07bcc90738ac698 : SUCCESS in 11m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/288d4986f15f404bb6ddad10614f0c3a : SUCCESS in 3m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6a1137a40cf94d99a74e5cc252c54380 : SUCCESS in 4m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/81c887395a284892b4e399696b96549a : SUCCESS in 12m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/834bcee7273a400d8be0cd17e48017bc : SUCCESS in 20m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/04008012d60f492a91ea814ce32c7ce0 : SUCCESS in 4m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5f4818c0645946a0afdd31796bab7bd9 : SUCCESS in 12m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f3727319d8c046a7b5358c7abf1378d3 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4d7e44a0dd445879eb688f377763240 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf37b80b21f544e5aed1e2d3e5f0b355 : SUCCESS in 20m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/428b3943816a420fb29b7a6c1078f38f : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b9890128b29486cbb2bd77b9e8d3508 : SUCCESS in 3m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/081f73f00b35430d91b63db3d0acb045 : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3f4016d7bd4a41aaa88b955fc15fcfbc : SUCCESS in 20m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b62e3452a70842a09bcb1d6120bb209b : SUCCESS in 3m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded25660f6934aae8d43d9ce9059f70a : SUCCESS in 33m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90cb2c0a7d7541d8818426526dceea71 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"99199b7857ea09cacdad3bb2a55625e5784facbc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-21 07:38:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"046011b74f6a7d3577a334693b8c3e7953cfba89","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-21 09:42:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc4540f75a6e8a8b4476a38f9b56424e99330c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:43:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"927004e27fe28cf08d1363678500b41492ce2f7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:34:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd258de78714a37b7e61407368e4cfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e4620f0e8f4abaa4798311191ae71d : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae43262130b4b94ac6a32625bbd8963 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e7dda1d85ca4ffebaf3f7a6ee7a34ba : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/976e3e5c71524d61a70cff2ffa14865c : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fc4ceb2baac47b48b2b14b2a09b9b52 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45afff03c2c141d895342c06ed7446f2 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a8316b2afcbd4c1a9dfd54780980c027 : SUCCESS in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f204acf6aa1d4d10b70e8873f46a3e59 : SUCCESS in 6m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/12db6ecafc3e4f959e4a98b36f25997e : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5ce455fc2ff34a80af7bda774fa30c69 : SUCCESS in 9m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7a2e057656d243659fe76c4f19616158 : SUCCESS in 3m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e268778eb0eb400ab502ef4906c1c527 : SUCCESS in 5m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7408e0c04d624f19bf98c0a385fe6a96 : SUCCESS in 7m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2e0c201849cf45e8ac28532b26315c49 : SUCCESS in 18m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb4457e3293e4ffd9da5214df255d614 : SUCCESS in 6m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/87b3ac238bc346268f32c5b7ecbbc5e7 : SUCCESS in 8m 36s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/968d4b406b084f729d03d4767da3f257 : SUCCESS in 11m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9cb8135c82e541258d219afa36d75471 : SUCCESS in 3m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3eef1675011142f6bda979ef0217efc0 : SUCCESS in 6m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/437ddc697d3749bbb83913a4c7e1d67b : SUCCESS in 22m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/37ca22d15065496b8f1967ade4662fc9 : SUCCESS in 6m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/159617c350f74e758e964d0c16b5c3cd : SUCCESS in 14m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cb1aac98d7f34c37920434698c4075f1 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd834caacf8544eb942af541e36c5763 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89007034d8f04479b140443a13ea3c55 : SUCCESS in 28m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fbc0f4cca3d9430ab0ac4bbd1ecb365a : SUCCESS in 5m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8545d32511294712b7b5b5528c97806d : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e9bf3ee384c4d81bf9074e6d9516673 : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/755adcef8faa437d8260c40147458d10 : SUCCESS in 17m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/052fb3bc32224aba84c198581f863863 : SUCCESS in 7m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a556838e9feb42c99fa1c5d3abc73971 : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e96e616c22049a997e445d9a2449153 : SUCCESS in 1h 39m 36s","accounts_in_message":[],"_revision_number":1},{"id":"72a099a7f2b4796b2d83492e001ae4cf4aebc8a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:34:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40c101eeceff79c7aae9022d8509d76fe27566c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ed82f556a5e47fcb8f7bb5fd2f5a9d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7186cf4e08704b45b639c4c3c8f558f9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf8eec125b72bd086e2248ee220aef81eea95de4","revisions":{"bf8eec125b72bd086e2248ee220aef81eea95de4":{"kind":"REWORK","_number":1,"created":"2026-01-19 15:51:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/973852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/973852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/973852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/973852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/973852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/973852/1"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 15:40:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 15:40:59.000000000","tz":0},"subject":"update constraint for os-ken to new release 4.1.1","message":"update constraint for os-ken to new release 4.1.1\n\nmeta: version: 4.1.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I301e1355e40a0e0210f643a8f765a81d311ca2d3\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nChange-Id: I848f462e1bce51c155c36513390a4bc9427a5f33\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf8eec125b72bd086e2248ee220aef81eea95de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf8eec125b72bd086e2248ee220aef81eea95de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
