)]}'
{"id":"openstack%2Frequirements~973880","triplet_id":"openstack%2Frequirements~stable%2F2025.2~Ifb5f0323e277cf47be3a1a104f06effbdcdad610","project":"openstack/requirements","branch":"stable/2025.2","topic":"new-release","hashtags":[],"change_id":"Ifb5f0323e277cf47be3a1a104f06effbdcdad610","subject":"update constraint for keystonemiddleware to new release 10.12.1","status":"MERGED","created":"2026-01-19 17:52:04.000000000","updated":"2026-01-21 17:04:34.000000000","submitted":"2026-01-21 17:04:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973880-new-release","meta_rev_id":"dd3a5d49995b399c54112556b9dce6e1e770285c","_number":973880,"virtual_id_number":973880,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 17:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-21 15:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-21 08:21:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-21 15:26:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 19:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 08:21:37.000000000","updated_by":{"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},"reviewer":{"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},"state":"REVIEWER"},{"updated":"2026-01-21 15:26:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"032d3fc4dcc7dd40a3eb2e7f7db12198a251c3dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-19 17:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5753907a9173aa91b2db50ad35f3d94bebe10c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:44:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e7a3b0e3a24ede97ef05d5423b2fd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2da8eb7a5e468dad5848e5d13cda85 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84eb2e17f7d648169d18c7790b99ca82 : SUCCESS in 2m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9e23a05d05d425abb9ea5c1c2181460 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c84c485049e47b4b229a7171c2baf55 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/203e9f0e8c4744bdade4531db3a5bdc3 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a8ca12715a04763b64b768db32a899e : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c88406f05fd49d0a1aa5dc43c4a82b5 : SUCCESS in 12m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b4d2c0f1818480a85ebf2523477c564 : SUCCESS in 17m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8da3f582e22d47bcbaf61e7d102b7e08 : SUCCESS in 19m 20s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ee5d316ac0bf4735820f6f28dcffb5cf : SUCCESS in 10m 32s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f1c1b32a1e374d399bf9ffb460e03d7d : SUCCESS in 4m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/70bd13729df245198e9ca7849de74348 : SUCCESS in 5m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f3101812562848d2a5e70f072f565a59 : SUCCESS in 3m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0b220547a63f4ca9aa69fb703e80cf9f : SUCCESS in 10m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/90be78888f754505a9da1315839569e6 : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ad1adbcfc83e4146aa9be4e733b3f950 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/48951e7da11c4561b8fa7eed687ab31c : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/23e83f255d0347c89be0dbb5788fa2e3 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf5f1521a66542348e86a2871e05c3d0 : SUCCESS in 3m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1d8cf7dc35524432816566f93371be27 : SUCCESS in 9m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d07e44c01d1547d3b964225dd01f7988 : SUCCESS in 7m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c5a1a428b32849639f649f762341c310 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/385d71be381749feb4503c5b144eee76 : SUCCESS in 10m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/086e846cea954a50af37ff73604fb896 : SUCCESS in 6m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/094174d019da4b85881f0e91e960e2f1 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3b092d17003b4e5f80de14f7d1714781 : SUCCESS in 10m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/13fb655428254474bfb6495534c00914 : SUCCESS in 22m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c029c3ee91184ad6bd8f206b3e8560d1 : SUCCESS in 7m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a556b5e790542a7a19e9ab8807fbd32 : SUCCESS in 13m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/47356de37bf44b7eb2748d6bb6f719c0 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc9f296e64a44638a43b9825b126cbc1 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/515dfdfe229345a093df7a175c99bf99 : SUCCESS in 24m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dac047e08ad242389c7234e58c433a78 : SUCCESS in 7m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fdb2f0c1a1944116bba623e48c13a974 : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84ee1b693aea49599987997554d93397 : SUCCESS in 6m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/836290ca5dd244c98b7bb30f5cde8219 : SUCCESS in 18m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/81e1ff70e8e9470c8f72867ddf24011e : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3081f94d88f6453caaa130c0b23216b9 : SUCCESS in 53m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da6b4348bc3d47e2ba1367f828c0d7b1 : SUCCESS in 1h 47m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4b264645f0c90455b7d3d112d35b55d07d01bdea","author":{"_account_id":28332,"name":"Alexandre arents","email":"alexandre.arents@corp.ovh.com","username":"aarents"},"date":"2026-01-21 08:21:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fa4a5529040d7e5cf781de8f0f8158b122b4c08","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-21 15:26:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"072809f0f2b9f7f3130a75d8e70a15e2837c89fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:27:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71246503083431e395db8c95b6cfd0539cce9ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 17:04:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591913dfd43f4f96959ba3a91ac9cbe1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41470f4fa8d9463884aa7ddb3505feaf : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c625ead16384b0aa6e2e78a32acdc24 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4963d79ddc9b4139bc56a656273e2b05 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e229e674526476d87335fbbeb5fe674 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0b660025e934507871a4eeefefca1d1 : SUCCESS in 4m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5560d08eeeaf40a997408aa21fff56e0 : SUCCESS in 12m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e830f962b8f5478cbdfb71af45753bc4 : SUCCESS in 15m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/57821312b66b44149195f00dfea0dd7a : SUCCESS in 20m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5093f3bd7ff448328f7f3468037589f0 : SUCCESS in 12m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ba708470209b4f25848b71f624031879 : SUCCESS in 4m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/55d173b3af3f41f6b54d9d10c11f8a63 : SUCCESS in 6m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/04cafe93b79d4b07a050e4ec46a1e150 : SUCCESS in 5m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bd6f86ca81a3443e954d9d13a3524869 : SUCCESS in 18m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed131729c0ff460a850156c6a336d889 : SUCCESS in 5m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e7d71328ecf1407eba4e6416a7b10d93 : SUCCESS in 10m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8037546501f742f795463a85bd740878 : SUCCESS in 11m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ddba65a63c144da2816bb8cc7a8a6b02 : SUCCESS in 4m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fdba0c2dc5504def83455b895ddbadad : SUCCESS in 10m 31s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5a2c06981749493a9e5e5490541199d7 : SUCCESS in 23m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/310256e537aa41dda4ccea1ab462943a : SUCCESS in 9m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1afe5cb9acc74a27a605dfd456706617 : SUCCESS in 19m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/602ef06ca20a4fb092ee6eb0d0150ec7 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/acb36f4f634440349006a3335539739c : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84db23641f454b61988e5b2b3111f17f : SUCCESS in 29m 03s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cf7dd52499c744c3a42383b13d489fe9 : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4312e2bf3b8b4bfcb760f05df80966b9 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b94cef3a8db43a1a4505955df7b3268 : SUCCESS in 6m 14s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3aca6d3a613c456bb3c9affabc8b2d17 : SUCCESS in 13m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0517181a1daa47c5b90c48aaee652624 : SUCCESS in 5m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0ec0237d74b40f6a956f4ea83e4c852 : SUCCESS in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaf41b9e0d9b4afd8ab4c731d67b8776 : SUCCESS in 1h 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dd3a5d49995b399c54112556b9dce6e1e770285c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 17:04:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c0507195b43195d9b4762a179a4bbeb1172fd31","revisions":{"5c0507195b43195d9b4762a179a4bbeb1172fd31":{"kind":"REWORK","_number":1,"created":"2026-01-19 17:52:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/973880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/973880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/973880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/973880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/973880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/973880/1"}}},"commit":{"parents":[{"commit":"7d16424510c49daa913fc20a6300427b72522231","subject":"update constraint for openstacksdk to new release 4.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d16424510c49daa913fc20a6300427b72522231"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 17:42:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-19 17:42:02.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 10.12.1","message":"update constraint for keystonemiddleware to new release 10.12.1\n\nmeta: version: 10.12.1\nmeta: team: keystone\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: stable/2025.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: If34fa0a33ab6ad1d4d250588fd07b9d9ef6034d4\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: Ifb5f0323e277cf47be3a1a104f06effbdcdad610\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c0507195b43195d9b4762a179a4bbeb1172fd31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c0507195b43195d9b4762a179a4bbeb1172fd31"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
