)]}'
{"id":"openstack%2Frequirements~974081","triplet_id":"openstack%2Frequirements~master~I92e7388461105d0e9880b3bee7f27fa9889d1ab6","project":"openstack/requirements","branch":"master","topic":"pcre","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-03 07:14:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92e7388461105d0e9880b3bee7f27fa9889d1ab6","subject":"Replace obsolete PCRE packages","status":"MERGED","created":"2026-01-21 15:03:40.000000000","updated":"2026-04-03 07:15:33.000000000","submitted":"2026-04-03 07:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974081-pcre","meta_rev_id":"95c5144b400d081f842cf20ed658b1858a2840a9","_number":974081,"virtual_id_number":974081,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-03 07:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-01-21 15:11:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2026-04-03 05:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-04-03 05:32:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 15:11:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-21 15:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-03 05:32:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"3bffb1837c1d04fbee2cae9cc13b7dbcfafeed1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-21 15:03:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03e8c2976b8fbd73ea9084571b024784f47e6fd5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-21 15:11:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"862b397adcab780d9ec3d22ba286543d2434603b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac5cb6a6a214d559fb00a52549ac352\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1d7554cd754dfe971c273b974aafc7 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72a7dc0c9fd04512b7f588813340576d : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc52f3dc85bf42ba91e1ffadc03551fe : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa1d951a288844eb984d123a5f9ac3b0 : SUCCESS in 3m 35s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/731fa24cc8054c9aa31f38983b47a9c3 : SUCCESS in 13m 51s\n- build-wheel-cache-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cbfd9002ab641538fdc6359fceb28a2 : SUCCESS in 17m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae1f7e0a82274eb3b30d6b2134b4b22d : SUCCESS in 3m 38s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/67fde05f70eb4dbbb83ee2c7922f18d0 : SUCCESS in 2m 43s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffa07e195dde46028ea2de3cfa8d064f : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3bfe1abd73c29345ffd50e0df05573bfcda407b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 17:37:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d212cf32610d4725b7ae23e38a945e34\n\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/c55d4199996d460ca218c48cbc0f4bbf : SUCCESS in 21m 42s\n- build-wheel-cache-ubuntu-jammy-arm64 https://zuul.opendev.org/t/openstack/build/d00b020ea45b4b41bcd3db46c87bc3a1 : SUCCESS in 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"21e0a739d74f656e2c54f9991c5a600d215fe78a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-03 05:32:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e48b8908a48a5ad9d8840a75e3f543121f19bcaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 05:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b80fb7e33ee7d70f9432b9423ca8fd2ca01802c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:14:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4da2997af654215a398d83f29a065c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b78fe4fc5b48b0a94802df4c0ed133 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126f0ec09c7b4ba59746bea87352248e : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0145d57622484d259ef1ba795231f7bc : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e04aec99ce14c979c9540231be74dd5 : SUCCESS in 4m 05s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08fad1f048484b0f840b0beedc668270 : SUCCESS in 24m 33s\n- build-wheel-cache-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84e8587a1a054d859ced9e3afcc2197c : SUCCESS in 26m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8abc02498d7c4109a67fa9606929cccd : SUCCESS in 3m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efc70ab243ce489b829ffb5f27997106 : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c87d01000f67da1bb18bb5fc356e88731d400549","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:14:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95c5144b400d081f842cf20ed658b1858a2840a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 07:15:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe896973481b44cb8ba17c631b5655c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd50565cac643cd89e52f9901fce8c6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05229e6d5576e5f16363730224962fb81ec05d49","revisions":{"05229e6d5576e5f16363730224962fb81ec05d49":{"kind":"REWORK","_number":1,"created":"2026-01-21 15:03:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/974081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/974081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/974081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/974081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/974081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/974081/1"}}},"commit":{"parents":[{"commit":"880e38a2d1deb39ae49e42430301287632f9535b","subject":"Merge \"update constraint for oslo.vmware to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/880e38a2d1deb39ae49e42430301287632f9535b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:03:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-21 15:03:27.000000000","tz":540},"subject":"Replace obsolete PCRE packages","message":"Replace obsolete PCRE packages\n\npcre3 was removed from recent debian-based releases (eg. Trixie[1]),\nwhile RHEL10/CentOS Stream 10 no longer ships pcre in favor of pcre2.\n\nUse the latest whereto library release (0.5.0) which uses pcre2\ninstead.\n\n[1] https://lists.debian.org/debian-devel/2021/11/msg00176.html\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/971428\nChange-Id: I92e7388461105d0e9880b3bee7f27fa9889d1ab6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05229e6d5576e5f16363730224962fb81ec05d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05229e6d5576e5f16363730224962fb81ec05d49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
