)]}'
{"id":"openstack%2Frequirements~974322","triplet_id":"openstack%2Frequirements~master~I6629edd31a4fa6c0c330e1214d85f26d275bb393","project":"openstack/requirements","branch":"master","topic":"I6629edd31a4fa6c0c330e1214d85f26d275bb393","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-06-03 11:27:18.000000000","reason":"Change was marked work in progress"},"38274":{"account":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"last_update":"2026-08-13 17:16:19.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-08-13 17:16:19.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-13 17:16:19.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-04 07:27:41.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-15 14:09:53.000000000","reason":"Change was marked work in progress"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-15 14:09:53.000000000","reason":"Change was marked work in progress"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-06-18 13:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-13 17:16:19.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-13 17:16:19.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-08-13 12:59:01.000000000","reason":"\u003cGERRIT_ACCOUNT_6476\u003e replied on the change","reason_account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}}},"hashtags":[],"change_id":"I6629edd31a4fa6c0c330e1214d85f26d275bb393","subject":"Add cinder extras hpe-storage-flowkit-py library","status":"MERGED","created":"2026-01-22 14:01:24.000000000","updated":"2026-08-13 17:17:43.000000000","submitted":"2026-08-13 17:16:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":106,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974322-I6629edd31a4fa6c0c330e1214d85f26d275bb393","meta_rev_id":"91f9b5ca4deb296cd2fc564a70c5e8b8f3644deb","_number":974322,"virtual_id_number":974322,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 17:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-08-13 11:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2026-08-13 15:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-08-13 12:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2026-08-13 15:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 16:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-02 15:01:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-02-10 14:50:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-18 19:19:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-14 07:08:22.000000000","updated_by":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"reviewer":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"state":"CC"},{"updated":"2026-06-11 05:51:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-17 14:51:13.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-06-18 05:30:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-07-24 13:48:27.000000000","updated_by":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-07-29 14:41:08.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2026-08-05 14:22:32.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"c7a7248c46fd9f111e0cb65e6818e8b5be29cc2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-22 14:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed2c2450d99cdaa2376320ab90923385b7259bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:12:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1af70c5b408647848eda45214ace63bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf60e9b46414abd851cd42faaa28916 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67604bf32791442887d07b27f1730b4f : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dde8173d1084b26875310b676dc300a : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35d995b6e0124aa28aa89fee265a7c7f : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a43c91c3986749a8ad62bcf96c3d6d3a : FAILURE in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8aa4f46316d24ab288364846bd3f7503 : SUCCESS in 5m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/89994b659a574d278baaffa4feb4093f : SUCCESS in 5m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/407b1ff4c0064a7882f94b326e4a198c : SUCCESS in 7m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c9b7fad54af149098a10a09433ad8b19 : SUCCESS in 4m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a6db14b30bbc4ae3a3d3b6f6ec2a0c75 : SUCCESS in 7m 34s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7e248d4362ff4af29b2673336ed58f35 : SUCCESS in 5m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7077eef7e0a7427dafb57a85d7a114c1 : SUCCESS in 5m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/824cb0defe144b1288535d550b6e6518 : SUCCESS in 3m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2e87909001b34745adfe8faa2ec14c3c : SUCCESS in 10m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a348c4fbb84f49929fc0bd98b2d6bd5c : SUCCESS in 4m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b6b7b76894c04fad860554a454647c1f : SUCCESS in 6m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ddea262c7b9348caab66e7bbe062d95d : SUCCESS in 7m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4bc2724712964bf9b0b3de466380326b : SUCCESS in 16m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cef2b296ee3b4117b914eafcd42fe779 : SUCCESS in 8m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c00aa167ef58430a9113205e9340fecf : SUCCESS in 9m 31s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9a381ee3492e4dc6b9d9e79dce3e6375 : SUCCESS in 10m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/24a1b86d72854dfb89eb94a7f976ffb5 : SUCCESS in 2m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0288c3537aa64555b4c959a67e951f36 : SUCCESS in 11m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8fd62d15d1274a95bfb1e43d198427d8 : SUCCESS in 6m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b1f8d370f19d4eb7a5590a4d2afc607a : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fc401de09541455db24208f6db060ce5 : SUCCESS in 12m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3861ef94b3e04b51a84708693cd8cf25 : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fb961dbc77a042ebb83e0cfb5725eb9c : SUCCESS in 5m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7cae7a36cc6e4d6782fa3ef66e12bd72 : SUCCESS in 12m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f6552a2cc64946008e4a250605914cc7 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/145d8256decb4fb6bf027fcf45f162ed : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f3ba6a411ea4f699c7c9ec66fe6da49 : SUCCESS in 23m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b636259a1d7b4991a92a90bdecae21cc : SUCCESS in 4m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b71ca8b3bef74f5c9f9d08baae9293f9 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5db7b5b6bb67417e81d42d2332cbe4a2 : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9072d6fe43de468fb6b975a386d1ab05 : SUCCESS in 14m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d021ec3a77564332a2a8c034658b5f92 : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/753bb4f01ea54fadba95885e4745d352 : SUCCESS in 34m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ad339a0d3aa46018614de55ca4f3ea6 : SUCCESS in 1h 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e14e8f6be492e53eb98e14cf7233bd7389d3e6b8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-23 02:55:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4c9bcffc8c1e0364cf7c29e10ebc1bd53bb57ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 04:29:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25c6c1698ec04257834c78152f039c6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb00213fbe2477eae6476b192e58c23 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769524c11712486989bb149c8a421a86 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53d871a52cd44430b05eb57eab90d60e : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a44ec3ccef6e4024995dc31943ed3ac3 : SUCCESS in 2m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/373915743e0141b9b700cfb9be748fd5 : FAILURE in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60db26f92d8f40db80a377e0c1a6b6e6 : SUCCESS in 4m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5488391834e544dcb3bfebb288dd44af : SUCCESS in 4m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e0d4d7da53cd47af8d5da34245561778 : SUCCESS in 6m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/aade55b7abfe40f28df5ccab820e962b : SUCCESS in 6m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2249e25f3f9a4cc587e1fc1199f466bb : SUCCESS in 5m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5fc494152cb548e8a57ed34709cc8def : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b14bac939b4c4860a5ad82f3dbe4b319 : SUCCESS in 5m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ed2614a664c44fb79291a0c1d98f09d3 : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/249bebd6651a41bf8fdca9530f707d11 : SUCCESS in 10m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cba52cea3ce34c87aa13302723e14807 : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f679ae0c36a04e2ebffbf7d32e008152 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3e8ec5dab8a547e089bd1b7f7af6b2e9 : SUCCESS in 6m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/24db4df3afab4856b0673f02567a4f77 : SUCCESS in 17m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7131799641274ac6bfad77c7e64c85b6 : SUCCESS in 8m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa60eb414afc462c951724604d95ad66 : SUCCESS in 7m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/996a9886db5f4c36b3fc15731fb36ac1 : SUCCESS in 11m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/566f3f8668ba47a9bc25d1b238ff45bd : SUCCESS in 2m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b5f3779e93c941198b510091cec98378 : SUCCESS in 10m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/133a0e8ff5ec48ff961f554da70ad9f6 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c67fe9ff93c0457ba7d23190424c3cbc : SUCCESS in 5m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/503da5b46fd0461e81002f34ca3f45ec : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2abb140ad3de4d1ca676b6a0d9c0461e : SUCCESS in 29m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/06d3f27aa4194501b189039a6baa342f : SUCCESS in 4m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b34484a9b56e44338fe040f9d2d7420b : SUCCESS in 12m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e11ec15b37e64e4497ba41d388ccaf3b : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/673dc90b680a45f4bd7bcf1972a757e6 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e7def4a3bee4b6aa70cfee70b77003d : SUCCESS in 26m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0191b4c7adec43a4ad3496cd70790a13 : SUCCESS in 3m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/48eeceeb657344619f2a75603775ba22 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c0fa765e2c44309bd51d2ce16b32492 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e2408c60357d4a0988218f2014b0311e : SUCCESS in 14m 20s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f240bc5be63746769c2c4011f00744bc : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ea370c641e145bda511ecc9107268c2 : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/335d9516f0a3461195891f9a62b19b8b : SUCCESS in 1h 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bc90e530340f5230b2bc3567df2da3518d4bc331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-23 04:50:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac7eb05db4ae4e1487b90bc46bdc253d8fe901e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 06:48:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9be591ce4eeb45fbb8d58344b1983d96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec919490c4c4ecaa74b4faf931c6861 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cca78fe8ea249269ea8616a9fd1c619 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f17a445675f4cf993a07cdfc7a140c5 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbae9f4e727c4679b11ecc00ce671a92 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06d2ea75740b4a06aab24f14ce35418a : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a76feb86cb164cb7adf8a6ce9dea7961 : SUCCESS in 1m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3dcd5691371e458a8de093434a229dd5 : SUCCESS in 7m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/292bd2d896b249428e9e4c4f00ddec45 : SUCCESS in 6m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fd8a6c8bfd7e4f16998007727b7d87d3 : SUCCESS in 6m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d81047823bee4294ad2e98ef721b792e : SUCCESS in 5m 45s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a33682859f244bc4bf7c001be6b5340a : SUCCESS in 4m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4108b81b474741ff8b952a9b247b8222 : SUCCESS in 4m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2f78a3d8723e4e98b429e048e1b23d6e : SUCCESS in 2m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d08b9737561a4524b3a8674c1ba525a5 : SUCCESS in 7m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/34ad3bb669974ffe8726efee24cb1b2a : SUCCESS in 3m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/18efa50d7a6d49f2a8bf9b8f6ddac9f0 : SUCCESS in 5m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/997b73e8f6734bb4bd0e582a4e4d950b : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b1e7dfb14d8f405882de4fd7f81e202c : SUCCESS in 17m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/294e3531f034451d865b8e9c4e4cca67 : SUCCESS in 3m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d989fc878f684cba823c526f04fa2d09 : SUCCESS in 4m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/61f4ec2b07684db3831481056c0bd6d4 : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fbb162fa4ad848f59e89dab5d089b4bc : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bf8e7258da004a788b58552507b3049b : SUCCESS in 7m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9e03104223cf47d9b76ee8bf4870dbe1 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/99f7e5c4469b4516a69bb7b427a1805d : SUCCESS in 2m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0c88d110f5c7409bb0b68db9a19bc3b8 : SUCCESS in 12m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fa03b97fcb914e32ae79eebb3a088862 : SUCCESS in 30m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9d0eeb81316b42b2b82a9858ef8463b3 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0214987b63a540caa2d9722adc0f0e27 : SUCCESS in 10m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/078e50aff61c40ad844e0dfa2c95bfdd : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30b801cf4467451dbe9a18ef12c738e7 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c878aa6a28ad45e0aecd072d48ba8718 : SUCCESS in 25m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6bf437cabd9d40f19fcbe2efa164bcb6 : SUCCESS in 6m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/69b51413224946cfadfad17a2e7d89e8 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39c74af0c61041b29f7c6ae7ded0c271 : SUCCESS in 6m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c5567d9f60f24a7885cf22019958726f : SUCCESS in 18m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/39648b7a219a4d73a13212ecb500e5ea : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24daab0b4ec843469e0ef1932f1dcfd1 : FAILURE in 1h 05m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c0a67c061024d3abff09c9a10dbdafb : SUCCESS in 1h 52m 30s","accounts_in_message":[],"_revision_number":2},{"id":"08243eecedb793d39064a8dbe981628171223a2e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-23 09:29:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdd7aa39d80d35eaa6df243cf8178c3f4b0d36d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 11:30:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6cf2b2cf1ea4b4ba9a54ea5340805c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a299e6c0de7b404a96a800004460b91f : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd5c607ebe34183a77f3d6eeeec32aa : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0956ea4cb5ea4883a465f937bf68e4b2 : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6214ecf4439c4b4792360d866d6e4d43 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daf0fd92953e46ff95c6fee5c30edb91 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f15036533acf4329bd64a7a17ce9885c : SUCCESS in 2m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef0f6682f6ae455d9abd3e235ad313e6 : SUCCESS in 4m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3577a4240af43178a182a5882d4ea23 : SUCCESS in 6m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/491eccc3c82b4d5ab3e8a2d4224262f6 : SUCCESS in 6m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d6ee8a851aaf4e108a8d3d6735474083 : SUCCESS in 8m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ad5c065c9d3c48c6afb68ce78e4bd440 : SUCCESS in 5m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d5ef770b4f574bda8da1cebc3f1ecb2c : SUCCESS in 4m 53s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ad7f6d8e7328443086909edac38a39e8 : SUCCESS in 2m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2805841f48f24ede85b37b30317ba97a : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5e9c394c1c094589ad36cdee4222b5f0 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/45b0d30bcc774795867abac3c14e9873 : SUCCESS in 3m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/368cbfb0aef842d8bf11c82fc1f28318 : SUCCESS in 8m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/86218f3582ba4a0497a658556ead9218 : SUCCESS in 17m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/359bb6b52527411283b917e4f8ee87f9 : SUCCESS in 7m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/34704ee1d2954b1fbcb0dd9136ec0346 : SUCCESS in 6m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bcaf08a25a174a3d998023008372af94 : SUCCESS in 11m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cfa9462a80a94b61b494182b32edbe22 : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5fba003bacff4332890db500b5453645 : SUCCESS in 9m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/adde7e82b1104278aa50a20b4faf4752 : SUCCESS in 4m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2c2f04dc9ba34ba0af3d32ffba0f8dbc : SUCCESS in 6m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b3a391e8f5c341a9a4779644059fc6a4 : SUCCESS in 12m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0dd9000f4b6a45028205f6a76f8ac611 : SUCCESS in 31m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a73974061633496a9be59ea226d34781 : SUCCESS in 3m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c20a2c012b304d72b2068e1ad1006ac4 : SUCCESS in 13m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea500ef6e8e647ed956c299989762409 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c1300bb57304815974c7fb1d1929eeb : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a357bbdc88074b48a88ea2ae9359cd94 : SUCCESS in 29m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8d640e8f716f46b48de6f964e233b70c : SUCCESS in 7m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fbe67d661538495594fe6bca1e2a3645 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de6a1104c3c8451f9a9a933faba66623 : SUCCESS in 6m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/20d45274569b4ed8a59be5d508a1b7b9 : SUCCESS in 18m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d212fd61ff3e4fa58e3b9a16d52593e1 : SUCCESS in 5m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19548339b62c4836b073fa1e808abbbc : SUCCESS in 1h 08m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc15d7fb41204076b48f50d5b8641db1 : SUCCESS in 1h 55m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2eba34faba63ac2847857cc4ed101e71225cb9d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-02 15:01:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cea4d243c04408317f36f4b9db7c65f1dfd13b43","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-02 17:02:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9dfb1706017d4d59ea4b4c13bcbde81d4d0dd88","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-10 14:50:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6d9978799ff4a9824891e1e1a9573f87ddfc0f28","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-10 14:51:28.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"f408f2206920cefefb2390c596346304fe4e4ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-11 10:15:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"24cb1898af63d203e3bce5aa2fa62150a9b5ce8b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-11 10:20:55.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"05d800d3514907fd747f52e1dd5e28fc639cba3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 11:08:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44d93b7f3105421883e3da98326bee56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2d9d61e41a440ea256180fea440d65 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134b59376b0a473a8afdfb4fcd2a95b1 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b9a245c84d84bf5af0436d68552a6f2 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0663a51f0bf44d67b842d01cc9b75007 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae52688a9443434893cd0a520ba5e997 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f28f2981f80543569a914016aeb349d6 : SUCCESS in 1m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1dda9b9701424c1b98ca31985511748f : SUCCESS in 4m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/80e2a004a7bb4ae09887f1b30a2016d7 : FAILURE in 4m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cfb2adfdf9a140c39d31e21e167bdb5d : FAILURE in 2m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bcfdb85fbdd84f299a45430656eab1cb : FAILURE in 4m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0095e5f47dbb4d0aad4c98f693df472c : SUCCESS in 3m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e5b3e480ffa64bdb987d3ceb7ecd4831 : SUCCESS in 5m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e4f6136ba80e4fc09ee2f8b6bb26c6a1 : SUCCESS in 2m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5bd6ba3325a740e9bd575b4bbf0f4f25 : FAILURE in 5m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4fe70fb231e34dd3a01d9d5e225f6308 : SUCCESS in 6m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7b406867701b4a97bdd9e03243e74869 : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fc123c2a47c549d0ad486bae40c589ec : SUCCESS in 7m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/974b9a2dcf1541edae1c0cdfbb68e887 : FAILURE in 2m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aef7bdbc26214b6ab186dfa638093407 : FAILURE in 4m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/413afb9bf4d0440487d387589a7c7647 : SUCCESS in 5m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6c1ded14ba8497b9fe9a5da50c1362c : SUCCESS in 12m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ddfb55a4f7294dfd97ece8be9c17f5fa : SUCCESS in 1m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/187aa47f8f124c9385b177b0b444e5b1 : SUCCESS in 9m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/82aa9fc514e2462b9932eba4f1338005 : SUCCESS in 5m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/443420db17094022bf9fbe33d606354d : SUCCESS in 6m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9817f3586f4949c285b4dd2c54a4cd66 : SUCCESS in 9m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a2918000a7ce40838b3cd3902b0f3e3c : SUCCESS in 30m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d3b6b3b54b754c53b53aedbdfcfd2c75 : SUCCESS in 7m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/191a1cc05427463e9b2a8a5a7bbb2bba : SUCCESS in 15m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7932ac6749cd428fb6d6041701ef128b : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72f3846a5a4c4a9aab1da46bb79747fa : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a8fb8d28fe64b5dbce770c2dc37ca84 : SUCCESS in 25m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/18f0df19f7cc4a35aed9ff264262263f : SUCCESS in 7m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ee5b6a70727d4ded9d5c7d771c7fc457 : FAILURE in 3m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3263448e236744b2b38a05791b056656 : SUCCESS in 3m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/53ce0950a2764aa1996bfa297ca309b6 : SUCCESS in 22m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/661ee3b2c7964c628e4366ee19d3d601 : SUCCESS in 6m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3603969eec5b49c28e01d29e2882cd2f : FAILURE in 50m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5071086f2e6e4fd08d96ab78dd36b1e8 : FAILURE in 25m 19s","accounts_in_message":[],"_revision_number":3},{"id":"50b9eb9aff1a7488a735cd565c15559ea8b2710f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-11 11:47:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e4b9683cf9fe30383c84da545067e0cc295d003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:58:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e1e0b8b9224402fa577c612f7578944\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134f73cae42d47f9851e94dcfc65a018 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ab1495bfa14ffd8643b9b94b71a458 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43242a7998854d6282cc557729d9b647 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a3d889eb81d41ce9b4241da7177cd76 : SUCCESS in 2m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f85a2226596043b38419457102f94952 : SUCCESS in 2m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c8e40098cf24fc3a6ded13ef7affdcf : SUCCESS in 4m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d6fc7db5208942b8bf9188b28bdacdb0 : SUCCESS in 5m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/962bf35651e64aeaa49c6b3e3581a641 : FAILURE in 2m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8277010edb074e319a8942d42a70196e : FAILURE in 2m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cbd0c333c75e4a09a8a059538cef4a25 : FAILURE in 4m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/662058513a5a4ccbab4188ebbe217a45 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6017306adc4248f7b78338f9a3757796 : SUCCESS in 3m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c9b844aa40e34fafa782932777f58a89 : SUCCESS in 2m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8a3c810d13b24aad85d46b724793dc64 : FAILURE in 5m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/28031e8a61b34f32a83886737ec24099 : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/128d78de11a04c07ad63a915eec78fc5 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9afac18e93184dd48f780dace420b4c6 : SUCCESS in 6m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0539eeaac6e3475a91e840c73a164ebb : FAILURE in 2m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63d08dba98d94e559f8bdccd7851d2c0 : FAILURE in 4m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0139ce3e98d34e979bac78622db775db : SUCCESS in 6m 48s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b5d1d5978a124950865f9cce77217978 : SUCCESS in 10m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2a965feb872a4f28acc10e36da507cc1 : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/24d869c1865648e9b3d9f49c876b50c4 : SUCCESS in 9m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/182b2112d67e4016942f66157a3ba982 : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c91fe52e277e4ac4a97a16f57c2978bc : SUCCESS in 7m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/227d8788b2004d548a312d15582862c9 : SUCCESS in 10m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f9695fc5c64a465c8418158930c7acd9 : SUCCESS in 33m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1271af5acf224175aadb3351e68ed6b4 : SUCCESS in 7m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/67a4f9df608e4d6ba3cbea169fc836d2 : SUCCESS in 8m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1e1b648f733e4ef2bb34f72e260f0ef9 : SUCCESS in 2m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6fcd4b0ce0b8436ba641a98652e254ae : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c40df183d16c450a85ce92d111c9ad34 : SUCCESS in 31m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/24c8b30a20554451a5e9b348f57b6854 : SUCCESS in 4m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5860a0ca0bf94df09d345bb7ebff5d41 : FAILURE in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43b8296e8179417c8562b7ccd7774c9d : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d70dc1ea492a4b6caf6f5283f6d9a035 : SUCCESS in 18m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/70168804896d483cafea59e3ac27f7c3 : SUCCESS in 5m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55e6421edeaa4586b7daeb4514332993 : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70fb4cabaffb4b05882dd05eb191923b : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1bafe9823284d3c685511bb22c42e814e623e989","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-18 14:42:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a3d4ce100c59530788f842a9e49589466b69bbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:33:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f6788abb5647c9bc82df55fb03ccef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5ad2ac7ed84401ab4499f88be918de : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7884d3a23c43449af8e1b5cccb6098 : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f94529c3719a44909bacd7743771b26e : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88555db2289a4a18a1eeb96553d741db : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0bd15e42e9c24d0284773eec643b15ff : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a5b5b10ac5c4b539b64810be75478af : SUCCESS in 3m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/11be4ce5192a4db78ba7706af669b50c : SUCCESS in 3m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f9cafea5e7294bd0843368c182f5cc8d : SUCCESS in 2m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2326e3eee38e474393571b5e24029233 : SUCCESS in 9m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/128cb32ce2814132a38f216d3fb9751a : SUCCESS in 3m 22s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f2bfcc3cd119413c96994c0a80c40534 : SUCCESS in 2m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0a3483b20e6b4d92a63a9f7830f2a0e8 : SUCCESS in 5m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/01f7f03de0e74acbadc3e67c15a0ac7c : SUCCESS in 3m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c832d2c3191743ea8da3adff5e9393fa : SUCCESS in 11m 29s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/19bac81746824e63b1c14bb7c22cf57e : SUCCESS in 4m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/43814c6816d447c0a775362488c84851 : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9d3fb2d4df23478aa875a31d6b2d5371 : SUCCESS in 6m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1862a66357394f0d8c9a0573ecab3d64 : SUCCESS in 18m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/120ec2e236764178b979f8a8249e03d5 : SUCCESS in 9m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/76e3e2b31693492bbfe176f6ddb8a800 : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aeadda6f04554840ab6840391deb497b : SUCCESS in 9m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a82829180039459381e4c08d65c25aaf : SUCCESS in 1m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/313f19a3036b48e089f0fe68041edad2 : SUCCESS in 10m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/12d826f60b374b998c225e1c8d0a6a9a : SUCCESS in 6m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a3e7e289b8f64e92ae5542bfa4645719 : SUCCESS in 2m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/58339e8cd19c453cb508af9bdc53f3ca : SUCCESS in 12m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/92d6eb50ad1b4c218a67751f47a3761c : SUCCESS in 19m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f5c7882e02f445dcbf55193f4550eba7 : SUCCESS in 8m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/47449957caef4b639d02ea5c97d8488d : SUCCESS in 16m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5d39a9bff80d4e7d9b1464229ec02d1d : SUCCESS in 3m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba5bddd106874cd086edefacbe85df04 : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e01c4989e00c48b98e366c2aa7f372ff : SUCCESS in 30m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/184618a2c40548d4a5d06f41a643d069 : SUCCESS in 3m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6da926d4532445f48770abbd91f6a63a : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b6a0344073945eeace56fbecb2a9c9b : SUCCESS in 6m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/56c70b3275964f35932d7c5237f560e3 : SUCCESS in 17m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aee4aa8b356045aba957012fa8d0b5fa : SUCCESS in 6m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec1fd6d11b804cf29f05999af33022f4 : SUCCESS in 1h 02m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/acd67c785e1b43a2bae2e79f9076e78f : SUCCESS in 1h 48m 52s","accounts_in_message":[],"_revision_number":3},{"id":"4256e2dacb23b0bc394cd3c1959391bc7b2e6f7a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-18 19:19:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e221f0b3e085e4f752fbbf8b707c415d1b556548","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-18 19:34:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c95ed38745d357b5627ce3509c01d5eba6ea73d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-18 19:36:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79b855d1aad8e8f453c881402ab152274d948f1b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-19 12:17:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e187d7d677ccb4f9851e95efd2f040c282487d3d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-19 15:44:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4501703a30ca033141ddaab4a466671ff4bfd27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-20 04:11:35.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"363c24106fd9feac280ce5386ec2ba6a86572517","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-20 04:13:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b36fadaa7c8494b4b8d6a5700d68068a266b3a3","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-20 04:13:11.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"e3c0d6c6ddc3364672bd343537ec30f9203cd61d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-20 04:20:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"842649c9acd191d8e8835316d0ac201784a4e23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 05:39:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9da613e69d94f38bf6c1553579fcc9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d029d08764547c18ce4a2c7b14bfef4 : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cea555642a444a802af2e89939b471 : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c384819223d4686a93257cf662636fe : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4186a16476a43de84b1926ae323a314 : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd717e4114a94604ac5d05432bdf0a30 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64e65874190e4ba1bd49db2555d9765e : SUCCESS in 2m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c6811163eb8f4bbbbff2f98abdc9d18e : SUCCESS in 2m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e3687b81ea334496a666cdae6fde3a61 : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/37f0ec55560249f1850bd1c7bf664c5b : SUCCESS in 7m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1de3e10a38c64024b12d1ce7c48b3f46 : SUCCESS in 7m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9180f320efca4d3784c757baf375028f : SUCCESS in 4m 54s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e64590c9111447cbb455217abdb00a93 : SUCCESS in 3m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/881d5d422ad74ce191cddd8c3d364fec : SUCCESS in 6m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a5deb8b792d8495e80ece5aa8c2511a5 : SUCCESS in 2m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8470b2a46f314b29ace8133306de1979 : SUCCESS in 12m 44s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8cbcb59368db49bebb709f9cfa9dd7da : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eba79e48ba4e4e4187cfc468b2c9d0eb : SUCCESS in 6m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4e807a75e8d7469bb6b87f078db41d10 : SUCCESS in 7m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/73aa7a4d7c974f8d9439488b66a81e62 : SUCCESS in 17m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65d57352f23e4c1b93d9a1d33b05410e : SUCCESS in 4m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ecf6c647b87441f99968cf00eda422fe : SUCCESS in 5m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0d2b0e0c596e45a3bf163eb78300b8db : SUCCESS in 7m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/659af80c776f4bbea2956197ba2f4aea : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/978765c6757b46d8a34b6d99d89fcf02 : SUCCESS in 5m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/25777bb995824257a99034d167e073b1 : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2921fc60cd6a40cfb2d27d35b17c2a4d : SUCCESS in 2m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/37335d6875a842d198a9a05b712f8997 : SUCCESS in 9m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6d1b2be1bbe8451ca959d96dd9b70697 : SUCCESS in 27m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4ba20967786b4d35bbd71fea7ca36926 : SUCCESS in 8m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2085799b7e18467f997f403c463a94e2 : SUCCESS in 14m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f6c14f51c992417796c4a498e7b036a2 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/760f4c7ef53a4203bf33a408bf48092e : SUCCESS in 2m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/288130b3074646a6b43fd7e1da0f9990 : SUCCESS in 24m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d451670aad9941d6a335a4f5a49fad61 : SUCCESS in 3m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4c6417e1d0b3447eb3dace437a6f936d : SUCCESS in 3m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fefa09383ed54d0aa79f0b33b3d48397 : SUCCESS in 6m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9169cccc8809471299d248ed130b0e86 : SUCCESS in 12m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/969b90ab536c4ae68dabf6610acaecc8 : SUCCESS in 6m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68005526bb594918a1d043fd8f943e80 : SUCCESS in 53m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e715052555154c939a1aa15cab94c2b7 : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":4},{"id":"22d74e8fddbe81eb072b91e1a6126eeaa023e57f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 11:14:28.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"01dba2e98573ec1a0345ec854116895a7a708dbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 11:14:41.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"dd01060d2d0e08104e095957edbe8b97be28ec27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 11:51:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff186bbe95c40d2be3dd45c7e573cec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d237a29b2c5943a9bb9f78d6a8e5541b : FAILURE in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3541f149eb44453ebf16af5cb96bd99f : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62e38aa93f20426ab69b5b33eed2efb1 : FAILURE in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a79ba7e0da3437c805e98054267024a : FAILURE in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ee354276ce64ee5a5da465d7f9b99b0 : FAILURE in 3m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa4c8f3bd856452292078a6c55f22742 : FAILURE in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/604ed89a889e4d2c9e3bf4f545d8da13 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7235734574b8465bb989de5e555553a3 : FAILURE in 4m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2c45de54b82c447e858f36a38c869677 : FAILURE in 2m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1169c795730e488b886d91c6564942c5 : FAILURE in 3m 34s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/52f7afe2fc314dbcb76a28da8feddd9f : FAILURE in 2m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2787d0c4922247dfb59e39cbad751d55 : FAILURE in 3m 09s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cb6aa9c4884d47a888ececa7efe5d7ae : FAILURE in 2m 43s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eeca24bb49a34116a31bbf66f17fd77d : FAILURE in 2m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0530704b708642aaadf5e3ff082d2a09 : FAILURE in 2m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5c0e99876a044608a78891080c851e0c : FAILURE in 2m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4592e13d7c454c91ad019b8d98848c97 : FAILURE in 3m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/273375e3db2049f1ae4f5ec3edf6dd9c : FAILURE in 3m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2a915a15c68f47f2aec63a3126186ad5 : FAILURE in 2m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4e24ee400024b6a9d0e047bbe2d10fd : FAILURE in 4m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7f6fd8cd72c3446bb3605c3253fe50da : FAILURE in 3m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c866495ec5594517819fb44b42429c28 : FAILURE in 3m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b6ff771239bc427fbed6ae6a240406af : SUCCESS in 2m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/65efd9b7da1b42528a7cfb56159c10b6 : FAILURE in 3m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6ce64a6883474eff92e3f1c626807593 : FAILURE in 2m 35s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c450de13da0f44e6b6efa75ec4384989 : FAILURE in 3m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/659e15ebf99b40a980b488a0c0e1aa83 : FAILURE in 2m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d0c61434e366457ab011ba0922750da9 : FAILURE in 2m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca49ac71cfbb4b21af5db753d15ef745 : FAILURE in 3m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bc6b2093ab274286b64bb5eefcfed1e1 : FAILURE in 3m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2dae4be964da43d5a518e66c723c66da : FAILURE in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80cc3aadf4024d6eb811bcb095a8bdba : FAILURE in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd2d23923dc241a88f84c2f9b7c719b0 : FAILURE in 3m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0019357626f947ba8c5e18477c7fe2ae : FAILURE in 4m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ccb65c1520f14e43a0b6a7b11ebd8ae1 : FAILURE in 2m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ca5b351fa1945d39107de233374dba4 : FAILURE in 3m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a0a0e65ed5434336937e18bbac39ec44 : FAILURE in 4m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8feb0c7db4f24393957abe785db3e808 : FAILURE in 2m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18aa827b39694f9c8ddfea594a33178d : POST_FAILURE in 6m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9565d150f80045cbb7aedfe9a13c63a6 : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":5},{"id":"fa90e94fc5bb52d3b8060ba34fcef1e990e8e6ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 12:11:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"13226a6139b9172b1c2f1eda1aece034b586c89a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 12:13:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"dd4d7dbb98168f681909e5c2d1bc5e398ebe36bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 13:13:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ba3576daf0dc8801c1f0027349aa12f4deccd0c6","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-12 13:17:25.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"90145ca4151bc3f411cfd4054c26df348a0ea59a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 14:51:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef224c0b295c4fb38eb05e8af9d45967\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84beccf2ef24fcfa3fb32029f11f16f : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4cc652868874e8884a937785cf0ae66 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9007818df4cd4e2f964d09fca2c2ac52 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c4584b9abe9464182a0965390356ebd : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f43962c6d92f4dc9b48f69d3eda7ed5d : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1aaacf2627d44517afd2eb1cfd044942 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b409a51d2c5941a5bec498c53855fb63 : SUCCESS in 3m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/433d3861bb154532b5d97991f108e518 : SUCCESS in 3m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/efe3b21d55204084b8a5a52e65a5b612 : SUCCESS in 9m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ceaf1287e5024da3b3d08b67e2bf325e : SUCCESS in 5m 52s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/dc0967069042424db443f66031b64dfc : SUCCESS in 5m 21s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3cb73ef75c734617bdf70c45bcbae1b9 : SUCCESS in 7m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1809491c93ad4fb59f30196618ec9dd3 : SUCCESS in 5m 03s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c74b9175dcbd4a0194da172f5948467f : SUCCESS in 2m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/76a75bd2bc7543cdb1a0a3ee1e76e350 : SUCCESS in 11m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b2ccc33efc8442f5a48f0bfec42a843a : SUCCESS in 4m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d4b008a4020d437d9e37fecb2a8db2fe : SUCCESS in 7m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/913bc4776fb94a5fa73276c1e3ef506d : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d1acc0f9564a4c2fb2d995f320235d04 : SUCCESS in 14m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/127ed8c38929451d9fce2c040e66a7b9 : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7772961461034ae0b63216b2ebf5fc4b : SUCCESS in 8m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b5527c1ba4374f268d800e8fab75fd6b : SUCCESS in 10m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/604b16944e684dcfacdb39f29e443d4c : SUCCESS in 3m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/96c062fc38604d60af57987169500009 : SUCCESS in 10m 55s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8e3f6a79f6294c978305a7940251006a : SUCCESS in 6m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8f15e4be3c51448298eb79252dd28987 : SUCCESS in 5m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/268c11de220a4ca186c09f01e71178fc : SUCCESS in 12m 07s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/623390969f6540eb80ff219ef9dcaa2a : SUCCESS in 21m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a91609a41e4644e9b66f26f7260eec0f : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d4c39085229745488add8b767868b997 : SUCCESS in 16m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7a467895e05a482ab155db7c8021bc15 : SUCCESS in 7m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7d00d21339e497393c8aa13895ea2a0 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3c8370622984b3d8edccfe329bf2ca7 : SUCCESS in 28m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/81860be640c944d787aac65d12425f24 : SUCCESS in 6m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c0ec04ebdf6e411faf06fcc9108fdd67 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfdf08b2fdd44f0b8d4af8d1452dd52e : SUCCESS in 6m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ae2156d25c5748978bf5fc5311b58e47 : SUCCESS in 19m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4542a4cad9fa4844937b94aee33caccc : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dfd77a52b334238bd93010931ee327a : SUCCESS in 1h 05m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78a6507a4c5340bc9c78705c8e2fa962 : SUCCESS in 1h 33m 03s","accounts_in_message":[],"_revision_number":7},{"id":"7f04aec183da68c1e05e56cb554185f5b676be43","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-13 11:56:02.000000000","message":"Patch Set 7: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e73dae85f6e7bcfbe71c4c9a0c519e8a2eb93b34","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-13 12:29:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"58a7e1e4e36db1c4f15099ef46b88f35d73ebb62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-03-31 05:43:56.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3a087964405368ed482c97b1688d1ea5c5dae096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 07:28:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4fff33526a846f294952a4459b7296f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e0a5e631274138bff6a8e72b36b872 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be92dbbcab64767b72ee7b8568d7885 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c970f56bef94e9da2399155234bb8db : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc5e7ce822fa4ecbaf70625f2958f453 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e47f4b0cc3424db29dc0987fbcc0a77d : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f908ec3774b74becacdfeddf29383bae : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ebca1198fa24c5e84f5120fa5f0b0b8 : SUCCESS in 3m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/01392ccf4c8c41eb9de3f6820d8e5abf : SUCCESS in 5m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/353d838b49eb4f9b87eb8de97f1c0402 : SUCCESS in 5m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3a5d18bd7b1e40049e2838220f9547f0 : SUCCESS in 5m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e65e1fde40e34953833139b35f27e949 : SUCCESS in 5m 35s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7910be55ec174258ba0a18557c4665fd : SUCCESS in 4m 29s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/447ed0fcc0bb4697ac2a4878508f081b : SUCCESS in 4m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/769f3ad0b3e444bc99ece6fe490e26a9 : SUCCESS in 3m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/765adec90a5e4827af7c4730d4bd0890 : SUCCESS in 11m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4f31ff6e2e2345e78e1d0b835be72b0c : SUCCESS in 4m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a22f1abbe1294680a1513042ddc5ef79 : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7dc90fc88bfd44988c60da72170f5da3 : SUCCESS in 3m 57s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0804bc59504c4ec2bae97f665d953baa : SUCCESS in 16m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab25a632cc944148b063acd89bba716b : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1678fe4980354feca5e989d8a906d974 : SUCCESS in 4m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b32a2a5e18db4151a0f7687d3edc3beb : SUCCESS in 10m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/463c043254454cd3b9fc314c38b2423d : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3f5667e3c142460eba2a783be22d373a : SUCCESS in 5m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/50d2a0ec11854c6ba0da677b50040099 : SUCCESS in 4m 04s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dff5da8118624db5867d1a2832809e52 : SUCCESS in 3m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8cf226cbfdfe45c89f0c5bc8780664ed : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2155b9b1b67f4c03b5b8fd98103dc234 : SUCCESS in 26m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/551bceca83124e2eaf2d687c36dd78e6 : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0387671d9cab4ddebd0c5bf873a16f3f : SUCCESS in 13m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9748d9513c5040d3823aea15add4b62d : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dfcfcdb2a3ce4379a75b1070862e318e : SUCCESS in 2m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/853b7d90fb8043428352229f33ef9bd4 : SUCCESS in 22m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ebcbbf0f4af3442cada79e3af6f04c9a : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8f2845b156c04b4a8bf4e96810d82956 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/640de8f0f9374e41be4de6677508cbb0 : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/89721d8e43fa4fe99a33b2722759306b : SUCCESS in 18m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2292e623fabd4e3183b5d3f933c04b7a : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/086e8c4450ae4ab0a488424635c4e155 : SUCCESS in 1h 00m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4bafff283394abdbc9a75bcedac7503 : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":8},{"id":"adc409ddd1a0af26ffc44dcd0d03589808c26e0c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 18:02:43.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3e68ff83bad122d6d2295a75281ca1ebe11180cc","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-02 10:07:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bebd99988917edbc2af69f96ac592c9b4394d985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-06 07:36:41.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7e18447cb37f4202299bddd1da1417b7da1ecdb9","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-06 08:40:27.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2a98301587a734db07f18a51e712eb0b5d08f1ea","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-06 08:41:30.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4034d8b79dab265301450b7e27e5dd0886b439a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 08:58:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343562273a66473e96bc5a6e0cd31b3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7bd6be722547079a9d09214d88bb31 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb9cfae944d4bd3985b6e03526b8dd4 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b35ad942ae1420bb69efb24e718c5c6 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cab833172c644bcc9267e7e0a6586768 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6559dc37e5024affb68d6342639e14b8 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/104894e7f0994869b28c43fbc92258c8 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbb84765cca94881bed678bb0d75b202 : SUCCESS in 3m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ae88566ffb8a4a5798deb9d49f9bf1dc : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/250f9d31262e438b9f34a82056d25c3d : SUCCESS in 4m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ca5695f28d2f4f669dd8e59e8511d99f : SUCCESS in 5m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b6e87d7680234aeaa1c701b200cdb86d : SUCCESS in 7m 10s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fbc3502bf98243adad26164465bcd018 : SUCCESS in 5m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6b9ddfe4b9bc414e996646089e21380b : SUCCESS in 4m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1b99b47343c340f382571e752c9e09f1 : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/920683dbb5a54164a9da6415b186af7d : SUCCESS in 11m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/487705f6b38e440c9ae99f132a984b96 : SUCCESS in 4m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/817184628beb42ca902f91ab7d15d4c7 : SUCCESS in 3m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2d8d93d89b2f40b98a20a1bd0b84b202 : SUCCESS in 5m 43s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b162da2a44b141ae9df5494152852d3a : SUCCESS in 18m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4d195d834424029941384d781541f7e : SUCCESS in 5m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5c89896ac25e46b4a7e78a35748b25a8 : SUCCESS in 8m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8184f410cd3f4fc2aaded78767517fed : SUCCESS in 12m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7eafd592990b48ab9173cbf4e14316ec : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/06b77c49315a48089cd4efe596bc25fb : SUCCESS in 5m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c9ccefe641044bdc9d912035552798f4 : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7c7c4f27734d406f8eeda4cdd6bf745c : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2be61762f9034f0993b78c46009c1537 : SUCCESS in 11m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4bee25148211408793425761633e49fd : SUCCESS in 27m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b7db4cbd1ac244a8ae989c1be7a73fa9 : SUCCESS in 6m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d3fd7300bbfe4636bf10b1c21bfb0c24 : SUCCESS in 15m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/41991f3a51bb49cd8fbdd174f5f0739a : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd3b1b39f2d244989c6b33e42376001b : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1da066d6408249609373310d33eedca0 : SUCCESS in 25m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2c4f73e7eecf491ea82f946b97dd2819 : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6d0f8e2ad9854deebef128036ae0c747 : SUCCESS in 2m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdaf6d2b1e0d4d92b84902fbcae749ce : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d8cb0e26c45e4734b1f298987907b0b4 : SUCCESS in 11m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/25bd04eac51d46b6aa1ed3bd35f91338 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/237367abef1e4ca59b3463b4b25390c9 : SUCCESS in 32m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43ff47f579d143bb8967e1e034ee93c8 : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":9},{"id":"30065f1fa73b78d4cb6f62dce01323472b75df62","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-09 15:47:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"efaa594c6723d95c689533f2183659e9cf0297ae","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-13 05:01:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2c69c9790553e67c2a94d05707b4bf76e2a125c5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-13 07:35:39.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"de4885a41065b23d6efdfe4f85ab1c01510a3723","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-13 08:16:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0b74325fb72aa872012822468ddd4af866d13f8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-04-13 13:26:23.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d3b34e708d2f3ce2c7eac1e11734a5134c1554ac","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-13 14:15:30.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"96de0421bcdebad758d9801f40d14f910a8eaa02","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-20 09:49:23.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"8c1f5eb9746ee63561585da5ed9a31673c3e9f61","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-20 09:57:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2fcccaef58dffd30d6bab0cfc4276f1dcbafb1ae","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-20 09:57:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"f903610db1c4d3ba41160abde4e296c1e2aa4d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:11:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eed798ffb6d49c2b6a857fa49a3cc06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a520732ea366418ab2439a7239d31028 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6adabeb6220b43cc80f8084c9e372a35 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/624847e9cc544e29a76b4fd873e9daaf : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f8c7728be5248cd8a328f93d2ccce1b : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09494201fb454fb4ad5090d4a09bc7cd : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18d5be668b61489684d45f9a64227b5d : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3372d9e70f8646d69906d6242d6775b7 : SUCCESS in 3m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d019d0e3894a4201922036034c976e62 : SUCCESS in 8m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/02b2932250b544a4879f8175a9cd60a5 : SUCCESS in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bf8308cdd9ca4a85bd05071f30219aec : SUCCESS in 11m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fc1ae8242abc43229dbf5fd5c8c3a1e5 : SUCCESS in 6m 15s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/4be9379cf86b495aaeb2c958da9413ad : SUCCESS in 4m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/377580b6d9ca4ce9917ab601a9260080 : SUCCESS in 5m 07s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/67645eccccb240dab64eac0161aea02c : SUCCESS in 5m 17s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b2603465ab594d3eae35bed4dbe9cf6c : SUCCESS in 12m 23s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2b506bce195c4ad7be420512674d5bb4 : SUCCESS in 5m 36s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4cdb798687e74ce4820bb7b124b0daaf : SUCCESS in 5m 43s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/578f95e46ec646188bd771a069c12437 : SUCCESS in 6m 48s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/caf2fed5fc1d4d27ba3bf3a9abd02fe7 : SUCCESS in 14m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eca5ad212c72474fa36ce41363424e6f : SUCCESS in 6m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6e8eb228f3154f23a3ca46c126c45563 : SUCCESS in 8m 17s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/7473f6a9ce6840e0ae8635bbec7d916b : SUCCESS in 10m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/63b6b721fb434cf3a181768972e3b00a : SUCCESS in 2m 47s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8ffcf73f02c64685bb7171f4c391a309 : SUCCESS in 10m 06s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/dd6f70e5a09643ab889617fb0a58e489 : SUCCESS in 6m 47s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/eeeb816b94384a99b8f59f634d60c0ea : SUCCESS in 5m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4ebf0deacec24bde80089e01a162ce7b : SUCCESS in 12m 43s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ec0c73d83db0441babe44c9646e6536f : SUCCESS in 24m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/281e6ff457874c7a9ef7754027cf0c61 : SUCCESS in 6m 59s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9b96bec1213f4743983dace76b9c9a9d : SUCCESS in 16m 18s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6f151191383f469b94a190af98c881db : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a3a7721d748450ca16e3c28cc252fd7 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcfc325aa9c14241bf9bd97e9547b190 : SUCCESS in 25m 36s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/caedfc38a5f544a49e6540fbb634230e : SUCCESS in 7m 20s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3667f23902b24135a0f6edb5daa13d2e : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ed51b256f93439e8a9e4af6baced588 : SUCCESS in 6m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/11246df55dc84ba7a48dcaf34dc082c2 : SUCCESS in 18m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/06f4567c69ee4845ab9de2acbe3e2923 : SUCCESS in 7m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb648aea150b4b01838332fad2f17aac : SUCCESS in 1h 09m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ad686a8fea44841988f886c5e3d7c6e : SUCCESS in 1h 43m 04s","accounts_in_message":[],"_revision_number":11},{"id":"7fca6d74502b1d2335c5435d2f6a8e9c024568ed","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-20 12:37:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"98fb1e1972a60c0ed096dd2292fbfec44cf9d35d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-04-20 13:42:04.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"2d97467729e8e1c6582854b5c8bc2e2e0a636790","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-04-20 14:06:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"da582cfd2f2b4bb2ca75e525d26395e63ee7b8a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-13 11:25:01.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"49e41ba9087fbdb4a7fe78b43e146b20290ade1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-13 11:35:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e37cf09536aa55215c1d949d9de426d2203b4bb5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-13 11:35:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":13},{"id":"21e2b6df17f19cf0d24abd0c9120f80ced22daa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:41:34.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ec3ae0179445709500a947f3a2c512\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc15e93edcef41ae872e20462de5e4d7 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef60612bb6f64e78bbfaead53c00d2a0 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/243ff7ed2ca1431c993b1615898ab140 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f27108626cc45bcbbeed7915ee98eb2 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20589d2e94b34b48b3586366cc93da72 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ef3dcf756354a09adaec521b2a6d3e9 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66017d46191142eea1799e108cededdb : SUCCESS in 4m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/375962531d8a42169b3660d530873eb0 : SUCCESS in 4m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/280ad3aa0b63464f90e55236f1d25b5f : SUCCESS in 5m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/71b3cd6308704e2fa599a8edab4e2ea0 : SUCCESS in 2m 50s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/1384a7a624a3408fa6747cd4368c2b58 : SUCCESS in 5m 37s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8332211b626042aabe13df0e6659bae0 : SUCCESS in 4m 51s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8fa1197002424666a07dd16f29bfe7f0 : SUCCESS in 5m 21s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/e8873458d68d42509ede3defb30adfb2 : SUCCESS in 10m 04s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b0f09051b4684ceca019885607adfa07 : SUCCESS in 3m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9e5a843b1a334485b40cbf9c516ff055 : SUCCESS in 4m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e963adb4d9404a66924422d65d475455 : SUCCESS in 4m 50s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/fc4df2f6d73f4e5a9901b087a0cf7f9a : SUCCESS in 15m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/101a9abb8be34017abd819e10613c8f7 : SUCCESS in 6m 10s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8793bab220674e7ca317a15e0d0295f7 : SUCCESS in 6m 27s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e11f10345ee24947b52e5e67a5029f31 : SUCCESS in 11m 58s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f9e7e08affc240c8b7e6dc7f28fe2af2 : SUCCESS in 2m 18s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/30532861ad524223b03785a9767169c9 : SUCCESS in 7m 17s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0d51c101b073487c9cfec588a6074b15 : SUCCESS in 2m 22s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/8d8914bbe81242389faa1365ed0fe3c7 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/45ce622ea57a42b9ac373a0e4b69e711 : SUCCESS in 11m 10s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/962224105f4b4d748962d7eddec2d3e7 : SUCCESS in 26m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2b528a423826418eba19511c84871c1c : SUCCESS in 7m 05s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6dfbfc59f64e40c18dc8435f0b272f69 : SUCCESS in 13m 25s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3b52628d448845ddac9a8be9e643b5fa : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90dd3a149a844f91818f324d556a080d : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/352cc9f5f3c1474e87b1f8879c7b508d : SUCCESS in 25m 17s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5536817b3f8149769f01cbf3f30ede76 : SUCCESS in 5m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/01925f2f032d4ccf8afe38c9f42019d1 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a80facdd5f04041b7c028bfea3e78e2 : SUCCESS in 7m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/8f29797140fc4ca9a8b7581600e270cb : SUCCESS in 14m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/20ddc9fa6c8d40058ce737a5914ce87e : SUCCESS in 8m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b96260fc5684d2d85787f104ab83118 : SUCCESS in 41m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f15ecfaae3e4cbaa1eb0d1cfecc3029 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":13},{"id":"256c7612a27bce087ee91d27c93eea2c99ffc2ab","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-14 06:58:46.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"123d3ad120f89d4f7c0fb2fe8fb92500a91594d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-14 07:06:37.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"0555bcf8e31591337fdd3c345edd3e20ff3d77c8","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-14 07:08:22.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2ba48510a68365134ba2749adf1af973e3ca6d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 08:15:18.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/933593360c75401d8c3b3606e5b5eb56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd680228223d404cad64c7b8c60b24d3 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651772aa76fa40fba5b085af57234808 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7946540b9654a2ca82d8d3445d036fb : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6ac268139b0429ea7198837b2b00329 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61d906a6c5d54e11840bdea2a6181eb5 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c49112198f64919830c058435718915 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30d9f0ef99e148d794c50f2d47a225ba : SUCCESS in 3m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0313659aba4c42fd9bc23c737aedf54b : SUCCESS in 8m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/20ae6354c60e4adfa967fb742b114977 : SUCCESS in 7m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5ff1087828b940e8a59b6e74182c3f30 : SUCCESS in 4m 08s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/4371432ac8f34431abf29100ecb5c588 : SUCCESS in 4m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5f893a7ce03241f1ade8d86bd032bfc2 : SUCCESS in 5m 58s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/7ca488d842fb4d8d8cb32b25fdabb3b7 : SUCCESS in 4m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8a70dccb282e4e8c8d3dd8ee8beca378 : SUCCESS in 8m 26s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/3d58c9e3d99e439388947f99a60e57e2 : SUCCESS in 4m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b5a38c10c8eb41bc8eb687bb275e394c : SUCCESS in 4m 26s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/fad3ee14bbb849b3b7d907a10b31cd32 : SUCCESS in 4m 09s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/64e19a18b6aa4914b3ecf06a86ff85ac : SUCCESS in 7m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5fc4411d130420fa4e789ea714ec05c : SUCCESS in 6m 57s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/ebf2b43abdcc49c6a17815f17de7cec0 : SUCCESS in 4m 14s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9effb4c9c8254cf88e3bb02bf4fbe161 : FAILURE in 9m 04s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/eb31186777f947c9ae7763dc03938d34 : SUCCESS in 2m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6cea1dc5b11c49779e0e92f1c423c8ab : SUCCESS in 6m 23s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/7dac17552c1749cba40262a83cfb54ee : SUCCESS in 2m 15s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/30818135fdf74196a20ed470b56f56fd : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/70de44a5da914658913afa51bc9e9961 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/885bc45219a44b42abf2131515e8f3d1 : SUCCESS in 24m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17bd5f7723c8405cb8098932935c4ad9 : SUCCESS in 3m 29s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/d1e0fa782cda4e75a94ecf20b26bf491 : SUCCESS in 13m 21s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7edf8cd3571a470cb0a8725800cd4aa3 : SUCCESS in 4m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e1503eef93442638f8012f901401efc : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3cb8050b3da04c338fe5bb9c23288989 : SUCCESS in 27m 03s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b2cb89468cc74afcb7a1f0f331ae4e6b : SUCCESS in 2m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/dcff6dd0b8704a11ac72d537de552423 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed7dfb51cf664bca9d1202e9e41d4194 : SUCCESS in 6m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/46c86e55c93646dd9c128c9eacd09b83 : SUCCESS in 7m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3d31e38f6f9d4582ab1c0e2c6aa1459f : SUCCESS in 5m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e41546e09d74f6eb46dec7c67e2ff9b : SUCCESS in 32m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47425e2fa8494b56bdb6b501727ffc54 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":14},{"id":"70565dbc2ca7a6968d623d426b1fc7a684cd5e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-14 08:43:34.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"af02d3f72af530afcda74702738ea72a428b3cbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 10:38:39.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1249794d27df4effa499ec221844d051\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e85f7c41f8468194cb3960b6de834a : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abcdb37b85f147b0a29b46a9cd9c94dc : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d8d539fbfeb453d99e08ebc4c7494ea : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe08beb39414748835084f73f70d730 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/28bbb2442a974e9ba411bb09d9e6b339 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/289ff611501543c18d3aa35a54d22c19 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae8becd9543c4ec2a2f8738b56f8dfdc : SUCCESS in 2m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/964d7af91d1743a78b1bad9513b5777a : SUCCESS in 10m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/831f4877adcc44709fc7e94aef2fbc35 : SUCCESS in 5m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/762ff0ea78f3423bb0f1e3f9f9941a92 : SUCCESS in 5m 22s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a3947cf0c5bc4cf7b6c9e844f83fb1a0 : SUCCESS in 2m 43s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/83a2f03e25914c729305461957c137a5 : SUCCESS in 3m 00s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/3572fd78a65b4173b1f04fe773c7bdfe : SUCCESS in 6m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c615baf7e5354d068a5e8293f78c9d63 : SUCCESS in 10m 19s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/c31650c56d534d1ebe3b1fb9a76c7ee6 : SUCCESS in 2m 52s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6e69e738acf144a4aa8da9a410ff1a88 : SUCCESS in 4m 45s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/dc21d8329e4045e69c0e59b74605e203 : SUCCESS in 7m 11s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/269fb9a3009e401282ce94621477fb87 : SUCCESS in 16m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88171fa0149344c0b477a7654505250d : SUCCESS in 6m 01s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/658f7b7df4594252a46a82ae8f98afe6 : SUCCESS in 4m 23s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/afb4be90ea454b85acb0e5698f73585f : SUCCESS in 11m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/004213b76d2748c8b64bdb079507ef8e : SUCCESS in 4m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d8b9220d74cf4bf5bd1a75106b34cb58 : SUCCESS in 8m 10s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a9cbd8a5070c4bf5a49bb3170e7a49f7 : SUCCESS in 4m 27s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/b702db8561fc4149b55c122da3347384 : SUCCESS in 6m 37s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a2088b7e08e44f4eb3c7167ec83dff33 : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8c7f9caf8142435884b045a58cc4f184 : SUCCESS in 24m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ebfa33cf4a374549b604137d1a106177 : SUCCESS in 3m 23s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/10f7b17625ca44aea13aeefa8dc31d55 : SUCCESS in 16m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/cf7ee8c519484bc1a5bf721ff8d6e118 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1bf9fd072ff64eb98856d07d7be62141 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4826b583d7a646d7838c59cfb6db6565 : SUCCESS in 26m 07s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/a08eb45626ef432ebe16248eaf1cc6ca : SUCCESS in 5m 06s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/cc7b66600b36411980624a6a41b9b449 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53011e0aff9a4b59a0be961bbc8982ec : SUCCESS in 6m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6c9fa85b6cb8438c9fba4284ed6aefd3 : SUCCESS in 7m 19s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5a3085158446417095ace3a373e60123 : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34ef86ad686b49628a11b9a91486f0e3 : SUCCESS in 1h 03m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2679e5ff6404347b11d9a38a0aea12b : SUCCESS in 1h 49m 38s","accounts_in_message":[],"_revision_number":15},{"id":"87aa20a26af3051c4c3900a5b87429fffaf9eb1a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-20 07:50:53.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ea5476792afd841060348d9b5e37b56ab9c68669","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-05-22 04:56:26.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f6d222b2d9a0821a1e90ce15a3b02f8493f0e30b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-02 13:24:43.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3e73144f047b760c11ca6bfbcf9d71fc2c1866d5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-02 18:49:08.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"01a3c196576f2e4d87d9cb594e7594f07f35dfba","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 06:49:12.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ad70a4f7c0603de6adf66fd41b407edccb0b0645","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-03 09:12:12.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"17c7981ef8cf67b530c1e1bd1de2e513be909bb7","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 09:20:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"26b53699b426fa924cf383877ee0a8a2d83e55c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 09:48:57.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"5f7e09d4c5b725ca52b634bbf6c42457335fcd7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 09:53:45.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"1dbbd905d8e4e0310dc6f34ea02137248d45cef0","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 09:55:21.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"2f951daa2c83cb580803aad6faf045badb639274","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-03 11:12:10.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"c3666eb9392c7c0580f97a1cbfa280bf31f72d7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 11:27:18.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n","accounts_in_message":[],"_revision_number":18},{"id":"1c96c76b1011818a334fc56e9e1e5513a4e3bff7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 11:31:17.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5260c9f1a4eb50ee977308a23b06323f420346be","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-03 11:32:25.000000000","message":"Patch Set 19:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":19},{"id":"a1aa64a405cb57d111e23e6b296ab2a4073e7e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 13:39:47.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef54ed32f84456392484f1d747f7434\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8433efc7ddbb4debac4c9a2412647b23 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd8552f23eb4eb69befee76347b5156 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbf1cff3f15f4f7c9f2c55764f3241bc : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cf43fe420114f3b9793f64416b626e6 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d601ff115584282bb0cc02030c4a42d : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a95c185146734de080428ba840e99441 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/468ffbc94eea4b6eb824332e8e140802 : SUCCESS in 3m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2bb7a4c03def4daa94be3be5b3551fe8 : SUCCESS in 5m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/34db89a5e17044a587aeeefe8138dad6 : SUCCESS in 3m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d7f018b29a194e368540826943e09d9f : SUCCESS in 5m 21s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/551ac3fd6fa64aec8d91c96cf063aeb9 : SUCCESS in 4m 34s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/36b45c77adf047719e88da2a00e8827c : SUCCESS in 5m 00s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/eae3ee05895e4e31bf9aadb9c0f3ba6a : SUCCESS in 5m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fdd7470caae24b3d9df454ecbe3894f6 : SUCCESS in 10m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/0191d1e0aa8b4e76b46a03ad64e45750 : SUCCESS in 4m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/228115234ef44c63ac7ff887b2861e84 : SUCCESS in 4m 42s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/359e38a28a424ef1a9dea5dcbd9babea : SUCCESS in 6m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5965fc28fef14815a26e4a2cf99d7a7e : SUCCESS in 16m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65442af70226401b9c52afc569685442 : SUCCESS in 7m 08s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9ea76b65dd1d4053b66e477f6949ca9c : SUCCESS in 9m 10s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8c0915956bab4661b66d9b049c80a2d6 : SUCCESS in 7m 06s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/0adcc506e16d4b1f83f1118c149a7a6d : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/6060864917484fc7b234d7031722aa8f : SUCCESS in 8m 41s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/dcf75886b1724426935325f089724795 : SUCCESS in 2m 29s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0b7abe45e0a641deaa9d96182d7b1b11 : SUCCESS in 4m 14s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b9f60d99129043adb2bf23b7b4e60d8b : SUCCESS in 9m 43s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a5abb6deac8f441e866c5306f1b26eaa : SUCCESS in 31m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f29795da357341f393237731625e2dca : SUCCESS in 3m 36s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/cde9511e792a43dc8db48b78cc7d8bc1 : SUCCESS in 8m 34s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/59171e294d5148a1b794d10b0a8267fc : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/813e5c0b6db84fc494b2a47ded6e088b : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2ae037425234618812fc2ca07c053af : SUCCESS in 25m 01s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5156b00863dc4982b7d758d7a8995d55 : SUCCESS in 5m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/a76565c0335c4d96ab348b56032903f3 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3182b43692af4b74883b8e2fbaae4705 : SUCCESS in 7m 31s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ead8c89cd0ca4531868689c24845889f : SUCCESS in 18m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5ce412ef97634446b187a4f89a072e33 : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d8ced76b0ac4812ad46223421bf028b : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13ad41d029b94622b23b3d6886bbf303 : SUCCESS in 1h 59m 57s","accounts_in_message":[],"_revision_number":19},{"id":"7d10980be85be7c76500a42f332786798df42e48","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-04 07:27:41.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"56eceb7b3a478118de8f71c296c2daeda9a276b2","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-04 07:36:29.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7c22421869689149e501626eb59df2d00134c4f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-09 06:52:00.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"a752a79fb0d2458ac8fb07625992755aa2f3fc5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-09 06:57:13.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0c55f696c2dc2f007d42f955550e188d65a9a683","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-09 06:58:03.000000000","message":"Patch Set 21:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":21},{"id":"cecd73c7d46f6e50a0d8a186746a15a858e3ae57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:48:47.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f02ea6012f6845428fa49ae576909474\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488e1608579b4917b88a4d1276185cde : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be303ddcc5064931b05d93acb3e6d93f : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5879592de2ce439097a797201c11826d : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1320951090d240adb538c74833845e2c : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c866008742724b8981f1eb9240754e74 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/422737b457d2461d91b1fff665eb9727 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e278dc4b6c95411293fadc5182e08fda : SUCCESS in 3m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ef34fe9d37ce4729b291da3f20642488 : SUCCESS in 6m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d839ae204ddf40e2b554ec7e2e9e24e1 : SUCCESS in 6m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/42f782a42dd94ad0a0c2cca9e803a3c0 : SUCCESS in 5m 21s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/f9fa9bd9d48e416ea1b4466421d2cd1f : SUCCESS in 4m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/616c0e8401624c39825c6ad1f7448c00 : SUCCESS in 5m 46s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/837187916d9847ff98f2a6998f041b35 : SUCCESS in 6m 20s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/c8d54ac52a1a41cb93b6337931819c76 : SUCCESS in 8m 48s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/e6c6bff24676455c90f915dea60b4410 : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/34377e01d723484d817f1a3ebb664e24 : SUCCESS in 3m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/686d012eec58403393002cf196617b93 : SUCCESS in 4m 11s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/f4aa65d374224f9c9477b45ce852b4e0 : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80dd7c0eedd949edb18260c79496c00a : SUCCESS in 6m 44s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/1afa710760ec4d6da60efa06283de106 : SUCCESS in 5m 01s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2dec8a9b17bf41e290e61edd2689150d : SUCCESS in 7m 41s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/beec0e21cb5f42a99b98e73620781153 : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/4d26d7f8ffec430482cd98df1f9d1d52 : SUCCESS in 8m 13s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/dc26a097ce524a86a914cc6e1f14120b : SUCCESS in 3m 52s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/84e2a3ea529b438f8b4dd05ef78b6e39 : SUCCESS in 3m 20s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/149e429a4e2744b1b8361b294a92a9f2 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c829ac1550764852b658b99aca19230c : SUCCESS in 26m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a4e44f10d574b2ba69333ed0d0da322 : SUCCESS in 5m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ea9bd7e86a1a4eb0a1d1ce903045c673 : FAILURE in 12m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d9327d0283c049368e1fc1d1c1ef95f0 : SUCCESS in 2m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6f70db1c7c54af4b24cbdd436bef626 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f11c11f09af643c0a5e4995c947a6a06 : SUCCESS in 29m 12s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5f08fd978ce94b7d8b7f8b304e5175cb : SUCCESS in 4m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/6b7b040cb7434fa69cf4ec595def6805 : SUCCESS in 3m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39728ddf67ba4216bf9acb0a3fdc7fe3 : SUCCESS in 6m 15s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/c93649e1b4c4495ea0985f28e4d2b647 : SUCCESS in 17m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5b9503e92e2a4ebcb31c60acc2ca0f75 : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/590fbdf9760844ffbdf5150b33203861 : SUCCESS in 1h 09m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30a8b7ba7da44170b1ba87524aa35144 : SUCCESS in 1h 50m 21s","accounts_in_message":[],"_revision_number":21},{"id":"4d017231831d882b59521da629bcbbfc49dff1a4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-10 20:33:35.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"f0ff1a26dc9a14c124ccd2403053acd2e58974b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 21:34:19.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9244d91f055940ebbf8dcdc282afa901\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34a354c373642c8a3f3ba12b4abe488 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e39f2306ab4c79aa9ac4d8819cf417 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79e8a73b79c94a81a9e74dd7d7bd9ec2 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c493c80c4b2c4e6089b929d3fb0572d3 : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d40dadda9d794c0ebd9ef83e8ec362d6 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c5b0bbc3cc04724a032f1de94c1c8d2 : SUCCESS in 2m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c87cd57c32b545969d3a9c72273ee6ea : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c71e6a2654024e678aecbf4f5d0b0f4e : SUCCESS in 3m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fbf263bde2d747d988117819a3616451 : SUCCESS in 3m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c299e29fe9104e539b0717c1f68216c4 : SUCCESS in 5m 20s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/fd1328bb82c3427194160948fe0522f2 : SUCCESS in 4m 42s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/7903a1dd59254a48adbb0e22439e8a59 : SUCCESS in 4m 47s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/6b3c462a3a9649a3aa9fdb1951db810d : SUCCESS in 6m 21s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6e43929225ed47dab4827e5b17eb486b : SUCCESS in 8m 36s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/3d3ccc943fa843578e603797d22b4de0 : SUCCESS in 3m 48s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/318846ad27ff427e93786f320d4bff3a : SUCCESS in 3m 38s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/4ba86a7634b2419c87ca06ebd817b887 : SUCCESS in 6m 25s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/90a7de4ad1f04f65a4ebbd62fa4af912 : SUCCESS in 15m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8097f3e1a5114bc78a171daa7cf727da : SUCCESS in 4m 46s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c19cfca00ff24191af114382cec867fb : SUCCESS in 11m 00s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6520f7e2421b4affb496874ef87212cc : SUCCESS in 10m 45s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a632dd37c14d4369ad4dd44b20cb534c : SUCCESS in 2m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c3ea15335a4c44b08a99c58783829536 : SUCCESS in 10m 57s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/4393abeed2434518a6719c3f9a916818 : SUCCESS in 4m 21s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2274faa39e0b4b11b59c5bffca9d1707 : SUCCESS in 4m 29s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/764a0dd1e4294cfbaef6b03b980f52cb : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8f78b8cb707440c494942c70df62a41a : SUCCESS in 24m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6221af0260144e20a7da049b5e24045a : SUCCESS in 5m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/404334c54b024b8991274064db961c60 : SUCCESS in 14m 01s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/bd08104b357c4da5abafe9976ed8dea3 : SUCCESS in 4m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4b127e07d9b4fafa16fb0e27409433b : SUCCESS in 6m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be49e5992edf40fdb1b0f22ae9106a00 : SUCCESS in 23m 10s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/c73afb0110fa4814af101dc8a5fc94d1 : SUCCESS in 5m 20s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ca7869717baf42ed9f0aed3f078366eb : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6417701fc71f417699857b95aa692d09 : SUCCESS in 6m 54s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/1ffeece7b3b149e59fefb4b6eb9165ea : SUCCESS in 19m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2b80dcba9dd14546b687fc8f66bb73b6 : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ae9afb658974dd49ee5a1d085d18613 : SUCCESS in 33m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/654c68343fbf47d0b3daaa3d5b9b9ca1 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":21},{"id":"4c8724cacf42cb39b42642e8e9bbfb7673db3745","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-11 05:51:36.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"3ad63efec682877c581498fb861c94af8bb492f5","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-11 06:20:30.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d786c1aa2e0e948854072a1bcd564e2282532eba","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-11 06:21:15.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8e122f99d6cff9e3573adce2f0f81da5db32f069","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-06-17 14:51:13.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"330308e6bdf8eedaac9f8ccf50087c2bfc048e93","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 04:28:35.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"8e6b554620b5bcceae319282a5780987fea5ab1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 05:19:36.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"5464d310a78ec208c9a016a8cac57be6679b1dae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-18 05:30:10.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4504faa89fc99931a5456d162287608ffdfa49b1","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 06:07:04.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"b17c71980e54020937a8488511bbe4afbc8957ac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-18 06:23:01.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f8e311aedb7f56f225d248b7aa72fe60eb40e3e3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-18 06:24:37.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"bae07a1100b90163fb72ee77330f0cebd040b0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 06:26:30.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f685dcee764b33ad494c8c4f65889a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb647f460e84b00a65eba529da908a4 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131874c91f2446fba97603730018030e : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9edad074d80c4d55968fc402402dc8ad : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fadc243fc38449c0964052fc1f43afbb : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1ab2a590234467e8cd87af204e88eba : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5aabe969dcf46f2aa8768b9c7e4897f : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc48c54664bd4c29b7742a8d7420fe9d : SUCCESS in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/269086e8204c498eb77b6dddb024a44d : SUCCESS in 3m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/852efcb83c2b4123ac331b69425ac07e : SUCCESS in 5m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/150ea02e1c014d4194a08045f2b665ce : SUCCESS in 4m 39s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/0633a0aa1f3a426eb0f35c288a4ed31a : SUCCESS in 3m 10s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/b553e6c76fb84ea6878988b9a219aab9 : SUCCESS in 4m 32s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f95d01a4b70349d2ade7bfc2632faedc : SUCCESS in 5m 09s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/1dc7965c99ce4e33929df22ead91f90b : SUCCESS in 6m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/4d3643592c01470e8643a14ab95112b6 : SUCCESS in 3m 33s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/b064398b7255419789d684b3335b9d61 : SUCCESS in 5m 07s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/4ac7d427f0ed49289230f6569366315b : SUCCESS in 5m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/07843f0aeca4436fadab1f9c68330740 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3196ca7769c3485396a185a45f2c2482 : SUCCESS in 7m 26s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/3b53f902cc2c4c309884a7afe9bb43ae : SUCCESS in 4m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/39d5277cf1cb4b1d8a005d0b971bec0e : SUCCESS in 6m 39s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5ca0e2d0dc8f454c8f7db06668c0b8e9 : SUCCESS in 2m 35s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f1eda74f38c644b7a58104203b61dd0e : SUCCESS in 4m 39s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/91cd2e0de6fa4e3d83ca6aaa28783ae9 : SUCCESS in 3m 46s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d49304bda3d84aaea880e39d541b2ee3 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/f7fbbdcaab134b79b90fbe5af14794b7 : SUCCESS in 10m 25s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0cb03e2b371b451ea686c39016d63672 : SUCCESS in 16m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a387224c4e50450fba857f90d24bb3cb : SUCCESS in 5m 11s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8b8682672d464a0186adb42619f5f250 : SUCCESS in 13m 36s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/723390e1b0f74bc89cb92e50bf04d722 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/694046e6618e4bcaaab6aa142f6f67ab : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/841ef2400e444773b2e8ff4cab221ef9 : SUCCESS in 25m 47s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/150ee3880d8c4c069f040cdd847afa27 : SUCCESS in 4m 22s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/56446937dba84b5eb628c627a3fb4943 : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c0a999340464660aabf9e72bc71d53e : SUCCESS in 5m 54s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/305b5913fcfc4915aec581a9c211f915 : SUCCESS in 18m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/53e21f1d5c5e4c6a89f40075aa42afae : SUCCESS in 3m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b7324762e0246fd9caf80eaf04ca503 : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/036c6a77344f421482fe59e0e3bd4538 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":22},{"id":"87a93507c91996bc41af3eba71825aa6a31e795c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 06:27:16.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"c2008f1a25281d52d574bca78fbe9efc283cc13f","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 06:31:39.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"e5c046f9bcefe84cbbda75080414af51a6af6eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 08:46:09.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed4ba784cd44edea25db5705af27848\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8face961cd24bb7ad9c97ec2e1f3887 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d28329d576a4dc8b35902be3434f58f : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d8a4992062940c0839fd42106fb9be1 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/663b5c6c13724afa9c1e780586c12fe5 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1b5f166bd3f460388b9f0d46913467a : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/384fed2b54e64a079d82b6049d8f8c49 : SUCCESS in 2m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71082f55810449a0b22bfe966d7a5e2b : SUCCESS in 2m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/132624cd2f0f4c7ebaba9f7420a476b8 : SUCCESS in 5m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/af4b33f08ba04f518ada381b5d8845db : SUCCESS in 3m 13s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/da3747ea54e84987b25ded48cb2e63ce : SUCCESS in 7m 22s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/498f8eaf9eee4974b81974f90a8e9740 : SUCCESS in 4m 10s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/6f3dd96a45a4423fb49552b750b1e124 : SUCCESS in 4m 20s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1bb2bff7b349460fae54e28b69890751 : SUCCESS in 4m 23s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/560e211bb16a4ff59cb32c4005411099 : SUCCESS in 6m 53s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/885af874f55a48a485f19c3f7c4b0504 : SUCCESS in 4m 25s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/45ffe47a23954685a475e769f473f184 : SUCCESS in 4m 56s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/f8b753187c734fbb82c85af46ed0613b : SUCCESS in 4m 51s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/33ebc3da7a5f49199b79f7aba880b2c5 : SUCCESS in 16m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/371dc189e3194027aa6d6112ddbe52f4 : SUCCESS in 7m 28s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/cfb6fc2297d645ea9fe1ba3eaad27a3a : SUCCESS in 7m 42s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b3292ec0c2bc415095faa0383790af40 : SUCCESS in 11m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a2337d31ebc24147b582a8b9a57882a5 : SUCCESS in 2m 21s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d1264275b6be4bada79b7e0a363f5c83 : SUCCESS in 9m 20s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/6ddd51f516334d30a3c806f4bd942be8 : SUCCESS in 2m 45s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2eb2a632bb864d10b758136d9f6e0dd7 : SUCCESS in 5m 14s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/eeb017897d30405683749edf8bab84f1 : SUCCESS in 8m 54s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d8c4d73044b54eb5a58aae746e9d3359 : SUCCESS in 26m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/84650ae3bde34abd8e7da8bdd7a429cc : SUCCESS in 6m 09s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6b0e2762c6a449d59fcf767700dd836f : SUCCESS in 10m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2ef5e8d168ed44258c710e79eec06f4f : SUCCESS in 4m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ef6436cc15b74b6d97bb12f411d0a1da : SUCCESS in 2m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/892e7c92cb75467db10f8f1072ed6118 : SUCCESS in 24m 57s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2842e5977e62477a87208ac3ae23af7a : SUCCESS in 5m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/9ab56ce1c52145f689fbfd508887b7a9 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9eb6020d4e734b87a0d732775aee1562 : SUCCESS in 6m 25s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/2c9fd8a7f52a40ebbba3c2105ddb2508 : SUCCESS in 7m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/48d3ff2f0d684d639206277d421d23dc : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4064a3739c5468e8233060ec41e82c5 : SUCCESS in 33m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72479246c2f74c4bbdd68edceb12acc2 : SUCCESS in 2h 08m 45s","accounts_in_message":[],"_revision_number":23},{"id":"befaca2ea9edc1f0fa1b001a296c0036160bb99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 10:29:50.000000000","message":"Patch Set 24: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"20cf5b3b2bb370a139ce8c4def05f33f59152d47","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-18 10:31:28.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"d39faad9883b513e9c547d8edfb2c52c1ed1de4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 11:27:37.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c791e27d1c944c60ad6dc96edd7c0f33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429f1273b65146a89598ba855cb924fb : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1785fb83f8d4ea09fd884bc05f6409f : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24303a77422245d0949efb4c7ab1f70b : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/862db7d503854990aa7e4795ca94445f : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48b5b0f1ff334bd78602a7e8a9332770 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4367da4234f4e04813f83a2dcd29f27 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b694467f77a74fb5b60bd517e6f0d080 : SUCCESS in 2m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b5c8d6640e574d38a47e8dd3cc5724c6 : SUCCESS in 5m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2760aa8d364c483393c0d755d4c89193 : SUCCESS in 3m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1d62d6ef750b4a16852c6915c37f057a : SUCCESS in 5m 23s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/f512499cca154d48b8a106a091db5866 : SUCCESS in 3m 08s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3b52e51b68c14788b082383688d007a9 : SUCCESS in 6m 23s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/9b18569579824e8182d0daa82a3ac93f : SUCCESS in 5m 13s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a946c38738ab48cdb82f034ca2debf43 : SUCCESS in 6m 35s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/545d26bb36ed4d458d5a05337613194b : SUCCESS in 4m 42s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/9bcb607e659b4669894d752aaeb9d557 : SUCCESS in 3m 09s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/76e1db6cfe3b4efdb0eeb9db53b0e89c : SUCCESS in 5m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e9bc5040ae4c4df3a79bf0757b58d6e1 : SUCCESS in 17m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d9df9f4ce944a09bdc2b11f5ddcd925 : SUCCESS in 6m 26s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a641076e84404452b061ac12d3006040 : SUCCESS in 8m 56s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8e335c899442487682341764f849af01 : SUCCESS in 11m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/13a2b51666424534b0f1087d45f49fae : SUCCESS in 1m 49s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/27d78a538b1544d9abac266b50807224 : SUCCESS in 10m 02s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0e999c5828e24c1996bc1cd6483fccbf : SUCCESS in 4m 09s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/99bc5d977c7a4f739e1e31fd777ee72f : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b7b3b7d03aa0403cba99d00731185c13 : SUCCESS in 10m 27s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/efa5ba33317b4bc2adab44696598c1b2 : SUCCESS in 26m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/57d358a9a714419eaca318825e580f63 : SUCCESS in 6m 33s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/fafd2970f5f74ed6a4b373cc64156348 : SUCCESS in 15m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/197f32c2c0994309ad570059aec1199c : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb5736bc2d214ec8a53c717ec1160d5d : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c683069231a64656936fd4ea4faafc64 : SUCCESS in 25m 10s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/2eff9e0bfd1d4a38a0427db1d91d48c5 : SUCCESS in 6m 29s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f028c05e038143d9814c77ec52c6fcf3 : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec08f85c6dfd4ffe96c8f9a8de46f956 : SUCCESS in 6m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/990b00a56e054ff0b5f019aff1558f63 : SUCCESS in 18m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bb995076515b4dc08a52fa4bf86aa373 : SUCCESS in 5m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485d3026bdd24cb8a6efb976a2233567 : SUCCESS in 33m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4636bb22e6a4beeb7dd577f33480012 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":24},{"id":"c9e2350fbd69a471a4cf21b2fc682b0d6a13ebb4","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-06-18 13:51:11.000000000","message":"Patch Set 24: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"db20d158db520d91091802b16dce1abb6a58a7d2","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-06-22 06:54:43.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"859a34f1bba383cf44424d3262537cc739d96d03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-15 14:09:53.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nThe following files contain Git conflicts:\n* upper-constraints.txt\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"1a3874799d26336c5039daf08997a73ed6200d1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-15 14:16:14.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a8c278cc1f687e9d3b0e2f90d75ac75b8b5e9e84","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-15 14:16:41.000000000","message":"Patch Set 26:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":26},{"id":"f17b2163126ac188e6adcb2634cd5369d55c3807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:31:47.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220e55b766214da8a46e0ef4fc4a6c89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a777119b6964585848d8e6960b5eb9e : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0f0a58b62c413d8e57f95a5fb20edb : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0f97820a2e342cba98ee080f60ea6ed : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e400536c3a14414a8df1380fb34fa93 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34394675a50447ca843f240bf9e8107a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b1c23cf3678453dbb7cc078ab3489c3 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/484d987e22994fc28c8147462a4b92c5 : SUCCESS in 2m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/75d81f2791fd4d09af7ed37f04b45fdf : SUCCESS in 3m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ce30e51093e94757ae5fb76b3d696d7a : SUCCESS in 5m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8547d06eae8a4d70ad550f79bb268ed1 : SUCCESS in 9m 55s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/5ac603dbba4044a6833c2e2d012dc040 : SUCCESS in 3m 56s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/93a665055ddc42efba7e398639c201c3 : SUCCESS in 4m 38s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/6caa6c7ca26648b3b170e3184631b638 : SUCCESS in 5m 35s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/dfaffe312a3149fa94ef6576c3183104 : SUCCESS in 10m 21s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/edb0638236834dc9970b2e53ed6b2c03 : SUCCESS in 4m 36s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/31f37a4239ac4cfcb8ab485ca1df77b8 : SUCCESS in 5m 01s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/824ced79c79d4d7e86044915f1335b20 : SUCCESS in 8m 01s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a4563ffacc1040beb19c0e949bdb76a0 : SUCCESS in 16m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70a3e86a849b4fbfa38780163a7e93a7 : SUCCESS in 7m 03s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b50a0483ae49413883a5a1d314db5560 : SUCCESS in 8m 00s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d5d69effc0b743cea6e4490c88ef823e : SUCCESS in 11m 13s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/302118556129419aa5c5ca040d59f1ea : SUCCESS in 2m 24s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c48549f1fadb4e349a1a25334ac2ab70 : SUCCESS in 7m 53s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/fb4373f0662d4a419c32a88528b3f87b : SUCCESS in 3m 52s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e41f71f9faaf4f9fa45ffac8979b0813 : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/baa2f06aa5ce4cf28c4deefe2012ed85 : SUCCESS in 10m 04s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0136ec7651b64571bee55902c4d3bb6a : SUCCESS in 23m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/621d2ceaa2164e53857d1b739212513b : SUCCESS in 6m 24s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/f26d5085cf554822b6f40f51595dc42b : SUCCESS in 12m 18s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8de98a43d8b54d84b509bf9a02375620 : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55b9d105fc5d40e5ada29dc7c797bd08 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88b2a3b1b0c24def9637e96dd87fbc8e : SUCCESS in 26m 01s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ec6ad7099ca74c0381e9e2e8d1441ac4 : SUCCESS in 6m 09s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/037c993f011f41df9f9bdae30f7bcf37 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9f77389be5e42b08e7685b5bfc196bd : SUCCESS in 7m 52s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9cdef4994ee04f34ad7cc24a802b471b : SUCCESS in 19m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/498cd7f66ebf43af97092e34368fa20d : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe07535396e940f48218b2d1fe26d9ec : SUCCESS in 1h 06m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5be99060675a475cb515d938e77e23d8 : SUCCESS in 41m 35s","accounts_in_message":[],"_revision_number":26},{"id":"339d21b9ef99d1401dc7f2a8f1353943d46f1f14","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-16 12:46:39.000000000","message":"Patch Set 26: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"d46e835e1b01f9828729cff261bf7109ccac0c89","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-24 03:53:27.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"53acdddbab1a157000a96645741c204969f4d766","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-24 13:48:27.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"567f97294f874109fd9bcca8fe0de4259e8b06dd","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-27 04:10:41.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"5cc577771daa89b076a262050cd51b772c9da7f5","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-27 05:11:19.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"89a32cc7ec7a0f1ca5609f2008aa9686cff89c17","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-07-29 14:41:08.000000000","message":"Patch Set 26: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"6808a14b038cc0160cedf1784e21d782f734abdf","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-30 09:25:50.000000000","message":"Patch Set 26:\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"6df7c6f57aa0de9ea8389c5e39c6d88726dbadbb","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-31 03:44:04.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"ad17cdd630bc2ef881a30dc7ff525279d58f5a7b","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-07-31 19:40:21.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"13021e011febfa6676413e583bf641907ad2ccbb","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-03 05:15:36.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"b5e0a9a6679d6b85436743ff3a996c47a63e09b4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-05 14:22:32.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"44e509ce2729f402c964d0572dc06817e9844e78","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-06 10:57:35.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"50dd65af134717f42a1b51761cb519cc26213620","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-06 11:02:33.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"53fff8ab193bb90b370761aca78a231b3d4f7791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-10 09:18:37.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"e4ef7f59ac35fb5212dd0f608c0424b8073de302","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-10 09:24:11.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"9f07269359ec3d5ada748e61716bdb01e7d0a8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 11:00:15.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/575dddc612b54baa9587a9e9b1b5ae35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda518b5698d4504bbbbd891df241560 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf44e6c67f3547c3abd27454f7e62456 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d3efd594a5b4c6eacb840fc192c8a09 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fddded0c9804e5e84829532fd05c14c : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0cc6b8825c3a44108cc40c989600343b : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbe927fd637d4916bb7a29e8be023d42 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4184eb5a1bc440ba334c82100a904d5 : SUCCESS in 4m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2399d814631d4baab63d91f46871519c : SUCCESS in 5m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/28557bd49ce34d3b9c11064ff35bb9ee : SUCCESS in 5m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/20de8c34c93f4440970bf1d65210ff91 : SUCCESS in 9m 45s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/16420d76095e472888fa01135e1dc310 : SUCCESS in 6m 09s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8574e419dc314629ac04b31c03a5fe6b : SUCCESS in 6m 04s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e82516f3f095432d8c5aa1a7074bd114 : SUCCESS in 5m 02s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/192ea0251b4246f582a04082a7bad319 : SUCCESS in 9m 38s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/9e379bd49b9e434985964001f14e0b32 : SUCCESS in 4m 26s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/bc5b3d6b523840a1a52cc8b66813d61a : SUCCESS in 4m 31s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/007a357d8d4948dda103929925876a47 : SUCCESS in 5m 43s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7eb7656c71044a4ab133a3830fecced8 : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/368ce81ef4f344888d97d15a1bface85 : SUCCESS in 5m 50s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b48a55ce22bc4fe8941f6cb43f1c7e8a : SUCCESS in 7m 15s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e135388b64f24b1798497c247ce4d31f : SUCCESS in 12m 03s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/88a50bad58fa4d96b3dae05fe6bd7970 : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/1fe27f3f679442838490ba56f41f650d : SUCCESS in 7m 43s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/71ce4b7fe40a4aa1b283e533c0525ccf : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/0b340f8c4f5945d39b5cb455e98faa0f : SUCCESS in 5m 54s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/11c98403eaf64382b2f0c5e79da92335 : SUCCESS in 9m 06s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/dc22b71655684c68bac0dfac3bbac0a1 : SUCCESS in 26m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/acd2562e27c347de837f0a63e121a307 : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0160b2ab9c5e42e8b3d73dec99db267e : SUCCESS in 12m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6eb2939c3a1940749f97b1520046d538 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f09fb3bf50242bfa3694b1f9afdf674 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34659595940346e5850a3af532475d02 : SUCCESS in 24m 11s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/ddf1d7b7bd964947a64bddbacc863dee : SUCCESS in 5m 51s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3978f032a15d40938f6eeb8c93c0ad6a : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ec8bb33ed2c42b2990ee263fdc7b238 : SUCCESS in 6m 45s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/db59a146387c4bdc8d1743ed61b13d7b : SUCCESS in 16m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0d1730c71d8140be881f07704e673a40 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42dfcdd83daf4c16b9d395c56da32386 : FAILURE in 55m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9c8fc8fbdbe4f73a72ffacc60032d26 : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":27},{"id":"dafd049fb08401ca8b5542ace5b6cf31ab1dfc47","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-11 06:58:56.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"1582194a4bbab78a071e8d8b4ee70585f3ee0ef8","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-12 14:24:33.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"5694c20f2a5430d3614837a43fd881250118da53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 15:11:51.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df00fa09ba8744079063f0c17b28aaeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9c880db7dd49369ee5d960a92817f7 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfdcf1dada444adae44e19d9a8b18e7 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9de13dd4420a4120908fa458b95bc67b : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c391ec4c19a54843ace98734de21b54a : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2cb1fbb63ef450081c9ce8881b37978 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d56d0b88513e47fcbdb41d896e3c6a55 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d99f2a4925d84f79856941d450cbba3f : SUCCESS in 3m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/df7eaaf71a5541139262ea7b41ade559 : SUCCESS in 5m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e5bd0310d0944c08ad62adab504145bf : SUCCESS in 3m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/86edef30d1374776a44add6bddb744af : SUCCESS in 5m 19s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a9c1542ea41d4881b6626a4a9c10dffa : SUCCESS in 6m 13s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/95c456d2fff343ad8f0be12384f5a8ee : SUCCESS in 3m 34s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/cc80c0d43ae44f34bc5aec8b3c8a442f : SUCCESS in 4m 04s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f4aea9bd6b954861bd8ad3fb98f2c730 : SUCCESS in 8m 00s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/90d6c97bfdf2470b865dc9f3681e098e : SUCCESS in 4m 45s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/4a6e247f904740fcaa07509cdc8e6d69 : SUCCESS in 6m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0184f590d0424bf9a36c4e8780e87e68 : SUCCESS in 7m 05s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/af78db34ecc54aa8b3968c7e218e08bb : SUCCESS in 17m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f393f5bd15440aa927682e4974804c3 : SUCCESS in 8m 37s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/fc31f00bf8084408bb435a270eff8f8d : SUCCESS in 5m 03s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/b990908efc734bd29d6f97a0aeaa9e69 : SUCCESS in 9m 48s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/76e5bf9b443a42aea5b102556a7340a4 : SUCCESS in 1m 43s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d2888dbb619e49f4bc518f8fd02a657e : SUCCESS in 8m 31s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/91bd6f75e9244cd19247a96b434b8aaf : SUCCESS in 3m 10s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/eb641ed0660d482489822dbdd37d3eca : SUCCESS in 6m 23s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/60ba9f33e1c848079e94ad797888af11 : SUCCESS in 9m 26s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7b0210753a03492a8384c539889cb323 : SUCCESS in 16m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc1259ccad7044deac247b28336e84ee : SUCCESS in 3m 31s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b27107e5447f4facacf5ee69922de1d5 : SUCCESS in 12m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c973f50b982e4f7596b58d5f62415c54 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a7c2415f55247c986e4e8ef9053c709 : SUCCESS in 2m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7208bd61415a41938afe672668781024 : SUCCESS in 25m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/801edb0fc65643e7b490de1e1e3e28c1 : SUCCESS in 3m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/1f9ed8c3e8514053af57b7bcfe3d8d48 : SUCCESS in 3m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/968f9b6456654788a12dbce6b6b7efbc : SUCCESS in 3m 41s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/7f419c0a4b5c4e61ba97f941ca4ac9d3 : SUCCESS in 8m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/841a943df2cc436783ede20864796893 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3049c53eee54acf8a9d0947d05ce286 : SUCCESS in 42m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c8d742860034da39d98ddd61e5b4dc3 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":27},{"id":"f2d18cf2fc49f17b4c05b7bd76260285dd5dbbda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-13 08:11:30.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"97ea01246d0f556295f868c2408fa80ec1cf0ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 09:50:10.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63f9cf4811b74bd1a4030d3b081fe064\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd5af68b389442269c961755f3bb1459 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5561c62cf2f346ab8aa0553cd7b279e0 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff1249fb032642d3882ffdcdd17621a5 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3a3c4592f464edfbaaeed38588ee190 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f589a36c861412b946ad7a80ce4b51f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53ae1a3f52d449fb8f6beca06491205d : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82f1a6548b2e479889c87be38b4d08a2 : SUCCESS in 3m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d160ef0ea23f4875bbab9d8fa1dabe49 : SUCCESS in 5m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9b44282a3d87491a9a18870e3ded6321 : SUCCESS in 6m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/20fc3e1fb37f4769bc6fed2adfbd8761 : SUCCESS in 7m 44s\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ebd81d769b21404795ef27dfcc8de94c : SUCCESS in 5m 00s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/1402adc9220e4f98ab024a3b804984c9 : SUCCESS in 4m 37s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/8a2ec42d15044f61b118a1329c204188 : SUCCESS in 5m 45s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6ca9d4f7d8b1405c8920c4340ecf8267 : SUCCESS in 9m 30s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/06379773e6b14c45b09acb6c53df3c72 : SUCCESS in 4m 41s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0042aad28da749978e43b22bee6e3c28 : SUCCESS in 4m 35s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/58441db23abf4b1890afacfad486f423 : SUCCESS in 6m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/9790c96f1c8c4a1f966b2abd5db0a960 : SUCCESS in 17m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a8213791969743af8764e61a0aa04f97 : SUCCESS in 7m 42s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c135860f93ca4534b23e756fce4175e8 : SUCCESS in 7m 14s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/63d5b1e030634883bc5a566bb975178e : SUCCESS in 10m 30s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/4f1a29f06ff54e168e8e7b0e2c0d42f3 : SUCCESS in 2m 41s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/498bc5c555d147bfa9c2e383c1b1506c : SUCCESS in 7m 39s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/895b7cdafdd04eb0b20fdb6743403a9a : SUCCESS in 4m 12s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/c2c7f298699d4c0aa5893989ad91d96c : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/d33d3f87336748a89f00b81a7f16d547 : SUCCESS in 11m 21s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8cd208422c884e35a752d96ed9e1063a : SUCCESS in 28m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c8a5bcedac224f8082daced9af5fc271 : SUCCESS in 6m 03s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0d501aef61e241af9cd840d80fca420d : SUCCESS in 12m 45s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/a8bb64caad8e4ba6879558882fe60fe3 : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3cdbb6984334d6a9dc9c13a59138ae6 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c45dcd72eab34d8197a158e8cb337dbd : SUCCESS in 28m 06s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/adab0cf4c24240f38a8f722255ec06d6 : SUCCESS in 5m 22s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/fda5dcdc19cf47709a08e7ec9deb5e7f : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f337fe0e00a14563b85c19be8ec34934 : SUCCESS in 7m 28s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/fcd645a022b34957883d691587e55a64 : SUCCESS in 17m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0beea6bbf827444ba972ebdddaf6bd83 : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b64b66340729455fbd10bdc3c437da55 : SUCCESS in 1h 00m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc3dcb30dcdb4bbdaecbabb3b8331359 : SUCCESS in 1h 32m 29s","accounts_in_message":[],"_revision_number":28},{"id":"8cdb0f207c21a5b24a3558dcd940ce1fc61b6e06","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-08-13 11:23:44.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"f0ebfeba81109d8577fac334fb653b5304cca987","author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-08-13 11:35:55.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"5847057664fdbef2f712bbe610f5db4cdcdb3fba","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-08-13 12:59:01.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"841b5e22f03446edededd8fcdc782bbff1e67149","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-13 15:35:17.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"5e0168eca7c333293983677ae9606678de3d5f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:35:58.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"339374fc097f7894b759a6aa5f88d51cdc606f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:16:17.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8279ce2d38dd46e2a3149aad228c1f88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb188b78b72c464e8e94b04c816fffbd : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eab800e0ff34705a06203d252d4b89f : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c299776aea1148f88008679567b5b16d : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c60f5b3497ca48fc8e16ba2eafaaf0e3 : SUCCESS in 2m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4758d531cd34475a5bd58b7de4d9128 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcf4cb736fe84832a25c7b8ab462fc8e : SUCCESS in 3m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b8d034d4b29447b2ab558de79e350fa3 : SUCCESS in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/87da9fa0d98f4efb86744395b46bf431 : SUCCESS in 9m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9345c0f61d1b42abb4ffdbfbcd75df35 : SUCCESS in 3m 45s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/35dcaf42b97846b8b4486749bcea1e8a : SUCCESS in 10m 27s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ec3f251e372849859b5daf6d44673d40 : SUCCESS in 4m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d29155c0544c4a85826063e48fd7e460 : SUCCESS in 2m 52s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/10017fd75ae446d08b56b139740b4185 : SUCCESS in 7m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7f0d77f2109f4fc8bfdc5125a5e355ba : SUCCESS in 17m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca54b733222843d48be490a34171e782 : SUCCESS in 6m 18s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/840c0aa20c7e462fa0f569a76545f8b8 : SUCCESS in 7m 30s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/9599ce906acf429bbd40768ac4d109f6 : SUCCESS in 8m 46s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/7cbb297055d54920b9880f4c6a3127a1 : SUCCESS in 2m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/39edb243d9284efb867a94f033594993 : SUCCESS in 5m 00s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c1524fb1653f4c2da3adf6ec413abcba : SUCCESS in 25m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5e0329728cef46e89d97d54d080ac01d : SUCCESS in 9m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0524ddd577944326a0ba5163dfdbc237 : SUCCESS in 10m 20s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/24d184dddf9544a69365d6ddd1b5d197 : SUCCESS in 2m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d61e10d5a25f4f8cadd70da27fb0f3f5 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70177079970b495c99b76c0cea119f58 : SUCCESS in 28m 49s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/3a3f97a4229a466881be99899e905a92 : SUCCESS in 5m 44s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/d768f08ac3354ef190f18976a0d5a7f9 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05008fe600fc4900b296d323b00e309f : SUCCESS in 7m 00s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/ccab073831954ce5b406269927ef3df1 : SUCCESS in 15m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fa6da6349c764e30bb2742e0c34c7f9c : SUCCESS in 4m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31571f648838419b81a1cc30abd7a2b3 : SUCCESS in 1h 02m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28977f296ebd4eceaf7cf66517ca9412 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":28},{"id":"b1177de3539076b78b41616d459fdfb8df427c4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:16:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"91f9b5ca4deb296cd2fc564a70c5e8b8f3644deb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:17:43.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba89104d9fd747f68943b396ab83fe91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56b1541c3d94aa298a4dbbbd4c6f233 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"120c032cb9bb5573d909c386374e4ee6bf971098","revisions":{"e3c976d671d48c8f941676ed6ac4476fcb6de67d":{"kind":"REWORK","_number":1,"created":"2026-01-22 14:01:24.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/1"}}},"commit":{"parents":[{"commit":"81d14c423cd4fe2f7e6d020bfb758dce71c763d9","subject":"Merge \"update constraint for python-heatclient to new release 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81d14c423cd4fe2f7e6d020bfb758dce71c763d9"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 14:04:18.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit library","message":"Add cinder extras hpe-storage-flowkit library\n\nThis patch adds the cinder extras hpe-storage-flowkit library\nfor the cinder HPE Alletra MP driver.\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c976d671d48c8f941676ed6ac4476fcb6de67d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c976d671d48c8f941676ed6ac4476fcb6de67d"}]},"branch":"refs/heads/master"},"c9d152e65ada82766b025df6392c06e33d8ccb81":{"kind":"REWORK","_number":2,"created":"2026-01-23 04:50:27.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/2"}}},"commit":{"parents":[{"commit":"81d14c423cd4fe2f7e6d020bfb758dce71c763d9","subject":"Merge \"update constraint for python-heatclient to new release 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81d14c423cd4fe2f7e6d020bfb758dce71c763d9"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-23 04:53:05.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit library","message":"Add cinder extras hpe-storage-flowkit library\n\nThis patch adds the cinder extras hpe-storage-flowkit library\nfor the cinder HPE Alletra MP driver.\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9d152e65ada82766b025df6392c06e33d8ccb81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9d152e65ada82766b025df6392c06e33d8ccb81"}]},"branch":"refs/heads/master"},"9ab952f905ba699ae4ebbe888541adb387b920e2":{"kind":"REWORK","_number":3,"created":"2026-02-11 10:15:31.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/3"}}},"commit":{"parents":[{"commit":"81d14c423cd4fe2f7e6d020bfb758dce71c763d9","subject":"Merge \"update constraint for python-heatclient to new release 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81d14c423cd4fe2f7e6d020bfb758dce71c763d9"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-11 09:54:31.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit library","message":"Add cinder extras hpe-storage-flowkit library\n\nThis patch adds the cinder extras hpe-storage-flowkit library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ab952f905ba699ae4ebbe888541adb387b920e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ab952f905ba699ae4ebbe888541adb387b920e2"}]},"branch":"refs/heads/master"},"f607ab6d3eb00a20d471efd01b085dd896d4ef49":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-20 04:11:35.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/4"}}},"commit":{"parents":[{"commit":"81d14c423cd4fe2f7e6d020bfb758dce71c763d9","subject":"Merge \"update constraint for python-heatclient to new release 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81d14c423cd4fe2f7e6d020bfb758dce71c763d9"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-20 04:13:01.000000000","tz":0},"subject":"DNM - do not merge - Add cinder extras hpe-storage-flowkit library","message":"DNM - do not merge - Add cinder extras hpe-storage-flowkit library\n\nThis patch adds the cinder extras hpe-storage-flowkit library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f607ab6d3eb00a20d471efd01b085dd896d4ef49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f607ab6d3eb00a20d471efd01b085dd896d4ef49"}]},"branch":"refs/heads/master"},"6aea502b3428f83f75465772240e8d9542ba8ec7":{"kind":"REWORK","_number":5,"created":"2026-03-12 11:14:41.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/5"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-03-12 11:14:41.000000000","tz":0},"subject":"DNM - do not merge - Add cinder extras hpe-storage-flowkit library","message":"DNM - do not merge - Add cinder extras hpe-storage-flowkit library\n\nThis patch adds the cinder extras hpe-storage-flowkit library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aea502b3428f83f75465772240e8d9542ba8ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aea502b3428f83f75465772240e8d9542ba8ec7"}]},"branch":"refs/heads/master","description":"Rebase"},"b67eb895857149493a0c5b1ee8867514906507c6":{"kind":"REWORK","_number":6,"created":"2026-03-12 12:11:55.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/6"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-03-12 12:15:03.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b67eb895857149493a0c5b1ee8867514906507c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b67eb895857149493a0c5b1ee8867514906507c6"}]},"branch":"refs/heads/master"},"2fec39ddb46290ecd0d16e7710d96cf3808c323d":{"kind":"REWORK","_number":7,"created":"2026-03-12 13:13:03.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/7"}}},"commit":{"parents":[{"commit":"bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb","subject":"update constraint for horizon to new release 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb067082abc8bf3676a8c7e6ae8fa91a2c84b9cb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-03-12 13:16:02.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fec39ddb46290ecd0d16e7710d96cf3808c323d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fec39ddb46290ecd0d16e7710d96cf3808c323d"}]},"branch":"refs/heads/master"},"ec226929c9774a66020e9c54c810431074cba19b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-03-31 05:43:56.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/8"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-03-31 05:43:56.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: in PyPI\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec226929c9774a66020e9c54c810431074cba19b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec226929c9774a66020e9c54c810431074cba19b"}]},"branch":"refs/heads/master","description":"Rebase"},"e42f26447f05b3d4ee694da5e640855b3fa23720":{"kind":"REWORK","_number":9,"created":"2026-04-06 07:36:41.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/9"}}},"commit":{"parents":[{"commit":"2897d874a900d28bbeb8ed59a78193477c76296c","subject":"Merge \"Make cross-barbican job voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2897d874a900d28bbeb8ed59a78193477c76296c"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-04-06 07:39:42.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e42f26447f05b3d4ee694da5e640855b3fa23720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e42f26447f05b3d4ee694da5e640855b3fa23720"}]},"branch":"refs/heads/master"},"1ebfba99017337cbff569e6342a69d9c45ad3b6e":{"kind":"REWORK","_number":10,"created":"2026-04-20 09:49:23.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/10"}}},"commit":{"parents":[{"commit":"954a2cc76030c7154969416f0cdfe1568f986c37","subject":"Merge \"Add the Sphinx copy-button\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/954a2cc76030c7154969416f0cdfe1568f986c37"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-04-20 09:49:23.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ebfba99017337cbff569e6342a69d9c45ad3b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ebfba99017337cbff569e6342a69d9c45ad3b6e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"e42f26447f05b3d4ee694da5e640855b3fa23720","theirs":"954a2cc76030c7154969416f0cdfe1568f986c37","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"3bfbd889cfd086c7d43e7ab782e3530d05376345":{"kind":"REWORK","_number":11,"created":"2026-04-20 09:57:19.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/22/974322/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/11"}}},"commit":{"parents":[{"commit":"954a2cc76030c7154969416f0cdfe1568f986c37","subject":"Merge \"Add the Sphinx copy-button\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/954a2cc76030c7154969416f0cdfe1568f986c37"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-04-20 10:00:19.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bfbd889cfd086c7d43e7ab782e3530d05376345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bfbd889cfd086c7d43e7ab782e3530d05376345"}]},"branch":"refs/heads/master"},"e8edb6f42094d71d0cc91479c1f812d59672bd3e":{"kind":"REWORK","_number":12,"created":"2026-05-13 11:25:01.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/12"}}},"commit":{"parents":[{"commit":"6015fe784a11a9066457cf45795423ca585443bb","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6015fe784a11a9066457cf45795423ca585443bb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-05-13 11:25:01.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8edb6f42094d71d0cc91479c1f812d59672bd3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8edb6f42094d71d0cc91479c1f812d59672bd3e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3bfbd889cfd086c7d43e7ab782e3530d05376345","theirs":"6015fe784a11a9066457cf45795423ca585443bb","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"bc99902b8f245fecd26c7a2c88b1c6ce74031173":{"kind":"REWORK","_number":13,"created":"2026-05-13 11:35:03.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/13"}}},"commit":{"parents":[{"commit":"6015fe784a11a9066457cf45795423ca585443bb","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6015fe784a11a9066457cf45795423ca585443bb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-05-13 11:34:29.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc99902b8f245fecd26c7a2c88b1c6ce74031173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc99902b8f245fecd26c7a2c88b1c6ce74031173"}]},"branch":"refs/heads/master"},"515fd2574a3533ecb2934d3f434108c1a1ee5fcb":{"kind":"REWORK","_number":14,"created":"2026-05-14 07:06:37.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/14"}}},"commit":{"parents":[{"commit":"6015fe784a11a9066457cf45795423ca585443bb","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6015fe784a11a9066457cf45795423ca585443bb"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-05-14 07:05:52.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/515fd2574a3533ecb2934d3f434108c1a1ee5fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/515fd2574a3533ecb2934d3f434108c1a1ee5fcb"}]},"branch":"refs/heads/master"},"b913cb51d8842947f1b69fb009f67a16086e440d":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-05-14 08:43:34.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/15"}}},"commit":{"parents":[{"commit":"655789d95ab02ab40f527fbd780b982c29bd2bf6","subject":"Merge \"update constraint for keystoneauth1 to new release 5.14.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/655789d95ab02ab40f527fbd780b982c29bd2bf6"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-05-14 08:43:34.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b913cb51d8842947f1b69fb009f67a16086e440d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b913cb51d8842947f1b69fb009f67a16086e440d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"515fd2574a3533ecb2934d3f434108c1a1ee5fcb","theirs":"655789d95ab02ab40f527fbd780b982c29bd2bf6","contains_conflicts":false}},"f8760017eea30e8a827c87590cce04c4f16d2868":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-06-03 09:48:57.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/16"}}},"commit":{"parents":[{"commit":"6f76e6442a25143e43bbd71cea2332f2e61e2187","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f76e6442a25143e43bbd71cea2332f2e61e2187"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-03 09:48:57.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8760017eea30e8a827c87590cce04c4f16d2868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8760017eea30e8a827c87590cce04c4f16d2868"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b913cb51d8842947f1b69fb009f67a16086e440d","theirs":"6f76e6442a25143e43bbd71cea2332f2e61e2187","contains_conflicts":false}},"397630b47dfe55dd1e31c41266170d268c63fd58":{"kind":"REWORK","_number":17,"created":"2026-06-03 09:53:45.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/17"}}},"commit":{"parents":[{"commit":"6f76e6442a25143e43bbd71cea2332f2e61e2187","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f76e6442a25143e43bbd71cea2332f2e61e2187"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-03 09:52:07.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/397630b47dfe55dd1e31c41266170d268c63fd58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/397630b47dfe55dd1e31c41266170d268c63fd58"}]},"branch":"refs/heads/master"},"586d1e9e9b0540a3c410a084bc7be863917eab99":{"kind":"REWORK","_number":18,"created":"2026-06-03 11:27:18.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/18"}}},"commit":{"parents":[{"commit":"bdc4e18ff354825af258288e11491b4e2ce17f35","subject":"Add playwright","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc4e18ff354825af258288e11491b4e2ce17f35"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-03 11:27:18.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/586d1e9e9b0540a3c410a084bc7be863917eab99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/586d1e9e9b0540a3c410a084bc7be863917eab99"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"397630b47dfe55dd1e31c41266170d268c63fd58","theirs":"bdc4e18ff354825af258288e11491b4e2ce17f35","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"53d4de2bca0efe41c7b8d6270ea0c12dda7d8ceb":{"kind":"REWORK","_number":19,"created":"2026-06-03 11:31:17.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/19"}}},"commit":{"parents":[{"commit":"bdc4e18ff354825af258288e11491b4e2ce17f35","subject":"Add playwright","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc4e18ff354825af258288e11491b4e2ce17f35"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-03 11:30:15.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d4de2bca0efe41c7b8d6270ea0c12dda7d8ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d4de2bca0efe41c7b8d6270ea0c12dda7d8ceb"}]},"branch":"refs/heads/master"},"7b91a351d0472bc9df552844c71a6f99e8b7cf02":{"kind":"REWORK","_number":20,"created":"2026-06-09 06:52:00.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/20"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-09 06:52:00.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b91a351d0472bc9df552844c71a6f99e8b7cf02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b91a351d0472bc9df552844c71a6f99e8b7cf02"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bdc4e18ff354825af258288e11491b4e2ce17f35","ours":"53d4de2bca0efe41c7b8d6270ea0c12dda7d8ceb","theirs":"df8c1d8697a531140de152d548ae02b731e1a235","merge_strategy":"recursive","contains_conflicts":true}},"3a6b8121970d23665da2f5837d1f7a3ed1d48f86":{"kind":"REWORK","_number":21,"created":"2026-06-09 06:57:13.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/21"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-09 06:56:00.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a6b8121970d23665da2f5837d1f7a3ed1d48f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a6b8121970d23665da2f5837d1f7a3ed1d48f86"}]},"branch":"refs/heads/master"},"64cb1338365836ad556604c25c3f937a8e1a8d30":{"kind":"REWORK","_number":22,"created":"2026-06-18 05:19:36.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/22"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-18 05:17:47.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cb1338365836ad556604c25c3f937a8e1a8d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64cb1338365836ad556604c25c3f937a8e1a8d30"}]},"branch":"refs/heads/master"},"73588d9215ec8170e5f831782ff4a3dca954d8c9":{"kind":"REWORK","_number":23,"created":"2026-06-18 06:27:16.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/23"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-18 06:25:51.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library\nfor the cinder HPE Alletra MP driver.\n\nThe library provides python functions to perform various\noperations (eg. create/delete volumes, snapshots, etc) on\nHPE Alletra MP array (block storage).\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: no\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73588d9215ec8170e5f831782ff4a3dca954d8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73588d9215ec8170e5f831782ff4a3dca954d8c9"}]},"branch":"refs/heads/master"},"a68bb538b7d6eb1b1d4fb438011bc2a608b4d76e":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2026-06-18 10:29:50.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/24"}}},"commit":{"parents":[{"commit":"df8c1d8697a531140de152d548ae02b731e1a235","subject":"update constraint for os-traits to new release 3.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df8c1d8697a531140de152d548ae02b731e1a235"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-06-18 10:29:50.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library for the\ncinder HPE Alletra MP driver.\n\nHPE Storage FlowKit (Python) is a client library for interacting with\nHPE Alletra MP storage arrays via REST APIs. It provides higher-level\nworkflows for volume, snapshot, replication, host management and other\noperations, and is intended to serve as the backend integration layer\nfor the new HPE Alletra MP Cinder driver.\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nIs the library actively maintained? Yes. The latest release 0.9.3 was\npublished on June 2026 on PyPI.\n\nIs the library good code? Yes Good, The library provides a structured\nPython client with modular API layers (v1/v3), workflow abstractions,\nand documented capabilities covering volume, snapshot, QoS, replication,\nand host management operations. The repository includes documentation,\nrequirements files, and test directories.\n\nIs the library license compatible? Yes, it is licensed under the Apache\nLicense 2.0, which is fully compatible with OpenStack\n\nIs the library already packaged in the distros we target? No. The\nlibrary is currently distributed via PyPI and upstream GitHub\n\nIs the function of this library already covered by other libraries in\nglobal-requirements.txt? No. Existing HPE Cinder drivers use\npython-3parclient, which targets older platforms\n(3PAR/Primera/Alletra9k). FlowKit is introduced to support new\nfunctionality specific to HPE Alletra MP, including higher-level\nworkflows and APIs.\n\nIs the library required for an OpenStack project? Which? Yes, it is\nrequired for the OpenStack Cinder (Block Storage) project, specifically\nfor the new HPE Alletra MP backend driver, which is currently under\ndevelopment/review.\n\nIf the library release is managed by the OpenStack release process?\nNo,the library is developed and released independently by HPE via PyPI\nand GitHub, and is not part of the OpenStack release process.\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a68bb538b7d6eb1b1d4fb438011bc2a608b4d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a68bb538b7d6eb1b1d4fb438011bc2a608b4d76e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e40839202f1d3762e432554e791e35f3465ec9a2":{"kind":"REWORK","_number":25,"created":"2026-07-15 14:09:53.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/25"}}},"commit":{"parents":[{"commit":"a051ed53de8d81ba9781d76e651a26b160c4b493","subject":"update constraint for python-openstackclient to new release 10.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a051ed53de8d81ba9781d76e651a26b160c4b493"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","date":"2026-07-15 14:09:53.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library for the\ncinder HPE Alletra MP driver.\n\nHPE Storage FlowKit (Python) is a client library for interacting with\nHPE Alletra MP storage arrays via REST APIs. It provides higher-level\nworkflows for volume, snapshot, replication, host management and other\noperations, and is intended to serve as the backend integration layer\nfor the new HPE Alletra MP Cinder driver.\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nIs the library actively maintained? Yes. The latest release 0.9.3 was\npublished on June 2026 on PyPI.\n\nIs the library good code? Yes Good, The library provides a structured\nPython client with modular API layers (v1/v3), workflow abstractions,\nand documented capabilities covering volume, snapshot, QoS, replication,\nand host management operations. The repository includes documentation,\nrequirements files, and test directories.\n\nIs the library license compatible? Yes, it is licensed under the Apache\nLicense 2.0, which is fully compatible with OpenStack\n\nIs the library already packaged in the distros we target? No. The\nlibrary is currently distributed via PyPI and upstream GitHub\n\nIs the function of this library already covered by other libraries in\nglobal-requirements.txt? No. Existing HPE Cinder drivers use\npython-3parclient, which targets older platforms\n(3PAR/Primera/Alletra9k). FlowKit is introduced to support new\nfunctionality specific to HPE Alletra MP, including higher-level\nworkflows and APIs.\n\nIs the library required for an OpenStack project? Which? Yes, it is\nrequired for the OpenStack Cinder (Block Storage) project, specifically\nfor the new HPE Alletra MP backend driver, which is currently under\ndevelopment/review.\n\nIf the library release is managed by the OpenStack release process?\nNo,the library is developed and released independently by HPE via PyPI\nand GitHub, and is not part of the OpenStack release process.\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e40839202f1d3762e432554e791e35f3465ec9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e40839202f1d3762e432554e791e35f3465ec9a2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"df8c1d8697a531140de152d548ae02b731e1a235","ours":"a68bb538b7d6eb1b1d4fb438011bc2a608b4d76e","theirs":"a051ed53de8d81ba9781d76e651a26b160c4b493","merge_strategy":"recursive","contains_conflicts":true}},"5f7cf5343415022231c2718dd0d9e245a7ec5b58":{"kind":"REWORK","_number":26,"created":"2026-07-15 14:16:14.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/26"}}},"commit":{"parents":[{"commit":"a051ed53de8d81ba9781d76e651a26b160c4b493","subject":"update constraint for python-openstackclient to new release 10.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a051ed53de8d81ba9781d76e651a26b160c4b493"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-07-15 14:14:03.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library for the\ncinder HPE Alletra MP driver.\n\nHPE Storage FlowKit (Python) is a client library for interacting with\nHPE Alletra MP storage arrays via REST APIs. It provides higher-level\nworkflows for volume, snapshot, replication, host management and other\noperations, and is intended to serve as the backend integration layer\nfor the new HPE Alletra MP Cinder driver.\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nIs the library actively maintained? Yes. The latest release 0.9.3 was\npublished on June 2026 on PyPI.\n\nIs the library good code? Yes Good, The library provides a structured\nPython client with modular API layers (v1/v3), workflow abstractions,\nand documented capabilities covering volume, snapshot, QoS, replication,\nand host management operations. The repository includes documentation,\nrequirements files, and test directories.\n\nIs the library license compatible? Yes, it is licensed under the Apache\nLicense 2.0, which is fully compatible with OpenStack\n\nIs the library already packaged in the distros we target? No. The\nlibrary is currently distributed via PyPI and upstream GitHub\n\nIs the function of this library already covered by other libraries in\nglobal-requirements.txt? No. Existing HPE Cinder drivers use\npython-3parclient, which targets older platforms\n(3PAR/Primera/Alletra9k). FlowKit is introduced to support new\nfunctionality specific to HPE Alletra MP, including higher-level\nworkflows and APIs.\n\nIs the library required for an OpenStack project? Which? Yes, it is\nrequired for the OpenStack Cinder (Block Storage) project, specifically\nfor the new HPE Alletra MP backend driver, which is currently under\ndevelopment/review.\n\nIf the library release is managed by the OpenStack release process?\nNo,the library is developed and released independently by HPE via PyPI\nand GitHub, and is not part of the OpenStack release process.\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f7cf5343415022231c2718dd0d9e245a7ec5b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f7cf5343415022231c2718dd0d9e245a7ec5b58"}]},"branch":"refs/heads/master"},"9deca88bec1737f45ecb1730f9a0ee7341709937":{"kind":"REWORK","_number":27,"created":"2026-08-10 09:18:37.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/27"}}},"commit":{"parents":[{"commit":"a051ed53de8d81ba9781d76e651a26b160c4b493","subject":"update constraint for python-openstackclient to new release 10.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a051ed53de8d81ba9781d76e651a26b160c4b493"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-08-10 09:17:49.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library for the\ncinder HPE Alletra MP driver.\n\nHPE Storage FlowKit (Python) is a client library for interacting with\nHPE Alletra MP storage arrays via REST APIs. It provides higher-level\nworkflows for volume, snapshot, replication, host management and other\noperations, and is intended to serve as the backend integration layer\nfor the new HPE Alletra MP Cinder driver.\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nIs the library actively maintained? Yes. The latest release 0.9.3 was\npublished on June 2026 on PyPI.\n\nIs the library good code? Yes Good, The library provides a structured\nPython client with modular API layers (v1/v3), workflow abstractions,\nand documented capabilities covering volume, snapshot, QoS, replication,\nand host management operations. The repository includes documentation,\nrequirements files, and test directories.\n\nIs the library license compatible? Yes, it is licensed under the Apache\nLicense 2.0, which is fully compatible with OpenStack\n\nIs the library already packaged in the distros we target? No. The\nlibrary is currently distributed via PyPI and upstream GitHub\n\nIs the function of this library already covered by other libraries in\nglobal-requirements.txt? No. Existing HPE Cinder drivers use\npython-3parclient, which targets older platforms\n(3PAR/Primera/Alletra9k). FlowKit is introduced to support new\nfunctionality specific to HPE Alletra MP, including higher-level\nworkflows and APIs.\n\nIs the library required for an OpenStack project? Which? Yes, it is\nrequired for the OpenStack Cinder (Block Storage) project, specifically\nfor the new HPE Alletra MP backend driver, which is currently under\ndevelopment/review.\n\nIf the library release is managed by the OpenStack release process?\nNo,the library is developed and released independently by HPE via PyPI\nand GitHub, and is not part of the OpenStack release process.\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9deca88bec1737f45ecb1730f9a0ee7341709937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9deca88bec1737f45ecb1730f9a0ee7341709937"}]},"branch":"refs/heads/master"},"120c032cb9bb5573d909c386374e4ee6bf971098":{"kind":"REWORK","_number":28,"created":"2026-08-13 08:11:30.000000000","uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/22/974322/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/974322/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/974322/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/974322/28"}}},"commit":{"parents":[{"commit":"a051ed53de8d81ba9781d76e651a26b160c4b493","subject":"update constraint for python-openstackclient to new release 10.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a051ed53de8d81ba9781d76e651a26b160c4b493"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-01-22 13:58:51.000000000","tz":0},"committer":{"name":"jlothe","email":"jyotsna.lothe@hpe.com","date":"2026-08-13 08:10:17.000000000","tz":0},"subject":"Add cinder extras hpe-storage-flowkit-py library","message":"Add cinder extras hpe-storage-flowkit-py library\n\nThis patch adds the cinder extras hpe-storage-flowkit-py library for the\ncinder HPE Alletra MP driver.\n\nHPE Storage FlowKit (Python) is a client library for interacting with\nHPE Alletra MP storage arrays via REST APIs. It provides higher-level\nworkflows for volume, snapshot, replication, host management and other\noperations, and is intended to serve as the backend integration layer\nfor the new HPE Alletra MP Cinder driver.\nhttps://pypi.org/project/hpe-storage-flowkit-py/\nhttps://github.com/hpe-storage/hpe_storage_flowkit_py\n\nIs the library actively maintained? Yes. The latest release 0.9.3 was\npublished on June 2026 on PyPI.\n\nIs the library good code? Yes Good, The library provides a structured\nPython client with modular API layers (v1/v3), workflow abstractions,\nand documented capabilities covering volume, snapshot, QoS, replication,\nand host management operations. The repository includes documentation,\nrequirements files, and test directories.\n\nIs the library license compatible? Yes, it is licensed under the Apache\nLicense 2.0, which is fully compatible with OpenStack\n\nIs the library already packaged in the distros we target? No. The\nlibrary is currently distributed via PyPI and upstream GitHub\n\nIs the function of this library already covered by other libraries in\nglobal-requirements.txt? No. Existing HPE Cinder drivers use\npython-3parclient, which targets older platforms\n(3PAR/Primera/Alletra9k). FlowKit is introduced to support new\nfunctionality specific to HPE Alletra MP, including higher-level\nworkflows and APIs.\n\nIs the library required for an OpenStack project? Which? Yes, it is\nrequired for the OpenStack Cinder (Block Storage) project, specifically\nfor the new HPE Alletra MP backend driver, which is currently under\ndevelopment/review.\n\nIf the library release is managed by the OpenStack release process?\nNo,the library is developed and released independently by HPE via PyPI\nand GitHub, and is not part of the OpenStack release process.\n\nNeeded-By: https://review.opendev.org/969373\nChange-Id: I6629edd31a4fa6c0c330e1214d85f26d275bb393\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/120c032cb9bb5573d909c386374e4ee6bf971098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/120c032cb9bb5573d909c386374e4ee6bf971098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
