)]}'
{"id":"openstack%2Frequirements~974408","triplet_id":"openstack%2Frequirements~stable%2F2025.2~I657b175766f10fe5bbf0fef69367fc906948e031","project":"openstack/requirements","branch":"stable/2025.2","topic":"new-release","hashtags":[],"change_id":"I657b175766f10fe5bbf0fef69367fc906948e031","subject":"update constraint for python-ironicclient to new release 5.13.1","status":"MERGED","created":"2026-01-23 10:45:45.000000000","updated":"2026-01-23 16:51:36.000000000","submitted":"2026-01-23 16:51:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974408-new-release","meta_rev_id":"ecd691222d012bbf8b0c8073852a24702ec719df","_number":974408,"virtual_id_number":974408,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 16:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-23 13:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-23 13:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 12:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-23 13:48:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2ac38f8060db53d6837ebb110a7c930eeeabb1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-23 10:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfd91b4c7f171234216399e26718056cdcf592fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 12:28:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c742ae12dd342098cc2ca61f7e3101e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e8f0ebbd5f481e8eac273f35994117 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e572cc1155a84460835eee4ce8111431 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e9806c1e90c471db08a37dc8e3fafbd : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a317e9a25cf246bd8480f08baf5cccdd : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b29964f5dcf442518185e2a142aa1892 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fce63dcec2a44630b7bca33bc23cbde7 : SUCCESS in 3m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/abc2844e7ffc462ea7e177605ff61bc9 : SUCCESS in 11m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/73cde92c8bea49c7b5bda9b95ef8a40c : SUCCESS in 15m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c731dffee95643e18af8227dd6beb2aa : SUCCESS in 20m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cbde4748227e4f269c4e8b3db808af2e : SUCCESS in 25m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8cb783b480624f4da091eb389893ae2b : SUCCESS in 4m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/740b18ac2db54f138aa4404db42f86d9 : SUCCESS in 4m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3ff2fc965ab6453aa10fff28d031c17c : SUCCESS in 4m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b9e6d1c16e24414ca6f164f0251b7579 : SUCCESS in 9m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f3d7f3bf5b684df1bd0c7df989421709 : SUCCESS in 4m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a77564f5cb724d53830422eaba82edd3 : SUCCESS in 4m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2f91f50c9c8e4049a4ec690470905bc7 : SUCCESS in 7m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/24b6c0cf67c044f7845e8ef6525d0970 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1029ecdcb6f749e48ae9d674a0e4cae3 : SUCCESS in 3m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/00800e8bead14541a7ed3c45aee07313 : SUCCESS in 9m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a50e09921735474691ca228cb9d33664 : SUCCESS in 12m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7599fd360edb4ccba489ecedc96c1da6 : SUCCESS in 3m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d55f91f470574572bdac9ac10a3752ff : SUCCESS in 8m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9bfa926ebba14dd59b10c7c0484814de : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3e7a2ccb5dea4db98d141bcd8cd9a2bd : SUCCESS in 3m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a5e6e8a37e9d455fa915c8af7fa07d6f : SUCCESS in 8m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4a6d898409f6435f8ea4e910f76b726b : SUCCESS in 18m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8e909735a573402a919e58c7d179be66 : SUCCESS in 6m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/46519f13744f4186b390d82875f1ef25 : SUCCESS in 8m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e4d5a345583a4a969ede41d5c5090a2b : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fb945f71f03b44d094ee38b0b91f4f16 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1127ddc9bbcb45c5bdadd6893bd5d587 : SUCCESS in 29m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9b5cad91072a4d53a95b8e51f1f63a90 : SUCCESS in 3m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e39b5646f7a444e8bbf64fe860b47bf8 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abc7aa4a7b77478aa94ee0df890ed785 : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/431b2dab29d04c71b69733dd7dd5cdee : SUCCESS in 6m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d8c1bd6a79dc420b9ad7f02261700a05 : SUCCESS in 3m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27cc2098880b4b828a5e1e0438936e4e : SUCCESS in 1h 07m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1128f104c2794bbd8da243f39e9fa140 : SUCCESS in 1h 39m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eba05cc042fd0446c10346030dfddf5b9fc93153","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-23 13:48:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"988b32c18ab6cb41d76cfca4c8cbc0e3e61332fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 13:49:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"378cf0731fe5a9a38e18bf713ee85e685dfeb75f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 16:51:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98a1b6e9f754723be00509b764d8dd3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543a9ddfd26c4bdc985599d3c4f18d90 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cce337774e8343ff8f82cf297bfe84ab : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cc5c0683c74483786ed0c577b3cffa8 : SUCCESS in 5m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/632efff8a67240309f5b2a8bb39c50f2 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c13c9d7be8914d7baf75e239f6373fc3 : SUCCESS in 3m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/377d9754d61e4f5a943aba3a2a7c7ae1 : SUCCESS in 4m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f5373227e50c41a2a73749eb2e0045ba : SUCCESS in 18m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9e69013e780b45c3a01d86c7de26918a : SUCCESS in 21m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3abe17635a414102aab6daf948b23e2c : SUCCESS in 10m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bf6816772da042ccbce7a9cd412fde37 : SUCCESS in 3m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/64f10b5566be45158a2b3eb975f85e78 : SUCCESS in 4m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/66324c7f9fe74b8987e8450aa9a24fa2 : SUCCESS in 7m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4d1da2cdaaa7496a92e62741a076effc : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee1451928fd744eeb584294c17a7f6a4 : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8a2c45b562a940259170ce58b9a014c5 : SUCCESS in 9m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4cf7a791b88d484a930f8841ad9e3fe3 : SUCCESS in 8m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/adee1c4a75a141b9a141801866b6c073 : SUCCESS in 2m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9033afdbeef84056a969d15d1da1940b : SUCCESS in 11m 30s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c0d00cefc5144895aed3156cfc704b69 : SUCCESS in 22m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92ff0a1496f84d969538601fe6a42c8e : SUCCESS in 9m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f03f61173b6646fdab7083eb6669befd : SUCCESS in 15m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/56ba9b767d0047c884ee16dec10ab39e : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73c95e5986514b7187cd1d4590861188 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4ec1816ffb1481c85cd79094c666d26 : SUCCESS in 30m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ee3526c8e09843e299525b496d00fcf8 : SUCCESS in 6m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/94e20df28b0a4cccbb51f1a238aa0c28 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d593858a8e943c88c43428de1ea29c0 : SUCCESS in 4m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b460416d83524a0b973c6cf216600cd2 : SUCCESS in 19m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/25512eaef85e4f05a4528b08a2bd453b : SUCCESS in 6m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6afdb3581c154d29a4889e2f1e8f8030 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ccf48d6d9d747b493537f4bdecffd74 : SUCCESS in 1h 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ecd691222d012bbf8b0c8073852a24702ec719df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 16:51:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d40384730d72e0be170e3bcad584e1f10eceabb","revisions":{"3d40384730d72e0be170e3bcad584e1f10eceabb":{"kind":"REWORK","_number":1,"created":"2026-01-23 10:45:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/974408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/974408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/974408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/974408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/974408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/974408/1"}}},"commit":{"parents":[{"commit":"5c0507195b43195d9b4762a179a4bbeb1172fd31","subject":"update constraint for keystonemiddleware to new release 10.12.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c0507195b43195d9b4762a179a4bbeb1172fd31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-23 10:35:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-23 10:35:44.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.13.1","message":"update constraint for python-ironicclient to new release 5.13.1\n\nmeta: version: 5.13.1\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: stable/2025.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I418d2dedb1a378d06449f9c33d81a0820fd4ec62\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I657b175766f10fe5bbf0fef69367fc906948e031\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d40384730d72e0be170e3bcad584e1f10eceabb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d40384730d72e0be170e3bcad584e1f10eceabb"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
