)]}'
{"id":"openstack%2Frequirements~974728","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I36904336fb1b97ee026b36bffded3d3a9f8eb553","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I36904336fb1b97ee026b36bffded3d3a9f8eb553","subject":"update constraint for python-ironicclient to new release 5.8.1","status":"MERGED","created":"2026-01-26 14:00:21.000000000","updated":"2026-01-27 16:52:02.000000000","submitted":"2026-01-27 16:52:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974728-new-release","meta_rev_id":"0842aacf8918ea500f89fa5a632314a4961afac2","_number":974728,"virtual_id_number":974728,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 16:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-27 14:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-27 14:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 15:56:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 14:56:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"de9cdf4872b3c86f3377a529fe0161bd2412f4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-26 14:00:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5baca21deaa6be01b0399c88cbfd62478cc546c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802a7139bdba4b1aac6f6f9812acdda4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48da80532e4646388fdc9c4df49aa903 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c801a6cd764cc48a1150d3c45699c5 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b46a2afe53b340e0974667c0b4be6ffd : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4132afb0c8234cc381e873f58d0b7274 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ccd765ba9de6487396cda6a15727dead : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d52f51a73a64a569cf65bf7799e6710 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b452ef47f5b4046b60039322e208827 : SUCCESS in 8m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b534c911ea9495989219920daabba6d : SUCCESS in 16m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cdceaec9b1894087a87e2e0414029c94 : SUCCESS in 15m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a9c9f9468d714caf998a65c8d18a80a6 : SUCCESS in 17m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e6fc30853dfa4a7ba058752d4b9344cb : SUCCESS in 21m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0d2b2d8f1ea84cbcab0ac43bae7ba349 : SUCCESS in 5m 09s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/bc931738d8b043ffbaf62fc360be2339 : SUCCESS in 5m 57s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7d1f0c19716e47cca16dbafb991892d1 : SUCCESS in 3m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a019b368200b4ca8a723e435fbd4771f : SUCCESS in 9m 49s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/91ccdacf77a74775976c9854c55cf8e4 : SUCCESS in 5m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/622189310d614e3784279d027c948e9f : SUCCESS in 6m 15s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/873ac83686434b75b1396f49b42c6ea2 : SUCCESS in 7m 10s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b3ecb68f24c54bdfb41d7512f30320d0 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/385934d621a840739b542f5441ff1ccd : SUCCESS in 7m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f880864b5a894364ab5bba33f822b394 : SUCCESS in 9m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cf39197d9d85461bb58db51206f36c8a : SUCCESS in 10m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6ba76197969f4802a65c7f64ddff740b : SUCCESS in 2m 24s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/e321b2ba5e124299b6f5850eba0fe0d0 : SUCCESS in 8m 31s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/8b1ae4777255401794c8cb1854fd60b7 : SUCCESS in 3m 20s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/429e5dc107c94486add60c397b8a8f7e : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f94af781c0404cba92de57b6cab03a4b : SUCCESS in 11m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a1852fd3970e4aa0a9513da4002ccb03 : SUCCESS in 21m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/327458d3c671462d815aba445ac33af8 : SUCCESS in 6m 38s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/efd0f806ad214d2db7d9a36269e79bf2 : SUCCESS in 14m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2bd09ccebedd427d95faa8fdc697c18b : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2dd68cd935442df98946fe8cdf0db98 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9afbb631bde416ea59e6687520a7bbf : SUCCESS in 24m 25s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4a4b3816e4924ff6ad7ce63010084e88 : SUCCESS in 5m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/66b5474c82d0467ea3ae23dadd620525 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c821a61a55e4e1aaa3a4571aae20691 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7009daee00db42d7adcf6d71a19312c2 : SUCCESS in 17m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e91b43fecaf4f6bb9ebe33a3e06887c : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ef24de5b5f64a9dbd45f4a2ff9e97ef : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1cedbebe08fc56f902d10845d75b89a48616e07a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 14:56:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d784e6a5dfb52f8e67fd822f2b876ebfa8d4ecd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 14:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bd091d3d567f9d4345e731400018a8488036253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684ce1d0fefa4a34b6d4e0c41bbd7e54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e2cb2d87d9445189034ccc63607f6e : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5469fda8eec64e5b94fe04f2e4d36802 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6649f6034514aa3a8387c0daf930990 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57cc0400af174724820eb3e12b62c738 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e2e9b5f66f743e09141347991a571e7 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63ad2aab70214a70b3fc63f6691384e4 : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7201b7ad5bca4539a0d2d1b1a3ad5164 : SUCCESS in 21m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8746850eb3a64d0281c330ea5f987a21 : SUCCESS in 5m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/633eaed4851f4888a3e444bf9c0ed361 : SUCCESS in 19m 46s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c224b1a729de46369e68e2ec414144bf : SUCCESS in 11m 05s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b1cfeb2d191d4247a48363f4aec3f33b : SUCCESS in 4m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/95b212f4fddd4d1b81d23d6f7a789274 : SUCCESS in 6m 16s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b6a51309367f4b8ba7c83f77cd16cf59 : SUCCESS in 9m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/eb199baf0f674e89892595fc09503c3b : SUCCESS in 14m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff443441ea7f43bfb1ad5f29dba50c13 : SUCCESS in 9m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/11fbbfcddc3b431a98f1494e7015c97c : SUCCESS in 9m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fa75fe2f1ebd40dcb53156800a968d1b : SUCCESS in 12m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/73f4edc9ad29456c8ed4c06091ceaf8f : SUCCESS in 2m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/271a5ae48d3c43419e5c48de0d47fcb3 : SUCCESS in 9m 05s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1d990e86fed148fdba9f2c36e51badaf : SUCCESS in 19m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92b41f85237a41c28b126e2ef12dd052 : SUCCESS in 7m 11s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/392fae66cdc24be4a86d7b6be272d2fc : SUCCESS in 12m 05s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f9d4c3eabd47473b94fc312a42529c67 : SUCCESS in 3m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6d4adc41d3c4ea09d8edd6691d46344 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca5b2f62101f482bbad2c9e93f9161ee : SUCCESS in 23m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3772a78b091146629c2ee6f7a98033fb : SUCCESS in 5m 03s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7363e3996e144421802b679ab4f41447 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d85c6f6943984656af0861e0f1c99e28 : SUCCESS in 7m 05s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2ce520003ddf49329f2546a905c6fecb : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c19745ae18414d51a1f6f9cfc5d1f025 : SUCCESS in 1h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14b631ca6d394eefb9a28b6a5e792390 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0842aacf8918ea500f89fa5a632314a4961afac2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48fbc6a59c2156a6874e96362eb4e25034aa2737","revisions":{"48fbc6a59c2156a6874e96362eb4e25034aa2737":{"kind":"REWORK","_number":1,"created":"2026-01-26 14:00:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/974728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/974728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/974728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/974728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/974728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/974728/1"}}},"commit":{"parents":[{"commit":"2c31269cffe9d6e5414822882d6158934df536ee","subject":"update constraint for ceilometer to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 13:50:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 13:50:17.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.8.1","message":"update constraint for python-ironicclient to new release 5.8.1\n\nmeta: version: 5.8.1\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I4ff1f317ffe7b229e507ccd0d42ede8e63202f68\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Pavlo Kostianov \u003cpkostian@redhat.com\u003e\nmeta: release:Code-Review+1: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I36904336fb1b97ee026b36bffded3d3a9f8eb553\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48fbc6a59c2156a6874e96362eb4e25034aa2737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48fbc6a59c2156a6874e96362eb4e25034aa2737"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
