)]}'
{"id":"openstack%2Frequirements~974732","triplet_id":"openstack%2Frequirements~master~I0ca60e147c658ec997276973ef33c3c2b2d66a77","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0ca60e147c658ec997276973ef33c3c2b2d66a77","subject":"update constraint for python-ironicclient to new release 5.15.0","status":"MERGED","created":"2026-01-26 14:18:37.000000000","updated":"2026-01-27 16:20:12.000000000","submitted":"2026-01-27 16:19:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974732-new-release","meta_rev_id":"0bcb8e1398d46e688fe1b4e380d36176c9128fed","_number":974732,"virtual_id_number":974732,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 16:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-27 14:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-27 14:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 15:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 14:55:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"f02d7ca28d313782be0df5de38a0e99607873cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-26 14:18:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5f3d81239eba28a85591323feede221980041a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2cf33c48d34d93b357fa448591f1c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b579eaa5ebf4d32bdd779b274ca2264 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7adf2f652c8a4cb1a190924da52c5752 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cb12c85b5a7420586f1b8cb74df1cd6 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bf643e15f9a4f13aaa71ec0e7453221 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b034669a158483c95541cc5856dd905 : SUCCESS in 2m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e38c431676e42f18b30ce4e32d71b83 : SUCCESS in 3m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3dc28a5dda794d4a924f68b673fe7ab0 : SUCCESS in 4m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/082bfa202cd94098a69e5972232e3467 : SUCCESS in 3m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/edec0214433e47b4a5b8d099a66a4d90 : SUCCESS in 9m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9ccf220e4f1d4e06ad71c2117f955bdf : SUCCESS in 6m 09s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/130b211907574cfcad3119330f909cc8 : SUCCESS in 3m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a9ac9287a2a4477aab54274a44546cd2 : SUCCESS in 3m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3e62d7fbc9f34c3f9047bd37887607f3 : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3fe61ddd38354f8186cb0cd6507bbdee : SUCCESS in 10m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2775f4f33cc54d82baaab29c34f270da : SUCCESS in 6m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c30cf6e87e60407185895c2dbd95d63e : SUCCESS in 3m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/aeffda2612a449e5939587ed991caa48 : SUCCESS in 7m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0ecadd7950144025818ce177b2bd45f6 : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c92bf91af0846ed92c24a2ca3dd02ac : SUCCESS in 7m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6603d5bf4bed4ca78eb5fe8769280fed : SUCCESS in 5m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b73fae8682bd488a91966a47f7c8baf1 : SUCCESS in 8m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8d07bf3cf1214c9aa326b938c71cff86 : SUCCESS in 2m 46s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fc1852fff5c4483b91b6d1ba2b80a3f7 : SUCCESS in 10m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7461b80278b54290926134f044f97979 : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b028adacec364db499e68f617be9609c : SUCCESS in 3m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/01f35ceac82a46069d94f40bcdcec998 : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b06056a72aa74adb8bcf7881601645c8 : SUCCESS in 27m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d8c091ff7faa4c91a106da4d97323cdb : SUCCESS in 7m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8a0f1b1ccf034866887c56838de28b75 : SUCCESS in 9m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fdfb046026a44c24ac62868bc0f014c4 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e511ac192b740ca944923eeb39a7fa4 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcac9844c86c433a894d6a1d36e24249 : SUCCESS in 26m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7358d47eb4c04b17b1027538476ea372 : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ecce25170dd44f298f3dcf9129853656 : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ac1d3367ca346d095b78da8e5d5e3c0 : SUCCESS in 3m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ed8dff3c7b3e4f3fbeabf85f32f1e8c2 : SUCCESS in 17m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b49f5f47809e49d0a8c7b6a2041dc3d9 : SUCCESS in 3m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13177c88f0ee4d579d718d821bb5b70e : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2941ba7ea95d4006af58d845b91e1660 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0727058fe78aff497e621a59dd9523fc043dfd2b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 14:55:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54729d6e853380db0d65b0c7cefc9e5a55a6ef4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 14:56:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1bf7eeaf45480978d41eb43c505f65e7892f0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:19:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2b16ce366e4246bf4261441f9e8a7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7e2709ff834eeba74005d70655c5d7 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d988ca5196e4ff8a2a0332347ed80ab : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e09c9cef1f4d49138d40ed3adbf394f7 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0dd174f74d64324acdeb4d464099b2d : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b9a0471db6e461e9fc50ed7fdadde65 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e9014c0fe14548489f54a93f6a97cfad : SUCCESS in 3m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6032ec56b1034887ab6a4dbb5151d822 : SUCCESS in 6m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d5c05b65cece4afc8d88ab404f5b83ee : SUCCESS in 8m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/65e6e846d2154deca5f55f031a729fcd : SUCCESS in 6m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/430a4b3cb82d4b75b194155eb62b9674 : SUCCESS in 8m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/014c6632bae64af49e5f716ad089dbd1 : SUCCESS in 6m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2004fd607f824b71aed43d3354923101 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/02f7c3b935e34feda9acedf3964f2dce : SUCCESS in 9m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b2fe0e54deef4fe893f5e16d6eac145d : SUCCESS in 19m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/365377725c2f4660bd33afbb713922bb : SUCCESS in 8m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e26f297a0a644d3e8e939610e709d609 : SUCCESS in 8m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7ed35fa152f8410fab0931e5ef37d814 : SUCCESS in 9m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/02ab97d89ccf41c9be1a23efd7fdc5a3 : SUCCESS in 3m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/519c32a5fa00411e8b0e4f2ceb466028 : SUCCESS in 9m 39s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ff30dd02deac4bcb815dfe73a7ab304c : SUCCESS in 38m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4610799ba06843f7896b5e56162054d1 : SUCCESS in 8m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f2cab07a60734f9a8be92b6ef6896c42 : SUCCESS in 16m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6d9da7213e094b1388fb7188f9ab4804 : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01311f1ba0f3455a840ab84794546f5b : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8853e1788dbb4f898beba34c3a98decf : SUCCESS in 27m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3c6821ef19154ca59892a088dc536a07 : SUCCESS in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cf5ca2d973ea4051bd623cf01c74b5b2 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5863ac73bf384bb3853c88e4c0b1e340 : SUCCESS in 5m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/51c051c8dda24285a0f287b92cb69665 : SUCCESS in 21m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/45b17ce22e664bff9a3c33cee5b35a0d : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b77be22e5774e2aa4e6e85ebbd11a4f : SUCCESS in 1h 04m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7aaa6716f904aa68242e4012c8c1f53 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bcdaf72d93ee941fb922783c695e800d0ba96936","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:19:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bcb8e1398d46e688fe1b4e380d36176c9128fed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0cb7797f4b94d9a819a3f39445c2a7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467f897bc7a14d4787672fae7ce85cf8 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88c3a718ddc83e74af9aa87594d8ac185e8b3a81","revisions":{"88c3a718ddc83e74af9aa87594d8ac185e8b3a81":{"kind":"REWORK","_number":1,"created":"2026-01-26 14:18:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/974732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/974732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/974732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/974732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/974732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/974732/1"}}},"commit":{"parents":[{"commit":"b6633b1e375b7fbb288f7c14e84cefdcfd1d276f","subject":"Merge \"update constraint for kuryr-lib to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6633b1e375b7fbb288f7c14e84cefdcfd1d276f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:08:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:08:35.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.15.0","message":"update constraint for python-ironicclient to new release 5.15.0\n\nmeta: version: 5.15.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I160958d554e670b81c40a837a3ed9a32712c844e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0ca60e147c658ec997276973ef33c3c2b2d66a77\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88c3a718ddc83e74af9aa87594d8ac185e8b3a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88c3a718ddc83e74af9aa87594d8ac185e8b3a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
