)]}'
{"id":"openstack%2Frequirements~974733","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I888735f1379b9b8319539006739e0afb94b0f0b3","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I888735f1379b9b8319539006739e0afb94b0f0b3","subject":"update constraint for python-openstackclient to new release 7.1.5","status":"MERGED","created":"2026-01-26 14:21:13.000000000","updated":"2026-01-27 16:44:51.000000000","submitted":"2026-01-27 16:44:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974733-new-release","meta_rev_id":"7a1b413bd02752ca0e2620d2415d78c9ec7cbf5d","_number":974733,"virtual_id_number":974733,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 16:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-27 14:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-27 14:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 15:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 14:55:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"20c6a6209e59c49cdc9f862b11eaf5dc55774ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-26 14:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22a0e87eb6339f73406644d9e037669eb8529fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:19:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4616b8973d148cb9f7c5c3d1ec38942\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49c0701ceb4245c591e52a6311480ca2 : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a2fa3d9b5194f73885d55c5b7ce3d16 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f214d5412e894d84be77fe6a2f9ae9e5 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a0fe75b311946fca96005803e75e134 : SUCCESS in 2m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8f485225e2341879bf023f486fb4b38 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22dda4b822054ccbb6c4213a6bd990a1 : SUCCESS in 2m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0057afba13454257b52ce8be8597b65e : SUCCESS in 11m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/398598eadbbb47c7882515ffc9e44e75 : SUCCESS in 9m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c8ea262163548718c0106af35749911 : SUCCESS in 9m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8086715ece3f4fc9af7cbb865c844e74 : SUCCESS in 7m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/253195b43ba049aeb285750b658c3ab6 : SUCCESS in 18m 19s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5239217c6a754f9e8ccfa175ac1c285f : SUCCESS in 3m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/197c44beb6004a2587dddfc7d87c667b : SUCCESS in 5m 26s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b39a651fb65642e5b87ee8b414c3921a : SUCCESS in 6m 13s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/da5b90a5330a470e83ee0a1c91bc5944 : SUCCESS in 7m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b956094fd7554bbf9469e16fc0aea676 : SUCCESS in 3m 28s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7dc133166fba4f7182f1579539917c42 : SUCCESS in 4m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/50f6562d100347e0a25a6fd6446964e6 : SUCCESS in 5m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a40aa11f91a3473f8a172e09f1083644 : SUCCESS in 5m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60614738e02f4c089e94e318fbc8d4d6 : SUCCESS in 7m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a359974d4303486d9e0e183c1c6aa236 : SUCCESS in 5m 11s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/15111cc8d66740d6b25ca11c2f735b78 : SUCCESS in 7m 03s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4939e700e50f4b37bc9d1cbef47c2033 : SUCCESS in 3m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c9f0d6582957497b9c5640154ae692b5 : SUCCESS in 6m 12s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3dbbc036ca4d45a7a306914bda1bda63 : SUCCESS in 6m 39s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fa8e69b6486b457690aed3e08c48510d : SUCCESS in 3m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/609560dcef9b48f5bbc42665dd316ab0 : SUCCESS in 12m 16s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5c9446a218c7400ea11452e6f755c520 : SUCCESS in 19m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5ed7d2e96ac848d99f8ee1545d147ffe : SUCCESS in 7m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/cc01efcc30fa46788c565ee29ad20937 : SUCCESS in 11m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/45e77058b58844099fbcf395d5dc0e1e : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8421a63d7868486c8b7f94046b5c4ff9 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b57cd8d4ef04a31ab141a370a18edc0 : SUCCESS in 23m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4124080d069f4501898fbc282f9bfd91 : SUCCESS in 2m 43s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/aacf781f23ec4db6b377f31c9dea7ab1 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba9066242b1549b1937c5556325c238d : SUCCESS in 3m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/36e6fe148a2d4dbaa4437c5c5e8d74cb : SUCCESS in 18m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70faba312dc94bb6965981bb56cc3b1c : SUCCESS in 45m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/978249eb7133427f8dd0aadf84ce8e7b : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"67a76d527489ff13dadf3d92e69efcac3e537439","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 14:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b217cbae70951e967d9259eeeff44cbc49572302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 14:56:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df7e24a6b5565dbb50cdab6f2c9eb9ce16470545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:44:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab81c52c0cbe4e36af62536fc03f996e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692507de94a2405a90094138d060c2a9 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b771fe194ce4c1491e19950ed379c08 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/451cfa55dc6d4dbfa5fc284876524cf4 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93b1f6ee1a4741eda5d54a8731f716f4 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5853e863235a47b19a2443b61e8b4d0e : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80d7fa1b389748839f083366150a1f69 : SUCCESS in 11m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df49801ecc0949e6b5164976cc70bfe1 : SUCCESS in 17m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/60b3d6b105474fa3be4d2ff0671c4235 : SUCCESS in 10m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b1eb3eb6c999408a824797af0bdb37bd : SUCCESS in 24m 02s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9b37c86e3c5b416ba79b727a8cde511b : SUCCESS in 12m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9cdb390e92104734b575709d24ab5ece : SUCCESS in 5m 42s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/81dd1f3174e04d0e8efd68d9343bed81 : SUCCESS in 8m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1c7aeb2de5f847d38242c93323f9f088 : SUCCESS in 8m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7cd699a79bae40239e41ee49b58f2a88 : SUCCESS in 13m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af26b553d7ec488383a5c6be3ad9c469 : SUCCESS in 7m 37s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/71b8def5296c45a99decd953d322dfe3 : SUCCESS in 10m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fcb7da2b39a14295b33b6c4a7b120dc3 : SUCCESS in 11m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7ea17ea5ddd94f52b92464ad26ccc1b6 : SUCCESS in 2m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/346ade7107164015852e32445b08e42f : SUCCESS in 9m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6484e5c3d15344378f4c0ca26267bb8c : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3e6602ba37fc4b39b5c398c7cb2d7ad8 : SUCCESS in 7m 32s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f20912e1668d46abb484a7888add4b5a : SUCCESS in 12m 23s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/32148c8faddf4fdd8ed935f391ebb4a5 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2d64c07b6f74b8f9b0303878d131150 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35c3179d9f464470a5c46794998981ab : SUCCESS in 26m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/92e3fca2880e4ef28089ca8aca29cb52 : SUCCESS in 5m 04s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/fa8df7451d5c454b9e917eefff4340dc : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a5a8951a44a40fda75d071e3869142b : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8c9296a2a42d4a54b38c91d9696e5afa : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4859333448264eb8a9f890c3d3d429ee : SUCCESS in 57m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75cd98bffe074f948748b7fb44852bf2 : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7a1b413bd02752ca0e2620d2415d78c9ec7cbf5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:44:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d26e01d73fa6affb5dcb521f8e6981b9e4645bc","revisions":{"3d26e01d73fa6affb5dcb521f8e6981b9e4645bc":{"kind":"REWORK","_number":1,"created":"2026-01-26 14:21:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/974733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/974733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/974733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/974733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/974733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/974733/1"}}},"commit":{"parents":[{"commit":"2c31269cffe9d6e5414822882d6158934df536ee","subject":"update constraint for ceilometer to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:11:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:11:11.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 7.1.5","message":"update constraint for python-openstackclient to new release 7.1.5\n\nmeta: version: 7.1.5\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I44e12a172c24b1d7a0e01ffd3f09aa47dda27674\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I888735f1379b9b8319539006739e0afb94b0f0b3\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d26e01d73fa6affb5dcb521f8e6981b9e4645bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d26e01d73fa6affb5dcb521f8e6981b9e4645bc"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
