)]}'
{"id":"openstack%2Frequirements~974734","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I0b1f7772a9db71ffaac74127b75944daa8662977","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I0b1f7772a9db71ffaac74127b75944daa8662977","subject":"update constraint for python-octaviaclient to new release 3.8.1","status":"MERGED","created":"2026-01-26 14:28:31.000000000","updated":"2026-01-27 16:52:06.000000000","submitted":"2026-01-27 16:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974734-new-release","meta_rev_id":"740ee2c6b0c921d5f05955fd0f5eedaf5a6f0c54","_number":974734,"virtual_id_number":974734,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 16:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-27 14:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-27 14:56:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 16:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 14:56:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ce7f44e537a159b0e2391ca9ebd05dfa7745e5ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-26 14:28:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92eba245b0461a703645c7d7199611197b343e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2893cc5d38c44e084f18ec24da4b079\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5790e529654cf8bb126ae421db8d26 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3affa96d5a7f4b9e843f5597ff290925 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6735844c64784bd5b7478a7b076cb1ad : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08323b72eb584db1a1ddcf72de652929 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e0aae9a3de041b3b912ae9e23ea5ad1 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20d96e44432042159c0cf0e2cae6bc1e : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/311ee2e4571545e699ac7a84dd21f281 : SUCCESS in 6m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b4ff84d188d647668e81b705b4f1e523 : SUCCESS in 24m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4e6adde98df74737a2fbf1a3811c31c2 : SUCCESS in 11m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/95ae540240f74297b8f116debe2a48fc : SUCCESS in 17m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d2202faf0dd94b288860a877771891f2 : SUCCESS in 19m 30s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8455e102c110409a930db7f1d957aaee : SUCCESS in 4m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c01e29673342431eb80567dd2445cff1 : SUCCESS in 5m 28s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/cb4c592acb79412bb8b5b05196dcfb7e : SUCCESS in 3m 45s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e4c8983f5438488c81d5fa76718ddf5b : SUCCESS in 10m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b09b16e906b640fe9d105910d1b8028f : SUCCESS in 3m 21s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5fd6e984320944dd8d11a7a74a3fa04f : SUCCESS in 6m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b436dfc97a6741e29b90b1e1106c3bd7 : SUCCESS in 9m 01s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a964a1136d534fef8e2dd2b229f94a55 : SUCCESS in 13m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b66e9bf7ae64eaeae5b2d461d0e9e59 : SUCCESS in 6m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c443f6a3f4cf4a18abf50a11c977126a : SUCCESS in 8m 01s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/760d030f6e0c43979630cdba3f3bb7bd : SUCCESS in 7m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/aab2b16c1f2d417db5bce906b24ca22d : SUCCESS in 2m 21s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/09078fa4ea0a4c87a8e93f829426cca8 : SUCCESS in 6m 24s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3934470228e64430a15622d0f5053fbb : SUCCESS in 3m 54s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1988cd716e6343cfa6de49df8ac69934 : SUCCESS in 6m 30s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f9ae17be855f421e8ef9408dda109727 : SUCCESS in 10m 13s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/83abc95e99cd4b14908ab0db0fe557a2 : SUCCESS in 14m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/58b508d2096f4530a8e9eb1b7f3005aa : SUCCESS in 7m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1ed963bec0104ba38e8e201b456da1d6 : SUCCESS in 11m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/86e520dc998f4daf9989775ccbf36f5e : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d565dc992a844dd5b506cf872b370d5e : SUCCESS in 2m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b146b944ac44768a2e0841444eff563 : SUCCESS in 22m 51s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c67af36d41974bbd9c92b347678caefe : SUCCESS in 2m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8f506c532ab044b089c52a7f250ceb26 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2fac891abe84dca954fb760cd70c1fe : SUCCESS in 5m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c7a82ef5cc5f4a0fa76819e349e442af : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a597bc1556ca465bb2686ec4d4c19b6e : SUCCESS in 56m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1015524af241429193f4802372c3231b : SUCCESS in 1h 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"304f90835f32d6639bc68e72c84d32fe5d599515","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 14:56:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7257ea67e754e5120a102efabbe67d2a43474376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 14:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3b40ca09b52ffa5eb82e59802d9e47aa75fe7c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c0adf12d14c4942be0854aa5d70bc07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/896f365ec3584802bb3710b6ee6d403c : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fbf01545d8346c989acf84481784057 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71fd8df822a54164b135db567b2afaea : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2c9e1523482477f9dfa46b898f6e7b4 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c92165e4f24c49eab78b47b310e10bad : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d31c20ce8b949aea6ac41e7b77b9dd3 : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3889df14ecc94d11a3055cf9cb10a371 : SUCCESS in 6m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cc7973e5a2ff494fafe0e49a35adda05 : SUCCESS in 9m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7c2bd8a2bfd74fb4bec2f8624d738f91 : SUCCESS in 8m 01s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6a93e74ab2f7440ca0045e017a5dd6ed : SUCCESS in 10m 47s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/32c16f634ae746b1927f340829c64b2a : SUCCESS in 5m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c2cbe93c5716418eb02940758ca5b405 : SUCCESS in 4m 43s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/37c94b79970f4d2c99433201810ae963 : SUCCESS in 7m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2ecc393019d3418a8da25d358b612c64 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/409c302755d74df1ac36f6df06300b3a : SUCCESS in 7m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7ed8dddf771d40c28ed463d546e46f8f : SUCCESS in 11m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b01ac0b34d2c4ccea1eaf8555fb16e45 : SUCCESS in 10m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/84e37aae7e764eb39bb58ef0554757e2 : SUCCESS in 3m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d63734afeaaf423a9d4e6733a62c58f8 : SUCCESS in 9m 03s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2379a44ff43d45968d6289a7b35e6d59 : SUCCESS in 16m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3065f205fae540ea9456c5efddabf44b : SUCCESS in 8m 02s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5349f420f28e4b71821966f3fbd2c853 : SUCCESS in 14m 12s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6e24d2cb5fcc477f8a15e85870aaccb7 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0d882c193424718aef764e164516e0c : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03611eb3224441cf9c3c43da323c162e : SUCCESS in 24m 43s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/46051593fa44489b9b0e28322e945f46 : SUCCESS in 6m 38s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8405b09a43c74718b4764b90a231bafd : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/412b8bf6f73e44f9a5e6d33414303a97 : SUCCESS in 7m 34s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/721435307de647c68492b430be617a51 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b282b2994af34670adbfab7042006f71 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5e78af0e51941cf8d572493cf994d64 : SUCCESS in 1h 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"740ee2c6b0c921d5f05955fd0f5eedaf5a6f0c54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b12904724c6eae0a0cda869b0c9a58be6eddeea","revisions":{"4b12904724c6eae0a0cda869b0c9a58be6eddeea":{"kind":"REWORK","_number":1,"created":"2026-01-26 14:28:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/974734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/974734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/974734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/974734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/974734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/974734/1"}}},"commit":{"parents":[{"commit":"2c31269cffe9d6e5414822882d6158934df536ee","subject":"update constraint for ceilometer to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:18:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:18:28.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 3.8.1","message":"update constraint for python-octaviaclient to new release 3.8.1\n\nmeta: version: 3.8.1\nmeta: team: octavia\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I644a89f24b436f18f7541a48541c3b2ce20c1321\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0b1f7772a9db71ffaac74127b75944daa8662977\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b12904724c6eae0a0cda869b0c9a58be6eddeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b12904724c6eae0a0cda869b0c9a58be6eddeea"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
