)]}'
{"id":"openstack%2Frequirements~974738","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Ic285dbdee06b9e7220c447832cad061535f26e31","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"Ic285dbdee06b9e7220c447832cad061535f26e31","subject":"update constraint for os-ken to new release 2.10.1","status":"MERGED","created":"2026-01-26 15:09:55.000000000","updated":"2026-01-27 16:52:10.000000000","submitted":"2026-01-27 16:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974738-new-release","meta_rev_id":"415610de4f5997a82122e9b1dd95636417550466","_number":974738,"virtual_id_number":974738,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 16:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-27 14:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-27 14:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 16:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 14:56:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2f8e30b689e7d48add2932798ac1b64d508d3826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-26 15:09:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c15aadd624a5556ca79f1e0c8e84c98984a2dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:49:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b265cbf9f7604fbb935d3fd92947d53d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7558156155304adf8cfd2440219e3894 : SUCCESS in 2m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1cf9bb1e49b448c9ecec41c933c59da : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/322bae38037940c394ba88644b57e0db : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/110216f7f9e24e6c835bb765b3cec447 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ce48320b13449dcbde825cb1efa5048 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/330fbbe52fee4c799143b5d8facb4237 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b20f9aabd89c4446889f25da8f827ff8 : SUCCESS in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/71fb423d9cfb4e98b653ec8306a5b8a8 : SUCCESS in 13m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7bc03b7d5b484fd495ecb4be041028ee : SUCCESS in 13m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/90f6dd47aae54461bddad7c34eb4cf80 : SUCCESS in 17m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/667506abe20c4e99ae9c58687feeccdd : SUCCESS in 28m 37s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2dd2c0135784445db65e7ed7a2503f7b : SUCCESS in 5m 10s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/f87be9b4036247b6b63329638df61566 : SUCCESS in 4m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f073f0c491ee49349f0d603897464f16 : SUCCESS in 4m 11s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4efcb03df2dd40d7bb449abfa39ff0e3 : SUCCESS in 11m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/dde54a782f6b447a96b88427809f9ca6 : SUCCESS in 4m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/12673b7d1fc54e3fb04c5415022bb807 : SUCCESS in 5m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c920993c1341484783e5e6538be5a604 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b896983b303c4513bb342a91f2a35659 : SUCCESS in 5m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fd36077f2fe451c8f67062cbf174913 : SUCCESS in 7m 08s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/775aecc03d0545329473606f7327a393 : SUCCESS in 4m 39s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d1cf9675c40a46709e3dca34b2360385 : SUCCESS in 11m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/68034bff11b34f51a831885bf12bb0d6 : SUCCESS in 2m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/699f6a6259d04c3f90ae053172ea5d6c : SUCCESS in 9m 13s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ba27d907024d4f3e97b134fd6426297a : SUCCESS in 4m 56s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/66ef7faecd744b2b9603bfa35fe73ad7 : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/31172fca97a143659bc22a5a85477146 : SUCCESS in 9m 34s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/92beb86d4df7414b98a2a75b61a58e0a : SUCCESS in 19m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a7e126c4404439da679b52bd1a20519 : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/304b70479ad94f05877dedef37090614 : SUCCESS in 11m 09s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f15667dca1644a2e8fe5ed6bbce2ac93 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd77784febea4faf98a1df2af87486b6 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/548c341896aa4c0fb02404d4b391602b : SUCCESS in 25m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6d0c8adbcb5a446e962e5ab4787071c9 : SUCCESS in 4m 08s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0d84c65646bf4652a81b0661ffbff0df : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bb789343c264daab5f500bef4a2d349 : SUCCESS in 5m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/eb8c9c0e422f4d63a982cd3c02cb6ca5 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb456f18990842afb5f362aff58f978f : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d546057e7b74dfcae07b3f3ec389333 : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"79930b865ae0d0552bb2f9ab73f84ca1fbf94703","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 14:56:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6142b792ed56e7e3c297c1635ea11ae515a83ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 14:58:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c6be73535a10824c26d49bac4307f5a2c632f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739ad059c85c40068a46c426df80b657\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95308efaeab4425c9e4dbda7b050be85 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3880c9a093c404fa1ac94f808c620ce : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc5ff2fc52aa423a93f744c93d301b9d : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a883ad8e35da4815b8ba6eed452da16c : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73163d717dda4164a42337be7afc3e8c : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/341efdf25bf141738a1a1962f3144a15 : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2bd97c8730bd472a9333c064783ef417 : SUCCESS in 13m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/533c1a043fcb4420bd970668f8dc740d : SUCCESS in 4m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/575eb9c73e1f4a9ea4cf7a45159605bf : SUCCESS in 18m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b6b5a33dbc944d308c1c2bdf62696277 : SUCCESS in 9m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2f2a49146cd54e6884a225c7f483951d : SUCCESS in 5m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5adf940fba7b41afa808675d66cb7076 : SUCCESS in 6m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d35ace8ec9424fc38229d92bfdd59e41 : SUCCESS in 5m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/60c5f25547384c0f8ddb79967e9c2efd : SUCCESS in 15m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/892ea4ceec82419c972dce064ff13852 : SUCCESS in 5m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cf2919565c0948c199d38ef5ae141544 : SUCCESS in 9m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8e2fa85b95214d72aa91a0c0c0d892fc : SUCCESS in 8m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d08c93609351410b852d4c37bdc64c0e : SUCCESS in 2m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a481ab5b3b694c819f47fea21a5c95d3 : SUCCESS in 9m 16s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1cf6401a8e84433dafd2bfad0f9c97a4 : SUCCESS in 17m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ef3faf9829364630ad5ff2345321e0db : SUCCESS in 10m 08s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9aca5bd1d1474e6c8458a087cff1727c : SUCCESS in 8m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5c2f527e8a6445009bfa906e82c3248a : SUCCESS in 8m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32c4492315724cd3a13a1581460ee319 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d0f724ffc784b81b8d0327f86d9fc4e : SUCCESS in 22m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b65eebd965d349f7b5f212f1ea9a5a51 : SUCCESS in 6m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3bd8c21cace24f6c89233e6b36369036 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82c841300f5d48cbb957de0bc7a49b1c : SUCCESS in 3m 54s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ebef8f25a11d459c924407710da9c53f : SUCCESS in 8m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/054d898b1b304f068fe85a261b1e8156 : SUCCESS in 47m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d66c704c246a40ccabcd5876f531a442 : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":1},{"id":"415610de4f5997a82122e9b1dd95636417550466","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:52:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"523352ed0d4a34a249ace1593c428342a8072052","revisions":{"523352ed0d4a34a249ace1593c428342a8072052":{"kind":"REWORK","_number":1,"created":"2026-01-26 15:09:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/974738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/974738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/974738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/974738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/974738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/974738/1"}}},"commit":{"parents":[{"commit":"2c31269cffe9d6e5414822882d6158934df536ee","subject":"update constraint for ceilometer to new release 23.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:59:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-26 14:59:54.000000000","tz":0},"subject":"update constraint for os-ken to new release 2.10.1","message":"update constraint for os-ken to new release 2.10.1\n\nmeta: version: 2.10.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I442286ef61cde3c2e22f8521b4753f6b1d65c59d\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic285dbdee06b9e7220c447832cad061535f26e31\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/523352ed0d4a34a249ace1593c428342a8072052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/523352ed0d4a34a249ace1593c428342a8072052"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
