)]}'
{"id":"openstack%2Frequirements~975006","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I0af055a6f3c9f51f912cd038c16ac3adaa1b163c","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I0af055a6f3c9f51f912cd038c16ac3adaa1b163c","subject":"update constraint for python-magnumclient to new release 4.7.1","status":"MERGED","created":"2026-01-28 13:27:51.000000000","updated":"2026-01-29 12:47:18.000000000","submitted":"2026-01-29 12:47:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975006-new-release","meta_rev_id":"c4bacb93c7a1662f52e672338cee387f8eee1e39","_number":975006,"virtual_id_number":975006,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 12:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-01-29 10:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-01-29 10:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 15:13:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 10:28:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4caa6250934264409c7034812ccb09a770819d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-28 13:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3cdbdbe597f5f5a6d990a0c4cf79ce31860d271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 15:13:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7942449f2b4342d7abad33589d18eb30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a567b747b3e94b41a777b9fac05c96ed : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b9675f129694a14a29d4775feca1052 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f77d6bfeebdc4e81924754057ac722a8 : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c24b59616dc49cd848085b5c5fa6e76 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2745578ad7384c5d9548e06d39e00dd7 : SUCCESS in 2m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a176d3f9fd44c8bb248a7ccb17d0930 : SUCCESS in 2m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06ae668a7ded4a3e8cc20c7cae203440 : SUCCESS in 4m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4fb4fde13136421da47fc98ec429d934 : SUCCESS in 14m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/836645c795f7480ea9e280f1f4a3b9d0 : SUCCESS in 8m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32c4c4ff6b3141bd993346772b497f3f : SUCCESS in 12m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4a37198be05a47e09bd005f7d5c82dd6 : SUCCESS in 10m 14s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ef7e7fc7b9574099a482def62e8b8f3f : SUCCESS in 4m 56s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/27ec5a318be74537a8b3230769f71c6c : SUCCESS in 6m 33s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ead5ef0821cb40f4a046ae19f67b9012 : SUCCESS in 3m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bcf567d944c848e0a4c152e0dfee6a14 : SUCCESS in 10m 32s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/850c9ab4e46b46aeb45af9bf4b290f75 : SUCCESS in 5m 20s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a60cb78ba28e49429ab82a6579f19c10 : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/66608294ac6745f1bde69593ed704d7b : SUCCESS in 5m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d47fc7062dfa480eaeb6ffbe7029db5d : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0a38a080ff94a8aaad55fefb3cbbd75 : SUCCESS in 4m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1f67c9b4250d41ddb65be5ae82eeca78 : SUCCESS in 6m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7e9e6f1a5af349e280dd6460d5046b9b : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4825013587fb428fb1eeaa1c1bfc4e2b : SUCCESS in 2m 18s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/af516df06cbe46dbba67447731c72e64 : SUCCESS in 8m 05s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bb89a2f484fd413f967f5852356077de : SUCCESS in 4m 15s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2bb4066d962f45b8876ce4244fe17799 : SUCCESS in 5m 02s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0b586be6940a49c18bef81a4beac1063 : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/7bf30f12dfd1497b88d0b85c9f6dbb44 : SUCCESS in 19m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c71085c76501455e8faf6235000d7689 : SUCCESS in 6m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/03844210c30b47978e5e86839a8aa11f : SUCCESS in 7m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2884628a8a4842e4b71cc05463e195fd : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b557157f9cea4ee0a9fce1358fed2ee5 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d2965208f0f46459d6c964e8f02e2dd : SUCCESS in 20m 49s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1e285cd8258d45de8ad6a9ac1067d5cc : SUCCESS in 2m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/89a353862d134ebab375e10cf5cc2848 : SUCCESS in 2m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fcae4418b944c9c81b47db9048661ce : SUCCESS in 4m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4b9b69cb2e5e4d80baee476c2a789df1 : SUCCESS in 16m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68e2fb83577e42f1886c7db036208ece : SUCCESS in 32m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf37c7f1b6d1440190a2f9a296ff65c9 : SUCCESS in 1h 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d51fdce7753790d82eaafa67e1a2a0496226f5ad","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-29 10:28:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ae1914335c67829830489cce53f12f655d39a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 10:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d875cbedb607fcb3b2d089a7ef07e108058bb3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:47:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c503adc393ae46d9becf2d7f61fe7494\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5774f80edb4c1db5c0e3afe539879b : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/870930a695da40af98c16f4c9d00a43b : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0005a42cc96a4995b3b0432bde6fe785 : SUCCESS in 2m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be5057436ae140dba1bcc2117d439501 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0875e23b3e7c4601b7efd6475aa2e45b : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5997df0f0e384988990124711ef22aef : SUCCESS in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ca4cb27aa5e8409ba4f765877a56fc6d : SUCCESS in 20m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c2d45f0a1b6e4cd8807a593787443f49 : SUCCESS in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ca44bca583e34352888a163197c953b3 : SUCCESS in 8m 38s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b53592dc234c4c4d898315dc20fe16f7 : SUCCESS in 13m 25s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e54ef3d1385b4ab9b10f3df6edb010e5 : SUCCESS in 5m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/72d732e427014858854346a43f3a476e : SUCCESS in 2m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5a1279c5f7a445358da4ac8ed38de021 : SUCCESS in 9m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c21d72ea65dc4eff8ead047954a3f996 : SUCCESS in 5m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0419cc46d8d54cf3879475048f70f432 : SUCCESS in 7m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ed6305114e5244e99526f7b21180942e : SUCCESS in 7m 39s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f46af740d3cb45f3aaf13079712d5c9c : SUCCESS in 11m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/472ccfe8a6214c529c38e1495683697d : SUCCESS in 4m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5e02799087cc4818a57a85263d529e3b : SUCCESS in 8m 36s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/64f0d47429474a5892a98d7c3e7cc9a5 : SUCCESS in 22m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/100498a116b54653b61301445631b1b2 : SUCCESS in 4m 15s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d911b4f3165248fb8dc42967faec6952 : SUCCESS in 11m 32s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7efd3849625d4782946dabf2ef9eea6f : SUCCESS in 2m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8478f78e4c048828fbf9fac038f43c9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15224e24f1f542c78fb28142ac1122ca : SUCCESS in 24m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2521916a23454b3eb4777bef91294de2 : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/90d7d6ca5c8840b5b3e092d0bba561c1 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e438c16ecbf40d59c0130e820bedaab : SUCCESS in 6m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c616f6651ce348e5bba41f0b1697ec76 : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d7997562c1b4b398ac36120999a75ea : SUCCESS in 29m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e5f6569e95840c382f6fcd88b707278 : SUCCESS in 1h 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c4bacb93c7a1662f52e672338cee387f8eee1e39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:47:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9314c00d70852299c63e41241e739172e8f79566","revisions":{"9314c00d70852299c63e41241e739172e8f79566":{"kind":"REWORK","_number":1,"created":"2026-01-28 13:27:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/975006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/975006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/975006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/975006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/975006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/975006/1"}}},"commit":{"parents":[{"commit":"1e473112b902584c1aeba589b6898f932dd4e46a","subject":"Merge \"update constraint for os-ken to new release 2.10.1\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e473112b902584c1aeba589b6898f932dd4e46a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-28 13:17:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-28 13:17:50.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.7.1","message":"update constraint for python-magnumclient to new release 4.7.1\n\nmeta: version: 4.7.1\nmeta: team: magnum\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia22812205b26eb892feba0725f274e5bd8f636ea\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nChange-Id: I0af055a6f3c9f51f912cd038c16ac3adaa1b163c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9314c00d70852299c63e41241e739172e8f79566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9314c00d70852299c63e41241e739172e8f79566"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
