)]}'
{"id":"openstack%2Frequirements~975403","triplet_id":"openstack%2Frequirements~master~Ia2de44376782188896213b669f05dadaa0f9c57f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia2de44376782188896213b669f05dadaa0f9c57f","subject":"update constraint for os_vif to new release 4.3.0","status":"MERGED","created":"2026-02-02 14:41:08.000000000","updated":"2026-02-03 14:24:11.000000000","submitted":"2026-02-03 14:23:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975403-new-release","meta_rev_id":"c6e3d01461bf4a27007f9ba68de3dd895b257fef","_number":975403,"virtual_id_number":975403,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-03 14:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-02-03 11:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-02-03 13:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-02-03 13:06:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 15:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 11:37:21.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-02-03 13:06:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"980044cda3244ec9bfbef05ab032e9497c26bb85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-02 14:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e57cf8388d26f93ff42a12779104b4f9d6866201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 15:50:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7280d8dda1ac4047a3c4fc80334948e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc8c3d1954f4f4f8fdb3be0b655f7a8 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae45363829e34f6d89f3e50d9cdc56e2 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7328ead0065a49b2b70e460a0262e3bd : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a74a3db0ea6045ab902f57c472f5ba76 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50d7dc6f3ee1496ab22b292bf602c13a : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8795e6a5b754a15abf9f29c28d92fee : SUCCESS in 4m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/85e728bfbb8f4861a6ecbbce6f8e9c23 : SUCCESS in 2m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3421e0df41af4f2cbebe84683f2081d5 : SUCCESS in 4m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3af380c5ebc8401cb8c2575a555e5b43 : SUCCESS in 5m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2f77ef10cb2941d6a702b8d789468053 : SUCCESS in 6m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/32772c92a0f94d6b952fc421c8485f46 : FAILURE in 5m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f61ee7f3d41e4bacb852b2405288a2a4 : SUCCESS in 3m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fdf337e6023946bdaf9b71e7e23d428e : SUCCESS in 3m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/13243b2ccbec467a9307d30d516c4220 : SUCCESS in 11m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/554831a6a2354c3580992eca051dad10 : SUCCESS in 2m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0ef65d5580b4463491e2c1e55cb3e4b9 : SUCCESS in 3m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2e717a32503a462c961a50ee7c226e80 : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/afd6a29f2294445a9edbf8e272b5bd78 : SUCCESS in 15m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6a6b28d9058412eaa0a6406ddb3c52a : SUCCESS in 6m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f803f66a155845fe87c2883b6ea3ef64 : SUCCESS in 8m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fad09984bb5d40fabe6273d4a343c5c9 : SUCCESS in 12m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7e1a562f649044a1b667950e5f3590ca : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/46b08825bc8741ec8c8337340f545022 : SUCCESS in 8m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/29527065dc614f519f3eba31e46fcdda : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1981fac79cf04d2c9dbdcef988822eeb : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c53aec26727c4a19b754727245ba35dd : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5734775bf73e487cbc9a8cd9107c5d62 : SUCCESS in 23m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3e394c87eb9c41cfbe9783770a1fc08f : SUCCESS in 9m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4a882661c80846ab9ba7a6617b4c5a17 : SUCCESS in 10m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea474ba35ebf4dd588fbd1922f6a9ada : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cb40a97ab3645d6aa0bc735f5fe8c96 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d8d09c515964dd9a920327ed7c6ece7 : SUCCESS in 22m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b9ed7578d9f94e31ae96e3545993c940 : SUCCESS in 6m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1fe7473198084cf0b7cd3701229d1409 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbd1f6bac9cb49da8779916aaec13966 : SUCCESS in 6m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2fd8aa38340a4c41aabd07905de223d4 : SUCCESS in 18m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9430824b23f044cf85fd4e5681b81a3f : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5eca436dfd8846d4b8781d7169062733 : SUCCESS in 32m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fb62847f9f7441c8771d154c429a69e : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5834d68d3f51e305e2a03d004b13cea9db7907d5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-03 11:37:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd0c790c935f82e8bb6c5d27ef620ea5ab8011bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 12:37:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf106e6e6c148dcaac84eab2941c61c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b398cdbd23041f98bb9c701cca66dd1 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c9310aeee2440089e4e9b4d83b77a7 : SUCCESS in 1m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ccb2f5d69f74689bc7f9a67b832f353 : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82ec7ef1a3a84fb09511e59ea13ec5dd : SUCCESS in 1m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1450d18eb3374366969757a26b85b1b8 : SUCCESS in 2m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bc63e08259944b6b59fcc58aa3def4b : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97109f50a3e04094b41283fe287f5753 : SUCCESS in 2m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/736b1f01d4894514bb9cab1b93726484 : SUCCESS in 6m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7e28f16d0ef24ccb8505a3b4660c87e4 : SUCCESS in 6m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f7a781c96f8d4eb28880d52c77c25d8b : SUCCESS in 6m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c7f2749e2e2241f9b49779eb07d2ffad : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/84432ad2ebb94b628cef2d343f93f4e8 : SUCCESS in 6m 32s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/038693a1c4d2430980aebc2605a030b2 : SUCCESS in 2m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a55da081e2ae48f58bec6dcffcc40857 : SUCCESS in 7m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b31a05e5f855496399623d963c4b46e1 : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2727425cb0884b98926926bcfd44839b : SUCCESS in 5m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/91beb7f2f9974d0dbc2a5c4615331bdb : SUCCESS in 4m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e136a62dca1142cbb6b904bf41358dbf : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea301d5bf6ef49f29249fc046778bb74 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/529ab09c85a6413a9c1497b26f3e2b1b : SUCCESS in 8m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f403a59f7b994cdf97d281dbbe52b18b : SUCCESS in 11m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/84ba1e7c1dc54ac2942a8fc3e39227d3 : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/67bc3b93f2804984ba17c91406c04e31 : SUCCESS in 5m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eb9c5e40f2774757b619db7cc05acc3e : SUCCESS in 4m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0b3acde22e8a4c1fbcc0a164bfa559ec : SUCCESS in 6m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/641924d041824733bb7b4dc6acba796d : SUCCESS in 11m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/44435326b1c6490e96dd0709026d84a4 : SUCCESS in 33m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64d83f2da01d479fadf60c1f1f69cc7f : SUCCESS in 3m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/431a4fe3e3a54d918cc8d43ce8b7b285 : SUCCESS in 13m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ab9009410b754c27a3e8e8765404934a : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba601f10b51043b3be2813dcfe6b1c81 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4a3c9d4fe7443aca10502212926304b : SUCCESS in 34m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7e233eb7b7854544a0c8cf8cd1a1e178 : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0dfb6170c6354a9081f226f173d7ea28 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e8f6a57bed24224b65eb3fb405dbe1f : SUCCESS in 7m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/61f2de59aec24d1fb1a92c36f07346f7 : SUCCESS in 6m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/09253c6f7422466aab22896315e3ec4f : SUCCESS in 3m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72da4efc370a4dadb381e6da2f6a04b1 : SUCCESS in 49m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a55fb61e28b04b408e556b617925d528 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"536a25bb9beedd7a4dbe5d60dbb7a41897333f81","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-03 13:06:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59fa5c756359e05de6d8461ecfad8c09f2bf0f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 13:07:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"410799719188930d742b2f04a16fd3fbf87f2b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:23:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd718e28c9f4ab39896c15316bd7eab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fc2bb3b0604242b19b5f3d45d417f2 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5564a29658624510a9416b28bb946869 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfc4062cc9f2455f8df7174cb2b25953 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c097fd75d13443499a07e95c204eb5d : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a829ebb415cf496a865e04c641478075 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9558350c358b49ec9b70aab471795328 : SUCCESS in 2m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da0744a1d9b34e2e9fa20ab629302af0 : SUCCESS in 5m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/534ec20b0c9e4155ab0f284f90e844f4 : SUCCESS in 6m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9f0cbe30ecaf4e508aabf90e96ef2b68 : SUCCESS in 6m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c381b3bedb3343a3bc65404fc739e780 : SUCCESS in 11m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c480fdde53ae43fb9da65f4f6da14396 : SUCCESS in 4m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c907817a6644495bb83c5f78cf0e3fbf : SUCCESS in 3m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/24cae1de529142c5816a337e21387b70 : SUCCESS in 8m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1b792ec2e6e041a9a5c197c2697c9434 : SUCCESS in 15m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfd8c16794974cadae2f6d442182752e : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b2a6238500d5459bb5004340a909599f : SUCCESS in 9m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36f2dbcb398743549ae48d741e49cd40 : SUCCESS in 10m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e61d0211b0c045a88ceb34fd965aa8e8 : SUCCESS in 1m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1945926c42024f98a2afd53d49ea186f : SUCCESS in 8m 17s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d6803dd309574706b37e0674055a8ced : SUCCESS in 21m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35f44ac6a87d457db66dceef313ecbd8 : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a38bee71d1a24255a45d1fe66c978aa5 : SUCCESS in 8m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8373cb5d5f754a4a87d00687da05d846 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61f10c245f804425b3ee48a380036b9b : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d53e67b45df47f2b483963332c0018b : SUCCESS in 27m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/200d8d7a792e4bb9ab4a1a4af42fc95f : SUCCESS in 7m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f75b9a638e29421fa15478c40b274d72 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/756750805f5a4be198c82fc0a81c00f9 : SUCCESS in 3m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/24ed9376fa0644fd8716a896a15613ad : SUCCESS in 21m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b124c37460654e3ca20ffe4d27de26b5 : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/507bc519129d4dce94bdae7285019a52 : SUCCESS in 50m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/405fc27b801f4805b6e4754af70979a9 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5078b317e87193db4f99830cdc9216bd4767e843","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:23:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6e3d01461bf4a27007f9ba68de3dd895b257fef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:24:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0bea72068fc44809b1360039dbea59c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217710afcb9a46a68c97a1f287aee457 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3bc8a52ae4767d5df609797e46c86ab633a044c","revisions":{"a3bc8a52ae4767d5df609797e46c86ab633a044c":{"kind":"REWORK","_number":1,"created":"2026-02-02 14:41:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/975403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/975403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/975403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/975403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/975403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/975403/1"}}},"commit":{"parents":[{"commit":"cc2a90b26f418a8d6a7e2a0ffb0586bd5a92f034","subject":"Merge \"update constraint for oslo.concurrency to new release 7.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc2a90b26f418a8d6a7e2a0ffb0586bd5a92f034"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-02 14:31:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-02 14:31:06.000000000","tz":0},"subject":"update constraint for os_vif to new release 4.3.0","message":"update constraint for os_vif to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I19fd01b7b37cfe12bb28f77f790c3bfdc75256af\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: sean mooney \u003csmooney@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nChange-Id: Ia2de44376782188896213b669f05dadaa0f9c57f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
