)]}'
{"id":"openstack%2Frequirements~975405","triplet_id":"openstack%2Frequirements~stable%2F2024.2~Icd6e7e86e47a532ecb6a0bdd0e18d37aa8ca6a4d","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"Icd6e7e86e47a532ecb6a0bdd0e18d37aa8ca6a4d","subject":"update constraint for os_vif to new release 3.7.1","status":"MERGED","created":"2026-02-02 14:54:15.000000000","updated":"2026-02-03 14:56:48.000000000","submitted":"2026-02-03 14:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975405-new-release","meta_rev_id":"68762c4ac43907ea488ef577c117a9afbc9feb4f","_number":975405,"virtual_id_number":975405,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-03 14:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-03 13:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-03 13:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 16:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 13:07:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"795dfeca7bcddf1a13b115ee74d031e51dcebe8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-02 14:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d48b2db7b4b4badaa87540a10653c27bf17a57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 16:38:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69890e58b187432a86d0c85965e09c35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e114ebdd8b04d2799cd651557328f5c : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64b45895dd8d4326bec7890a427a2dfc : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58e1ddcc22c2469a97ebb001cc62c012 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50e7017f67be47a798710ea0b67932dd : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a4e79ba3eee4b02bf81fe7b5190e457 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1bd0862fccc4e21bcfa3f51e51fd2f9 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e24fec8bb577455c951605757d6c26df : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eb5575f34f4c4b6baac9e4f9e3544715 : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6f9ab7bc535b48ac9daf32349e130a8d : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/471bee631a464b2b94e188cdf6801376 : SUCCESS in 13m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/301a139020ac4ed1a5bdef4fb02959ec : SUCCESS in 22m 35s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3ec803ef3609431d933b8a21754cf5df : SUCCESS in 4m 41s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/49fcefebff954791b88d498895505aa8 : SUCCESS in 2m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/6d01a2a76acc49778c6fdfbda5ff2e33 : SUCCESS in 4m 18s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/63fdfbb3f2934112b4d5c7baa271ec27 : SUCCESS in 10m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/be668f2ca27440058d241fd6ca75a368 : SUCCESS in 2m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6caf8b5764c14d4fbc1357417c46252f : SUCCESS in 5m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a08e69f2d9fb469d84912931c400dd9f : SUCCESS in 9m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b5f7906c0d3a4cafb9f6c060d0cb1ddf : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39b2870e09264b08986c2b0dd70fbdcf : SUCCESS in 4m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f618de39c6b849dfbd6b282aeba4b671 : SUCCESS in 7m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/14a3928815604d519410a11dda9efb66 : SUCCESS in 11m 41s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ce5108db18ab4d93af9c4f28b47bbbf7 : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/21ad842c2a1243868bde1c82e4d68c82 : SUCCESS in 8m 44s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/03ab2d4fd267499782ec939df7d6096c : SUCCESS in 3m 12s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/352fb5e2e31d4948af8c726a23b9e0dd : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/31e8acbfc2c94613b47bbeb8e0032a03 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/83826b9c5e5c42389a197685c1c9d7d7 : SUCCESS in 19m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c8ce565bbf0640bcbb5d5067729a7bec : SUCCESS in 6m 11s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d5f69506cc494d10bd3b17e0479b7181 : SUCCESS in 9m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/663c7353759643659c07ab74ef418058 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5393dc4c328149b98829f5c8ec5fa77c : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b899cb964e2f4db89156c97eeb7b08f6 : SUCCESS in 24m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/913fc9c122f04f219226f6ebea51aedf : SUCCESS in 4m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a27cca6f31c042c4ba286dda7b061df1 : SUCCESS in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cffcbeb91c934149aa26eb2f71c6f990 : SUCCESS in 3m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4d03b886e2cb4284a7cc031f936e4118 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66b593a04c8e419a8b005605029e018b : SUCCESS in 1h 02m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a758bb7c523545798d0bf5c7ec7f212a : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d14b052fb3be90e765338577c9069bb11e42bbeb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-03 13:07:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81b7fb58ec11627c828efd6762a424ea0e1ac7af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 13:08:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eab3c5de02b1e051fdd1e0c9c795dc420fd75705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:56:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caadc5d2078d455a8cc6c1d009548019\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebcbb3c47c594ce4833bef6ff380ac5f : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3645014bc2f640089c940fc20ea24f51 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a41dcdca0df44ceaad759b751872285f : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43b762ad0fe34503bb4edebae778e9cd : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e47ce91a1114b8482e814846e3807d2 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5ac7d27f5cc74ca590af3a1436e26d63 : SUCCESS in 3m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/07912b0156214794a29894aff64a0352 : SUCCESS in 15m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/94ce15a384594e138360dce73f2525f0 : SUCCESS in 8m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8176236b00e84a5e87bfface352e22f3 : SUCCESS in 18m 00s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/765eba934e4d4766889af13c42a43682 : SUCCESS in 7m 48s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/69a9a495ade746ab8e8bc8257fbe3572 : SUCCESS in 5m 08s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/16eb4ffbe0e9408bb6a84a430a0eeb71 : SUCCESS in 5m 06s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/00aa23416e194e54b53fe82fb259c299 : SUCCESS in 4m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8be6bbe1be33476a873bfbb07b16e2f7 : SUCCESS in 13m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09fe8e2fa7e9404d9e7112f871bbe2a3 : SUCCESS in 6m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bb9117589fb64e5ba926a5d15e3e40ea : SUCCESS in 7m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/523f233bb7fb4c68b78b1f095c3342c6 : SUCCESS in 10m 29s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/bee44d9b9e184bbf8b0a5b4efad18b5e : SUCCESS in 1m 37s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8da3fa28f3d04c9db9efe930b11d52f6 : SUCCESS in 6m 55s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3a6725f203db4ecf86bed7eb4b50126e : SUCCESS in 18m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e69b3973749442c9b9fff80f1256e59d : SUCCESS in 5m 54s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6c8ee429cdb640688ab14f00d0c729be : SUCCESS in 9m 08s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/04049a1eed7b4f178f0d00d245834815 : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e5d7f9dcccb4b03879be4968999a829 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/808abada241a4f359f49ae2ef686fe81 : SUCCESS in 22m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7a8ebc097f7a452cab1fb9d3b844e3f5 : SUCCESS in 3m 32s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/fdec5090df2c427e82d4c650920fd945 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dde018a9d0634cce934951d40bf101d2 : SUCCESS in 4m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ae70b0b5f1a9474c87d7e5163a34ebda : SUCCESS in 6m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/254952888d3b4686a664e44554fd099f : SUCCESS in 54m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bbc9be5ddd8427dab7174f401d6d02d : SUCCESS in 1h 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"68762c4ac43907ea488ef577c117a9afbc9feb4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92271810d9e66e3f8d5316c88b7d7bc95d293351","revisions":{"92271810d9e66e3f8d5316c88b7d7bc95d293351":{"kind":"REWORK","_number":1,"created":"2026-02-02 14:54:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/975405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/975405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/975405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/975405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/975405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/975405/1"}}},"commit":{"parents":[{"commit":"3f7620b81fb70f6133b26a442b5cce70d923280d","subject":"Merge \"update constraint for sushy to new release 5.2.2\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f7620b81fb70f6133b26a442b5cce70d923280d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-02 14:44:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-02 14:44:14.000000000","tz":0},"subject":"update constraint for os_vif to new release 3.7.1","message":"update constraint for os_vif to new release 3.7.1\n\nmeta: version: 3.7.1\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If7898a5bbedc6ad81182abf9e8f509c375fd89ff\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Icd6e7e86e47a532ecb6a0bdd0e18d37aa8ca6a4d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92271810d9e66e3f8d5316c88b7d7bc95d293351"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
