)]}'
{"id":"openstack%2Frequirements~975897","triplet_id":"openstack%2Frequirements~master~I7f8048246b00621232ad8a806973c9d67824c8a2","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-02-06 13:09:51.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I7f8048246b00621232ad8a806973c9d67824c8a2","subject":"update constraint for taskflow to new release 6.1.1","status":"MERGED","created":"2026-02-06 11:38:35.000000000","updated":"2026-02-16 23:56:30.000000000","submitted":"2026-02-16 23:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975897-new-release","meta_rev_id":"bbef946dc79a509b6b73d20dce5e51068d4babe3","_number":975897,"virtual_id_number":975897,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 23:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-06 13:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-06 13:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 12:16:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-02-06 13:06:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 15:37:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"4a2a847c8890106727d4ac32b46ad2764652475f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-06 11:38:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d61dbc1059bc0e46242c7f50f625cc85b28cd188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 13:06:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2bb6dd073dc4469a91c12dd18f3dbf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b4e60a873e4fb1bec62eb64c5252a7 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b371fd7095143b09e3bd71cef622774 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/870ab5e76894499dba6571133cf23eb8 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/096b1546320d46d296a886d07c991042 : SUCCESS in 2m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acc04f10472e459d94af9ae6f87fe5ba : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b085fa153bf45978953dca0b0fdc737 : SUCCESS in 2m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bef4a48a33b745fcab700e1b67ab2191 : SUCCESS in 5m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/239dc7ce0d9348c395af482ad59a2ba1 : SUCCESS in 3m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1c306c8a4f9347b6bd19f5ebc8a175b8 : SUCCESS in 4m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2eabef0e4a884009b2d69e8dd4a1b8ba : SUCCESS in 9m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/22dec8a607314a778a4f71208a5e3f32 : SUCCESS in 6m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/437d8e634b5b495eaca485ed502de67d : SUCCESS in 3m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a529f09d4a664b1895126b587bfd513c : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c85475b7b5ee4d518a078d93a606ad06 : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/37fea82a67d9465784872960d23e05d7 : SUCCESS in 3m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8847042553914184af795b4fd6396424 : SUCCESS in 6m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2cf9a8ebf794405ea518cbcf5718c967 : SUCCESS in 8m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1cea127c9c9546d98b4103af9fb6a31d : SUCCESS in 16m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7fa6fc2abdc4a2ba104e79c2169df21 : SUCCESS in 6m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a12bf496ae5e44b6b065dbb15fe41c1b : SUCCESS in 6m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bb4703e00f834b7dbfa4dca9da38d940 : SUCCESS in 11m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/45072aa6ae1f490b9c1aa5181536d494 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/04dd21f561f44486b30cc85ed60756d1 : SUCCESS in 10m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/833c8355981f4a7ebf0ae62f3be572b7 : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b38e8fc5ce8a4435a514938aa9d45615 : SUCCESS in 3m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/625473a6271848f081f200132fe6d810 : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/41f5192a5d8c4a09b0845665dd8943ac : SUCCESS in 22m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4cb5815332214aa4b3683ce65fbc49e1 : SUCCESS in 4m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f9f17b9b69194904bf61680dbc6bad56 : SUCCESS in 14m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8b1367100a384f7895c88505d56ea02f : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2befb971252411eadefbe5a65a05a10 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc202e6df11c4a13b5f557f59b639017 : SUCCESS in 28m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/93cd75ae34874c40a5c01a20746f4cd2 : SUCCESS in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3a409bdb7dfb4216841fe626c0cfcc4c : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f36028443c3f4813a28158c893244628 : SUCCESS in 7m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/36a548ce54264f4f999a46ff47d002b4 : SUCCESS in 6m 34s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d9d3483a9bd543a58298842f053968d3 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c19ddca1ad94855be1cbd858a0043e9 : SUCCESS in 54m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b757a5d5294c425996e92d5fd477c3a3 : SUCCESS in 1h 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"537ef5452122be43c7d2da82c0bb5a669551722c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-06 13:09:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7760ce36cc3b1b839b4e3998daeefb816569e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 13:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7c2724602a71e3556907f5546b882d4e5cc87ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 14:59:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/923b6fcb35474e41beae86eca9d25a2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635accac9d544036aaf44cbbe9ddacbc : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bccf2d3a8b244f28738ef3efc155270 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74058f4bad37488faf76e6391a6aa8c5 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f99054166a5a4dc4a4245f3beda8b5d0 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef94a23f21c94c229838a2c84e9ebd0d : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/089267d140e446b281959590227af484 : SUCCESS in 3m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e8fde44823ed4de6a110bc89f953c380 : SUCCESS in 17m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/001dccfa74f144a69014c689223075f8 : SUCCESS in 7m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1c19dc8892344c48b243549c8e0a2cd9 : SUCCESS in 6m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/245619c6b98d45259788f76a0d702431 : SUCCESS in 10m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f8ce4dc0b65b4459a553dbf4738d05e1 : SUCCESS in 6m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f7f779c69bc04fd5a2aeaf7e3af463b3 : SUCCESS in 4m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5f16afa2ff2347208657690a9df346c4 : SUCCESS in 4m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e1d91262c8df4a718ccb90d99696530c : SUCCESS in 18m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7963aafa6a054f1cb6090a900e0bfc3b : SUCCESS in 6m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fced71f057b34ff58ef7988ec67f810c : SUCCESS in 6m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/99e1b13a6e0740c19825259549195ae3 : SUCCESS in 11m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/66bb05e7dd5d495eaf94e9eec87276b8 : SUCCESS in 2m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3c347d47924b451e8b385ee48b7f9257 : SUCCESS in 6m 14s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d94fa29647684de1acf4634ff0da209c : SUCCESS in 23m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/162d36198c67434eb9f6d00026086a2c : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f45d382194a24af5a1e607285b6f94fd : SUCCESS in 13m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a8be7a69e2b42f4980f7caa5aaaaf3f : SUCCESS in 2m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae00d28b3cf3448caad1c839a48778cb : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b674a2ee4c8e42019e2e18491c680e49 : SUCCESS in 27m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/296f3e0267dd4141ab796ff82b381b48 : SUCCESS in 6m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2d9b2afd89e54a8b97396d71bd0726e7 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0010f5256a8b4d46b263701a525da286 : SUCCESS in 3m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f1c2e1dd85ec45d8a3ce68045f80b1fa : SUCCESS in 18m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f90604df1a764817ab1e21333fa9e02a : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/781e7bab569242c99f87e978f0a33b08 : FAILURE in 1h 03m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4bf340dd296423c9d1f99da04638d56 : SUCCESS in 1h 15m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2474808cf888aff8ed9d319f906bf94a14208cc6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-11 15:37:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6e91a26c4508e36778a81a1823ad5142e3d5e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:30:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bbbf2739c66462bb4b0a24597d67b2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c33f40a7aba4f188911f527504315c3 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6635bac0dfe40dc8a12da21d0f3dc90 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1bbc9e092864f9fbd03709df8d91d6b : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cb45f47e17f418eb8dbc4b9c6bf78b2 : SUCCESS in 2m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0de8f71b8bf4c1e9f02fcc5e8c1c7e0 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fae41999efb4a28b48fd98f2f9cb876 : SUCCESS in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9a8b4e5dad004ed58f2323888b43683d : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bb753841119848828b87d59f424477b6 : FAILURE in 4m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/eeef54dbc0a34f919746144167b28095 : FAILURE in 4m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a71bee99e186463e811128c6b9b46060 : FAILURE in 4m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/df62985f42664d939edcbfe3c01b12cd : SUCCESS in 4m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0bde1ad56a384a63a51e34539152c782 : SUCCESS in 5m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8caa27688f6044c29ef9d9f2aff7e0b3 : SUCCESS in 4m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b6e94467cc8d4952bf34b41018c9cb0d : FAILURE in 5m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4701c8232a314467a7f38f482369ac04 : SUCCESS in 7m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9299613479f34410b503c1ad5e25f1ae : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2cb9269273a642d7ba5f84b5ef705f3d : SUCCESS in 6m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cd656a9e17c24125abfa8f2c25c5dd00 : FAILURE in 2m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7aa4548db66d462f8e8efac80f1a4583 : FAILURE in 3m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/111677ddd05f4467aa004970bdc30d34 : SUCCESS in 8m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f40c7820e16a472c8d856a04d7fe7be0 : SUCCESS in 13m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/82ac0bdf99f04c8a8271f6a142425a6b : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d32d161ba1fc4df4861a5954cdee809c : SUCCESS in 6m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ff451ffd34ee474793412e3ae8cecb6f : SUCCESS in 2m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/84f5924ee2a5494f8dab8f0cfeadb2b1 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/976043b563df41d482a646e3869c75c2 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c2c1f963f83647edba3cf8e3eddc6c1e : SUCCESS in 19m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/193c041660cc4b07b3e4e219bc87317d : SUCCESS in 7m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/13c4c80e84e84fd3a11defb810b3dc8a : SUCCESS in 10m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e09e5cabd86745e284191942d1d7112c : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c67e887b57794202963ebf9cd4882f06 : SUCCESS in 2m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea93c3f7cf884d0eaecdd701f138e344 : SUCCESS in 24m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f6244716d43844f18a2897eb22273d31 : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/66139b76841448b78212c3901c417618 : FAILURE in 2m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe1197138b4445f5bc434979f060e10d : SUCCESS in 4m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/80ad28f7b7344ec6a508790727ba78b3 : SUCCESS in 16m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/30aef5b30a2f45fe8938ad00780857f9 : SUCCESS in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0965499169ee4072a061deb7e2dafc3b : SUCCESS in 45m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a83a4040eb9470b9b9a0c033c939284 : FAILURE in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"335330eacad082ff5b4ac2097916c3c9f9b9ccef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 17:14:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf19d8dbb1a45962aa73df0fd8fa6994f50b631e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e7c897607547a59e9a2c512a8929fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5946ee71a34f416bbb3a8dbbace5d082 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9d1df71389468e80434b5637247863 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8098bf7c234c4ce79448386a8466503b : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12d35ac9fcaa4b1885ce7b6b26c4ad3e : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c13d8a6626294da8ab980650647f38ab : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17b158208d944972a9f518a88f041e52 : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3989219b22d8499985836c524cb0e4ea : SUCCESS in 4m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b358f6795234349916569ac5a3e36f6 : SUCCESS in 12m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3914f14a90ab4440be5c9358ac6b2a98 : SUCCESS in 5m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c299e83e2f1c4db996958cb11d96a99a : SUCCESS in 6m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9dacf8674b164b0e947bca64c3603c61 : SUCCESS in 3m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d15e01223bd94496ad71975e22787cc7 : SUCCESS in 5m 45s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/edb970ea2b304f1bb3d9062a57985ee1 : SUCCESS in 3m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/357f53e28b2340ed896eea7bf68d7956 : FAILURE in 6m 26s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bda6c0e390384e41be36d5bb2803da71 : SUCCESS in 5m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/71b69db4d22848cd9cf69410fbbd78d3 : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/15718b4e6e534d929dcb8c073cba7247 : SUCCESS in 7m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/af95dd200d5240048e7d6e63d2e3154c : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e84453530c154f51a7c58361e2d4da7c : SUCCESS in 9m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7aed33ed2f194deaa993c1a553633e42 : FAILURE in 9m 49s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/96170872444d43f792d9c4fb734291f9 : SUCCESS in 12m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9febdf0c6c5148d78ff2f892c3961e3f : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e90c51e6a6db4fcbba96f5ad410c39ac : SUCCESS in 11m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ef03b900a35043798440b3d3fceaceaf : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c1510938b9eb42b69f6a4014a1cd7165 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c9490859e2f94994b1ef963c63cbae6a : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/efd06de65676471eb60a7bdca5bf0a33 : SUCCESS in 38m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d1522e67cd17423688904b5fa0be0c6c : SUCCESS in 7m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ecf418bf982342c6bfeae73bf4b166fd : SUCCESS in 16m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/837777de94d34cc8854f2bb2e8c0a986 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/182b4e19e9c648daa91780feca7ddbb6 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/234fca3ce6764d719237a418b519be25 : SUCCESS in 28m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/47d358d62ef442fd8b48d028d895e522 : SUCCESS in 6m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1ee2b29f5c214a5eb6a8f92e305124a4 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfe4c8773ef84ff199eefb9d383d6ce3 : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/47f3bb31986a43248e2c78ccdd42b14c : SUCCESS in 12m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bcfce0581c124136b935964cdeb3529d : SUCCESS in 5m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efac2e4e13c844feb2c00ca156f43d86 : SUCCESS in 1h 02m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8409f2c9f91e406faeb12b5264fe2d5d : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4c8aa9315dfd6c544bbc0536a01179c106356000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55465e488f4969181a13f07ff591a819104255e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:09:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39d961a026434af79434542c3e641f84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35038859888c431cab5a1149804c4b8d : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734556d2034b4174954479c2705924cd : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48126dadb6a48c1b88bf272fce3dee5 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0066f0ec0cda4801abb68c86394481f4 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0fde2fdfd1846928a490265d6b83abf : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/597328df1a444ecabc73c0a8d3de3528 : SUCCESS in 4m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f28882906e7542fe9b0e9393c1be0466 : SUCCESS in 4m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/006b7e83f25f41a0b75a45b64edfbc59 : SUCCESS in 10m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c781e138c44a48709787b80b96d430b2 : SUCCESS in 6m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/561b678ccbb147aca352c84e0c7b467b : SUCCESS in 4m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/417e40b92dc347a4a899a073d0ff8b11 : SUCCESS in 5m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d908d0d3bbc0488a9f2b8132b6a8f3d9 : SUCCESS in 4m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/00ff6052d63245a9a9414c294c5af26b : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/326cfa611797436bb6a7f8c813ca2bfc : SUCCESS in 6m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/67021c6d19604e75963afd9078f87478 : SUCCESS in 18m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e701f5681cb84c0e8cb242de677496bf : SUCCESS in 1m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3fcc06ac28974797a143c872dca6ff11 : SUCCESS in 8m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5c234d50dc4b499c9da4a3f272e34059 : SUCCESS in 20m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc819e2d9d234468bfd4681df3ba8867 : SUCCESS in 6m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8c1cf9764c4745d18aa2328f14227c8d : SUCCESS in 12m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/90c779b472c4401b9ac469f1f63c4bc5 : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1da08e379dbe45d7b817684f65d64d97 : SUCCESS in 1m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc8f688464564c32929e9bad781553cc : SUCCESS in 24m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c03f0073b2564190a7b74729e8976d66 : SUCCESS in 6m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/51538404f96748929e305881f9273413 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7977af3ebfb84e3b8b1ae467b10b2b11 : SUCCESS in 4m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/35c3232730a142f8bba9b71ddf6d0f0e : SUCCESS in 13m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/97cbcf12d7de42499f378e351d86281a : SUCCESS in 5m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cf7188f55b74737aa93cb44f3d5bf9f : SUCCESS in 32m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79884d1c27444af699b40c018dd8ece5 : FAILURE in 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"57d0c1462f17b1960067ff7a10751030553c1daa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 20:51:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4ae538531e946fb244d37f7422790c61825c1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 21:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba50fd488a52455bb634ad23cbd1ca7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4562cf056e48e2ba13a254d39a85fa : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861b3b19bd914f1398bf5da5b48fd40a : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea2f04b8c513452e8497e9c288da8b81 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff3740dba06c4cb0a2385df92c891666 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e38b603c238e4877b2201a2ae7c43bbe : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de35a5c7261f4b7397b1814267f1520e : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aef49b609a134f4998e3b54057da5f8f : SUCCESS in 3m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1283be916f7742d89201ef4ef224da5d : SUCCESS in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d4c543be88fa4a92adb3053ea15273b8 : SUCCESS in 9m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ebe6b27baeb34283acb4cb66f8dbc131 : SUCCESS in 3m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bd51f58e8504484ca4be9d89983f927e : SUCCESS in 4m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f0ac03ca75cf4f84834f5acd5bdcc95b : SUCCESS in 2m 58s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f1d40ae547544b7a9e53a929cd4d0843 : SUCCESS in 3m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5de1fab63b344c33aebfa3515934959b : FAILURE in 5m 34s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d51bb767e28642788d2b4d776d77a1f3 : SUCCESS in 5m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/41e92d94e99d4d3ba9b3bfccf13537d6 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/700c19ca75a545129ff243c890c1110c : SUCCESS in 7m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3b06a36b1dee4a4f85ffa59fe2951932 : SUCCESS in 17m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1129741534c3421bb51ab0522e900352 : SUCCESS in 12m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5ca4e032ee9547af8919cb561b5f191c : SUCCESS in 9m 05s (non-voting)\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/29a3def6789e49c9b70d09de01549c52 : SUCCESS in 10m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9202b4ba7657426f8796af0a87bb00e3 : SUCCESS in 2m 27s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4d89ca4fc6b7431b83389863ac3bed58 : SUCCESS in 9m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9229a14874dc426e8cd0ee0e3464f491 : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/01522f9bb178467babb7284535c04bec : SUCCESS in 6m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/77b82efd77264ed8addae7a443953d2d : SUCCESS in 10m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/74f8d64a0bed4512ae513946c9ba9a7c : SUCCESS in 23m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/05ca68f8c2d64757a75cf9043f2b7ed0 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b7dd477ec9a14bc8a9fbb91fbbe39d13 : SUCCESS in 8m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fb4e90c9678745619a0ce4ddcdc15889 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d9fcb02935e4954954c0820ac5c2a2e : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3bfbd815d109472cb21fe44273092784 : SUCCESS in 30m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d7ebaa0910d2446f94fd05af4a51c7d1 : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/449d2431470d426797ac885c3b8dfe5c : SUCCESS in 3m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86ec562136a04199815dac8625f31c5a : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9e03f18765a743d49a001f22affb9725 : SUCCESS in 21m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c6a5a481bdd44d1abb898b7dd037497a : SUCCESS in 3m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5641d1cd361492fbeee9e05dd7ced59 : SUCCESS in 1h 06m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e54bf9e190e4052b75d6981e688c462 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a56a907b8fd30aea0dbd16daac89b3e06ccf6576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 22:00:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"318a22b255a0c7968c0aa2cb36f0b70f3ab52ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 23:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/880274a745d64191b4748e7f6801963b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2518236f310846ffa70b75e97135867d : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e8ba22447e4496d9c86610c4189486a : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9310ddf507c24fbdb114fb6f271f1dbc : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac82e453a1054d338d9412c7095de02b : SUCCESS in 2m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a833ab77d4bd4eeda756a15480235aab : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d91049d2857e4c3f883b22b6e542f630 : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c38c77ad9bfe473bbf89a07d002aabd2 : SUCCESS in 7m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9671f755171b4a33a2d45a5ac71c7fa4 : SUCCESS in 6m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/51e2d60dc96c45a5a8cb6baa162b1da0 : SUCCESS in 6m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e08cb8ca400d4d14b44054e681baef58 : SUCCESS in 4m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/53d9c69bf6644b9db8fe0f6fa73e139d : SUCCESS in 3m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2d210ed3d5fe437bbc5afc77f264a503 : SUCCESS in 3m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c616d9bb0beb48dda5a45b198154e0f2 : SUCCESS in 17m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01058f7ef5d64aa49bcab767a151760b : SUCCESS in 6m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a02845ca2c63494285046f3f40057d18 : SUCCESS in 9m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9d2d68f0691c4c03bae44812cedcd1ef : SUCCESS in 1m 45s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c71338c6507d46b0bf7ee913ad383826 : SUCCESS in 5m 09s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ecee930c5d8d438d8aa1c16e21d656e9 : SUCCESS in 21m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7c3bdf40f6c4427aa9c0dcef61392c5c : SUCCESS in 7m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f346abfc725c4597bc53760dc61c1118 : SUCCESS in 13m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/46c7a523070648e6919b520c7c5ace45 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0129f52dda7d45eb8a98edaeb0f85c60 : SUCCESS in 2m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07d7ca67f16249d3a3821e752bc5269f : SUCCESS in 25m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c5818dfacf0a4aa7903a25728f50c720 : SUCCESS in 8m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4a187c93a12444e58d833e8d2f6ae92c : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59e12a80bed34869a236afa944f47b4c : SUCCESS in 3m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b0821811fd9a487c827ed8e292f8e2ac : SUCCESS in 17m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1786ca3912e24df9983fc495d1f838a3 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3aecb8fcf310444d86b0b2894d0cc033 : SUCCESS in 48m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f96d5823213e4abca202fbae615fe0c1 : SUCCESS in 1h 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a327d9ddaebb7aa3de73e9f1da69dca3a80b0fb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 23:55:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bbef946dc79a509b6b73d20dce5e51068d4babe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 23:56:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6a0c14f42a4e468f43e95292ac3531\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e7005f8e3b4fe98b981979b138a2a5 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"975d1f2e409aa5ead6cd1203c82d28f954fff45c","revisions":{"975d1f2e409aa5ead6cd1203c82d28f954fff45c":{"kind":"REWORK","_number":1,"created":"2026-02-06 11:38:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/975897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/975897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/975897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/975897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/975897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/975897/1"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-06 11:28:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-06 11:28:34.000000000","tz":0},"subject":"update constraint for taskflow to new release 6.1.1","message":"update constraint for taskflow to new release 6.1.1\n\nmeta: version: 6.1.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Commit: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Change-Id: I978860dd86d9e5ae06453c54cbca787d64412b79\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7f8048246b00621232ad8a806973c9d67824c8a2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/975d1f2e409aa5ead6cd1203c82d28f954fff45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/975d1f2e409aa5ead6cd1203c82d28f954fff45c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
