)]}'
{"id":"openstack%2Frequirements~976227","triplet_id":"openstack%2Frequirements~stable%2F2025.1~I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8","project":"openstack/requirements","branch":"stable/2025.1","topic":"no-module-pkg_resources-bump-pbr","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-09 08:23:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"38329":{"account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-09 12:53:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8","subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","status":"MERGED","created":"2026-02-10 11:10:57.000000000","updated":"2026-04-09 12:53:25.000000000","submitted":"2026-04-09 12:53:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976227-no-module-pkg_resources-bump-pbr","meta_rev_id":"7dfdc7e5fd77264453cbcfddc20ff99e6b814cc8","_number":976227,"virtual_id_number":976227,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 12:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-09 08:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-09 08:23:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 11:16:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-02-10 11:16:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-02-10 11:16:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-10 11:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 11:40:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-02-11 15:04:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-02-11 15:15:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-12 11:44:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-12 14:06:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-02-25 18:41:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-02-27 08:28:44.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"CC"},{"updated":"2026-04-02 10:13:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-04-02 11:54:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-08 17:58:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-09 06:59:01.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"REVIEWER"},{"updated":"2026-04-09 08:23:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f00eccdb3ef4797638f62fb3efee27283b200e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-10 11:10:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef47d80bc84f6c0ccb40e51cf097c511337cda84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:38:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f04b78414b2d479ca6bc494b4530873f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec06174049214d1e842300b136f1f1b3 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e375907add6b4f4c9b4ba2af92a8fbdf : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8cc4e6337c64fdd9f2b7e9682548062 : FAILURE in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e504702de09940c7b974bdeb3d4474e1 : FAILURE in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98569031370e426d87be1eb83517ae41 : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ca0f347708fd4a83bab5c898113155cb : SUCCESS in 18m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/274b85203d29444bb06f65e49be9676f : SUCCESS in 8m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dbdeb82d387447129fa82be7c513efcb : FAILURE in 6m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0203e40a0e754078826dd4541601d599 : FAILURE in 5m 58s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4c62ac9c5db5463d96e01afc50a65465 : SUCCESS in 5m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f90edbca9789427192e9e2d9b765c3a0 : SUCCESS in 5m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ab53f85ed981426badae22a32446e895 : SUCCESS in 4m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d986dd7885914b1dad0099898079469b : FAILURE in 5m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a32fee5285c94edb8c6ea8720fda0dc0 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6346b91503954ff0894490b57325a81f : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/66e1b87dc9f044d7802362e991b7332f : FAILURE in 3m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/46db5fb48a464acda9510569b232509d : FAILURE in 2m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35145a6694644e86879d25af006b9081 : FAILURE in 5m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2256550a910c42709d0d9e4796f8f7e2 : SUCCESS in 7m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/543e9d13864643e39d299be11c8c6d4e : FAILURE in 3m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dedb1863e0f14339acfefc72e7794229 : SUCCESS in 1m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d92319c336fd44b898e0891b2d876f67 : SUCCESS in 11m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b89513a6e5df4d21861c9e676edff482 : SUCCESS in 2m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6c97ad5ed94f4c8e9f7e6d6ecaf962b0 : FAILURE in 3m 48s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/df2fcd5a62e54dedb02efce412adbead : FAILURE in 6m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a1a4558d9d7b4bcb8c92ef4f5b807e87 : SUCCESS in 15m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6def01c3d1924b18b9c2a1fa54636481 : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/768c5f10befc4a69b67cfb3854311c0d : SUCCESS in 12m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/89d8871d52014996b68a210168a2f49f : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/100f9ca6b4364012ae2599b2a412a489 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7123398fff064404b27c3434060f3ac1 : SUCCESS in 26m 28s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/af8bbb7dac864c5fb94c67c7feb3d193 : SUCCESS in 7m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7cca205d43204244ab65390b03dd6c81 : FAILURE in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/734edd5d613c472fbb97e079e3e576dd : SUCCESS in 6m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c818a8f471844dcbb4f0b2e730bdae4c : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98da7dcbe40d4d1fa25c4f7c070f3d02 : POST_FAILURE in 13m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b5da34187194c7ab77331cdd57e333b : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cc7507a9f6122aa23ace497306f55e43c080c880","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 15:04:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42e5060550c8fc450bc4d96ef15f1e87e1dea7d5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-11 15:15:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae33ff4eea60cba41c85a2ae9223392e7ea77246","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-12 10:14:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c258d4c98d9ecefa0c231585755b478f1442b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 11:02:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc349a5b8fe84e3ab35c2ae38f6f53ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4745da948624200872525977fd1731e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cd236f949ed496e9df0c65dedc62fd7 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56f5e0fdc3614ee2abb672073d388193 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bc5d2fdc67c4b3ba80aa0dba915cb20 : FAILURE in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d7b14f9172143c08fe30d55ff0f811c : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e02fb42709934e649adb4bc83b5e35f8 : SUCCESS in 16m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4f8b0033a97495f9b212b2cd4988b2c : SUCCESS in 10m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f7bb05760264171935993295b72f82a : FAILURE in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/019cc620d7f74da993a665d8bca1a844 : FAILURE in 7m 36s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/47b972453eaa42d39d8b5890d97a6570 : SUCCESS in 3m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bd567d87c54a49baa72becdb2800a6c1 : SUCCESS in 4m 34s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/afddc2ab4b8c49e196ac34d0749b8af5 : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/44ccf5289df347c4b52d156b77eeca75 : FAILURE in 7m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4e88eb774c8849edb1a738808fb54a6b : SUCCESS in 5m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3c030bab816e4070bb4d260d10765e9c : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/be0a6a8d9353461dbe7983b473aac1c5 : FAILURE in 4m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fe12bfb7442042a081fae066bbef9fe9 : FAILURE in 2m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/101ac94f3a9242bd880be2fdc5dfd761 : FAILURE in 5m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ba29ae53c4d541bea153b80b2cd7332d : SUCCESS in 8m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7b2fb3bcaaf24c07bb3c2eacb8c4c0a8 : FAILURE in 4m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b29429e7e34c428aa88547811df1aaec : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6bed0786411941a1bdd254d64a8f0082 : SUCCESS in 9m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/480c303457b94c8d9dd7f595b8a583ee : SUCCESS in 4m 22s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/82381574fb574a398c61920287c828ef : FAILURE in 2m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b3f87f02ddbe4522a6f77c7178e3d800 : FAILURE in 4m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e052cc8c1e6d4624bc5dc8969095dd0f : SUCCESS in 13m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/81420a7388e3455ab19c64dea679c79c : SUCCESS in 4m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/43c22ac53c874d7180efbfbea8014ca5 : SUCCESS in 12m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/73de0d578c184c1f870de8bb2ec81407 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22f89272915342ffae3ddbd7c0cb1096 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3db0bfa879242c381e1faea92ffa067 : SUCCESS in 22m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9d7ee0fe50714ee2a7662b736e7f66bc : SUCCESS in 6m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/05629346c6ea41c1bb20be84bef31d77 : FAILURE in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbb508fbe2424616a9b6d23d46d9b985 : SUCCESS in 5m 32s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1a7177f0ef7e4805a7ba5fbd848c45bc : SUCCESS in 8m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6efd469dac641658e8c52af6aaf9e1a : SUCCESS in 33m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1d0b5743c7f4ee79f3c612de204b13d : FAILURE in 47m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f7b4ec5f0021533c21b1408315cabbfd580aa3e3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 14:06:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e8cd134d3fdc8744cb36324c2205370908a2139","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-14 09:30:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7259861b3ff0a16f8714c3935d8952ae8faa0096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 10:47:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3750050a31db4e3781783ebe948e5562\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae579680c9a4e328be45d2dd14abb1a : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fa9f59b45654e87b2411489498317ba : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79895930a87340b580bd4f29f7d073c2 : FAILURE in 2m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f58ff23c79184c44b300257e91763866 : FAILURE in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/193e28bada7749a4be0c0f0070c36e9b : SUCCESS in 3m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dfc91a1ea9bb4c82bb5ce72ebc5406d1 : SUCCESS in 8m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/206362138786499cbd035fd19d2a8369 : SUCCESS in 8m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b6e1a4969e65437f931b6cac0eedccdb : FAILURE in 3m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a7b411d5e5144609b9d75359657e3251 : FAILURE in 4m 06s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1e128fcc7a13417f968891e6745a8800 : SUCCESS in 4m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d7f77b8904414db4bfd19d8aa1197f04 : SUCCESS in 3m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/103403ba79e941a2a60f6424f04644df : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f3a1a847cf35448cb2d3077384e80e3e : FAILURE in 5m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0f49136b09b746a0ab4b83e84d28a312 : SUCCESS in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/46bba2ee64e74c5c99424a1eb6ef05f7 : SUCCESS in 5m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1c4aa17c8c7f4a608ae38373a906e09a : FAILURE in 4m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/eae3b127a7e14af087f6d4d35692a154 : FAILURE in 2m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98539d3dff714d7e8adda66821732186 : FAILURE in 4m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ed68340596f64abaa3845074f5e1b5a2 : SUCCESS in 8m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/659c7a5fdd5147cd9c3b0b8f5f45c935 : FAILURE in 5m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dce945822808438199b9184c3f064e7d : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/15498432b7b6442ca02ff79965197cd1 : SUCCESS in 9m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/39594f5c166c4c2cbf4519e7c0171e4d : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/197068958efd4a7eb2981ef932443772 : FAILURE in 4m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/533b4e9c7cb44e718cd60597b40a5dde : FAILURE in 4m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/244ca72da7d64a7b8151be9543857133 : SUCCESS in 21m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/75db593f81a14ddca88227573d27f61b : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0afe648a9fd14ac29b451aeb99f0a870 : SUCCESS in 12m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/37741aa45fbf4690b16fb55967c7a035 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a23fbc6738d84ced804ed0c13ca4ce50 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74ffd2991b0f46c9b151dab9bb953345 : SUCCESS in 27m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2ae9d837f57647c388f9e620df073aa2 : SUCCESS in 6m 37s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/64d88e36e9f245308c07b728219cf034 : FAILURE in 3m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/641855f94130440da2556e50c63d7c58 : SUCCESS in 5m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d0c7a145d79b4f6c80290324001187a4 : SUCCESS in 18m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca6c5895e9b04b5f849d0b229e3c4b43 : SUCCESS in 1h 04m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f81d074b63742dabd643d1193eeb3cb : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cd5ae265986299f1c2b6256d554eeab973aa75b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-16 12:02:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1e75e29c232ecfb79b4ca8a0f0c497b990f1d3a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-16 12:17:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d1aecc40aaa938297ba9eacb2176f9683c856c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:10:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/887f7c405d4f437ea4063b8184338196\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e85f9ba4c34edaab01b558c4e4cb18 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f6a8561a1804adbbce73f2af9a8c354 : SUCCESS in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f41adaac6ea64804ad6048a296fc1345 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e6e95c7ddcb4302ba3ff9792ce53cd8 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/744191220dc6468aaf5b939dd69c27f5 : SUCCESS in 3m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f818b8e6f33b4fff983cf4d246d5e0de : SUCCESS in 16m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bbcbbed2ff2444efa715fc108ac5b570 : SUCCESS in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1f927b3a706a4427b1692efde71a534d : FAILURE in 6m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b3d352b9c7a74f95b133ad128bf5b74c : FAILURE in 6m 37s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4c1cae25fe504cd183031fc0736177ab : SUCCESS in 3m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/93ab6d89af5c40519f8e9d9ad2e9bdf3 : SUCCESS in 5m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f6aa3ec463054a0ebf608d2a56464ea2 : SUCCESS in 4m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2a962af887654260a09256b608d93ac9 : FAILURE in 6m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/04495052369c4302b60151efaab88547 : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c5270891443c44a1bad512e209ceffce : SUCCESS in 5m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8d375588f8194507af4469d3381ade45 : FAILURE in 4m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2cf8c587d5ed41469836a61ef14c1a75 : FAILURE in 1m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb29f961affa45c08bfe43d4d3580ead : FAILURE in 6m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9b97bd8be55f48f481963db3c7733229 : SUCCESS in 7m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e9795c646f024a6b990997081b836fd9 : FAILURE in 2m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5b2419d859b24ed2a90443c3414a54fc : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ac444fa4648e47ba8061b8cf3133a429 : SUCCESS in 12m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ce5a46b29a534c5db80787fadb0fbfaa : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/47ae6a7e250f4775b97e6ecc6a6f174b : FAILURE in 2m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/14347edb3e6543199fbf703ce6d07e7c : FAILURE in 2m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/13575691a7794f76997611b335d7f85b : SUCCESS in 14m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/55e86c280fb8497e8f05cd63423efaf2 : SUCCESS in 4m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/58f1831fc020426ab4098d61df69f36a : SUCCESS in 14m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3b18e826022a4543a4eb5977da9bf4f5 : SUCCESS in 3m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9277ff851410450696e4beb901cb189d : SUCCESS in 2m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7a4bfae5b9f4ad281f753ef511bb202 : SUCCESS in 26m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/180cc296ce03459dba8e13a79cdbccbc : SUCCESS in 3m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b684ebf2007048a59aaafa43a902b78b : FAILURE in 2m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1405ac6359154a2f8b08121f10d0653d : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f35d0a253b0f49d19c052ac9d76b2b02 : SUCCESS in 19m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91981986ec474392bd5194f734ed412e : SUCCESS in 40m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1d82b6cca1943f2936b53ddce2d83da : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6104442f3e18edf5b51febfba5a69c9b40426975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-17 18:32:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e8ede668087b7fcbff37571494260b2a48e3ad54","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-17 19:05:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e72f23928d99e017b70bdaebbda7f3f72e8532ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 20:13:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b74d8d2a2dd8430e8e791093cc232c1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ff68da751641c992e45df03b22ce4f : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/920de06af2874883b3e278f6d5cbb953 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af57b3df57024cca934ff6421cce8ec9 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe51c882d56a46aab680a2f0fcaf91ba : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b34323b271d74845a4f43697dbce4296 : SUCCESS in 3m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5564c9b3a3614d6488f5cf7dabb603ee : FAILURE in 8m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d016718775ed4a338778b7199cd24a65 : SUCCESS in 7m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/271c13943d8a4bd3a8def21d3f62d30d : FAILURE in 4m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6e8769a930c3487bbe807ea94d45d9ee : FAILURE in 5m 20s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4be7d8a40e9b47619320185911ab456b : SUCCESS in 2m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e9b604c55ba146f587548521280c820f : SUCCESS in 6m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8a436fa19361440fb7aeba8457c4f9f9 : SUCCESS in 5m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c6d8d2d1fa0f49509d003b844e5d2ea8 : FAILURE in 3m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/640ff19aeb764610acef8db57a610f85 : SUCCESS in 4m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9f5ad41bf86544c0a0b53eb44dc545fc : SUCCESS in 3m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3e07968f32a145408570ea2cf14eb407 : FAILURE in 3m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/978114a230fd444093843dac031b50c9 : FAILURE in 2m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/358f921b577f41cda9095b310090af06 : FAILURE in 3m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/54889fbae0bb442db81d8a899ec01fbb : SUCCESS in 7m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b9024310782045d4953bf43951dac1fb : FAILURE in 4m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ba7350e4636842ee893752aa1b6b640e : SUCCESS in 3m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6feda51108064f7baa4a53bc580a9ad0 : SUCCESS in 9m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a460c721925f4944b5231aa624825f2c : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/190cc177cef14771bc5ff982de912e56 : FAILURE in 2m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3fa08c4d519243f08366bd05f63b2980 : FAILURE in 3m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/570fca93ef674f5da30bfa06ef260fb6 : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c69a066b95cb48cf8baeab1fea9be9e0 : SUCCESS in 6m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/df691e97b28543f8859ad28167172000 : SUCCESS in 10m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3aa3ca83191b4d238a651feb4d9ea609 : SUCCESS in 3m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/125eed123fc94b04a011038b5020b4f3 : SUCCESS in 2m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53eafbb51b2a41f4a2008d190e19c39d : SUCCESS in 26m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c1b828b375cb4a9a8e9e9a6b5300bf48 : SUCCESS in 4m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ef93b22e12ee4cbb847543cf6343dff6 : FAILURE in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d371495e7cf44dd6aa320593a4032cfd : SUCCESS in 6m 11s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6721e9ab6d1b4554ae3d748ee7c5c057 : SUCCESS in 17m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/addc26cbd57746a1a08a11ffbf5f31a0 : SUCCESS in 1h 07m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17fd01252d654707b2d48ef289d483cb : SUCCESS in 1h 39m 18s","accounts_in_message":[],"_revision_number":3},{"id":"62d970f6c217337a768d2bee4554d568046edd28","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-18 14:09:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87b6eb6bccb8b1fad5a2908180d88f7fe6ffa225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 15:57:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/759eb2e530a745bf92b951b3d3477446\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28db5a85676f41f6abb80b38b002520d : SUCCESS in 2m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28a9698bbf2643a99757a8bee78f5486 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b96e6644b9714a2ab0555969c3c3f28c : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/058c426a9ac74759bf56fc64cc53f538 : SUCCESS in 2m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb2780b108074d1ea8de96ee4fd00d20 : SUCCESS in 2m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e8ca5517f7b64a58a427d7c20abce45b : FAILURE in 9m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a95015d55fdc4795b101f4150dc8924e : SUCCESS in 6m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/048d2d9d3fc3444c9c311af60a997d18 : FAILURE in 4m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c76c06968524408a97ec4a27d5061ddf : FAILURE in 2m 55s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/489e5adb188548d1965e2175f34fce8f : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/47ab6359183b4b24b5a02bcb8dd1efbc : SUCCESS in 6m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1ec82e602b1940a38fa8b6ce212754e4 : SUCCESS in 3m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/de54fef254c64e8581bd8f4361bb162c : FAILURE in 4m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b9a1d0f7b9c54e56a7af2542a137056d : SUCCESS in 3m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/99e657ccf4d54aa6bc323f02ff9d302d : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e37e57482e404c458ece9cbe573de593 : FAILURE in 5m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a071239c21544af78510b7d7ee6a2a74 : FAILURE in 2m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e869a7457084afda60418c031e6011f : FAILURE in 5m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/35f2f448c2d84fc49b0724365721552c : SUCCESS in 10m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0f59687d108341698818881ffd63f535 : FAILURE in 5m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f26003f79825423bba745675c65791ac : SUCCESS in 4m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/54690933bb494eefad7e9e958ce15695 : SUCCESS in 6m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/90592da7688c428f9ebcef8811a5ddd3 : SUCCESS in 6m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/97ff7be921754d2cbc76fc3997ceaaf3 : FAILURE in 2m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b97e6ebbcf234b77b1edb5ad5e2f72bf : FAILURE in 5m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/19f56cce907b460e9d03998c9d012cc8 : SUCCESS in 24m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/655d2762067d456f8ec9beb56b1fe5b8 : SUCCESS in 8m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/19a3d62d6fdc4f85b1da793e6bdd54f8 : SUCCESS in 14m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c8bea6e5b4804c5b9289472c11492a07 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/245ee396d08f4dd28bacfcf17f807a12 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b92188c40074f718645c96b7464a6ff : SUCCESS in 27m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f9c5cfe43684ffebb93521d25ac0b73 : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ca124f2efc9a4849ae855a1950e43258 : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da8873877a754959975bab4a1026402d : SUCCESS in 6m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b79d6ed60fed4bf58ba3cf905b596fbb : SUCCESS in 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd52ffb714d44431815800cf77acd715 : SUCCESS in 1h 00m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef074e641bfe46579c8daf10e7c869f0 : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":3},{"id":"8393aa13766609c350a80db6e03c44eba88dfab6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 10:20:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"feabf35d87daa95cbc9aecb6c28018e3f086442c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 12:32:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d51a136c67c84f3bbe89a3dc5a9fd6fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be77431824dc4c7da1dbf312ff37dee1 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68d29a5aa3354bedb833784bdf1251c0 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e222c6a28ec4f0c9eda74582fd56bf0 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b9d19df5e1c48e996b2e1da1d395c23 : SUCCESS in 2m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9820cec4723b4b4b93c2b76fa658d88d : SUCCESS in 1m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c808db0140c4dd198ec57a8825874b8 : FAILURE in 16m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d32f8fc5a5e4fb4bc36e0420a5bee2b : SUCCESS in 16m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/504dea2f69c04741bd3eed9db7b9e1a7 : FAILURE in 5m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cc20cd1d941242d185082c31a1835c77 : FAILURE in 4m 08s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/963813ed0f5f4814bf48f801a92f3f5d : SUCCESS in 4m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bfbbe41c0f6b4031b553dda4199044d4 : SUCCESS in 4m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/848697aafcf84d8fb7e57b26544626aa : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/560caf997b6241e7aaf6fc681d48b144 : FAILURE in 5m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2383ec6927724bd6b2dd92856a2d201c : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bfc909595a494aefa04f998c6b7ff254 : SUCCESS in 7m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fa4bcb9583b74754a6baa3a992061281 : FAILURE in 2m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7ecdbefcac434832b8d977bc5b019719 : FAILURE in 2m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/500ba94dbeff43e184ba7251de677bb4 : FAILURE in 8m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dc2398c7af88437a985266de50bdc848 : SUCCESS in 9m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5aa340fabfd04477b92798fd1d7af9fb : FAILURE in 4m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cc898c943df74f5388285c09412da70c : SUCCESS in 1m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2540f7710b0b45a38214828f3fc4e50b : SUCCESS in 9m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8b0afd64643e4412a2c45e6853c35478 : SUCCESS in 4m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/63eb4fd2ce9949a5ad7d2a6d6900c51f : FAILURE in 4m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0cff4b270fb645228566ec02f5c7b5a6 : FAILURE in 2m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9bee0051b9ff457f99ded66a6ecce953 : SUCCESS in 20m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ef5602a72c6b4d97bd9ec9660e922b8a : SUCCESS in 7m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6ce88f31a1e0419aaa3670ff9c9ead2e : SUCCESS in 15m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/90775e5073c44f738342db6c5730f92a : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff2c5810c67045929efd3ae82b57dfab : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8dbae14fb194b91b74f3a9580b224e9 : SUCCESS in 26m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8accdb5773f74c8c8da10bd76efac6b6 : SUCCESS in 6m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b03316cc1e254e39ade06862c475abf0 : FAILURE in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b578ace52f6041379d6ee0a97eac1e99 : SUCCESS in 6m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2c630dfb6f6247a3b62e11a9a4d07e2d : SUCCESS in 6m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29724aa6edc746728b3edaff296327fd : SUCCESS in 1h 04m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4efa8267d9974e5dae15bfe031f94e16 : SUCCESS in 1h 59m 39s","accounts_in_message":[],"_revision_number":3},{"id":"59d2fb8689c76fd5ea1d577b5e8f4b6d4967e0cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-25 14:09:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b3ca00335e2361d4c9870062a803d43a52e6136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 15:52:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96c395cb7020447da0f65c0ae6bdfd88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00d3b1924e5444dbdb78e432def18e2 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4003fd37f36469896fe55f86167a1a7 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aebcf8f098ca49568bf6d49727e17428 : SUCCESS in 2m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76a53e1943034ffcb863339d7fc0e8cb : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff5e4f7463c04f08bab7e622b22e7f2c : SUCCESS in 2m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bbebf11fe9524145912c34aadbb47944 : FAILURE in 16m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6668f64262f54098a956bcf90e28a031 : SUCCESS in 5m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/72129362c49c4769886235650594af0a : FAILURE in 4m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/262cebe0d5de4dac9c453a761de9257a : FAILURE in 2m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fd228bcc9d0e4fba8aba801a71b33b92 : SUCCESS in 4m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f36e7fad63cf4006a008950d6258c35e : SUCCESS in 5m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ee436a4f2a4a41e0b175cea1c99182df : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7d2090391e9c4795879806d4679b68ad : FAILURE in 6m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/85d51279105b4907aff6421bbde83597 : SUCCESS in 3m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9f3883b68b5f4bc88e27407032f9891c : SUCCESS in 6m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/af08babef36a43bb8465074dcee6977b : FAILURE in 4m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/29ecb0b0e7754d4988700ae4b9c7e76e : FAILURE in 2m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f3c628f89a6406588edfc6ea6fc5c93 : FAILURE in 3m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9452857c7c984a49bc67644d6d603cfc : SUCCESS in 8m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9b707fd0aecc4872a519dfb085513ba8 : FAILURE in 5m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cf33fdd0f62b48daa4dc35aadc76fbe9 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/55d3bb1d3dd6411ca9d85fc4a24d5fb2 : SUCCESS in 9m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2a828754c2094da295ca1d68c7968e89 : SUCCESS in 2m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/547f5d513f154de581cf609a1ac25d5a : FAILURE in 4m 22s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/748e7aa8989141e0af1542cf7cf139e4 : FAILURE in 3m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/195ceda8534b440c8f92882d39e8424e : SUCCESS in 14m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f9e4495b7f104f8cb35f78df7bc8eba4 : SUCCESS in 7m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/800b9220938444f09c9e0eb96ca90f54 : SUCCESS in 14m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/91ee8b915dfe404eb4bbaa2be4086b91 : SUCCESS in 3m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dcdae9df4bc473ab23b52a34443f822 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20a8a5548f684f0285971491b9633838 : SUCCESS in 25m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d0fedbb691e447598e6865fd1b676fce : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/807f9161d6a34396bb3cef47460c01be : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adb47e2776794e4899de2ea56ea6bac1 : SUCCESS in 6m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/31ad056677ee46e2b6bda797758088f3 : SUCCESS in 18m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c78f5d26024423b29afc5142f56f5f : SUCCESS in 46m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c058252b79d64b188e92ce5e2703797c : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3dea5ba7beac1b2b50b83b3aab42de566d0ece0c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-25 18:41:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ea922e61792668a962afd43baaaf443a9f55215","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-26 13:06:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdc1a9fc4263150dffcf3290f08bd186cbfd3699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-05 15:17:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b6a0ee6f7a765b8f3bb268e7998b8f66290c363c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-05 15:47:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"270f23ca121f189110394601df3691eb1b5d8d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-05 16:01:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"66ef0e073cfcfead28a0816d3e5082f54e7d4354","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-05 16:08:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac77708fb975c5d32786a28568b33e1305ea4509","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-05 16:13:03.000000000","message":"Patch Set 6: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4762a44b38c04cbc7821abc62d3f0a391f2a72fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 17:45:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a55f2a071bd743d0988041a8fe4a8b2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc95aea8e83d42a3965d77ff1169a7a7 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0a3f84574af4db4845b91270c0e12a2 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1fdb324ffb0486abafa89146a2c9a2c : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37f4d1b9fdef4eedb24726b7c58d5809 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e2ffb64452348dd9f45ccaa243fd043 : SUCCESS in 3m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/45d1f42d0e044db580cc4d31b56024d8 : SUCCESS in 18m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1812608cf62843c4a75ca10bbd0fe240 : SUCCESS in 9m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cabd059404024af4bf451a1bb5bebcdd : FAILURE in 6m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/184957d29ccd4e189d6542f0afb66c95 : FAILURE in 5m 08s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f4c317e7413c46d3a7fa0c7c4e7cfaa4 : SUCCESS in 4m 56s (non-voting)\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/00bbabcd8da44808926c7faf8ce5d518 : SUCCESS in 3m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/35a1d160aac94f19b26dbe32e4d77132 : SUCCESS in 3m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8d292cc7218c41cbb8aade30c4a04e3b : FAILURE in 6m 43s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9b89ae99aa3645babaa27fe1cbe432b0 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fea8e89c9d314e2da6140ca639ec1510 : SUCCESS in 5m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fa7b713074df4a069c1a5b9ba59e36fc : SUCCESS in 8m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9c9e72f8d8e44c099739204268fcabfc : SUCCESS in 13m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ce2502b3a6e45d387c470cf3760ee9e : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/62369ef9c5384025a2ed97347cc17226 : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8c880597ce614adfa977e5e3baa94ecc : FAILURE in 5m 52s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d2b778e81eff416cab6dd56380b6b1a6 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c25f5c68f71f408e9b64db5ab9a58126 : SUCCESS in 10m 56s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7de0fa3f200d478f96c1ef189b4443c3 : SUCCESS in 4m 29s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e3611672189e45c5b668361337e119d1 : FAILURE in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ddb71e026aeb4b7bb743a3e258fda81f : FAILURE in 4m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/36b755ca33e84e91ae5de9b14f461c8d : SUCCESS in 20m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c9d3088593c74f668221f857898d4f5f : SUCCESS in 8m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6b773561ff564b03b0288b22b8ba8760 : SUCCESS in 8m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6544048591f844099c6415f9a020359b : SUCCESS in 7m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c85e5a7766a45009d3fcb2207a9d4b5 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d80e77a547b4f2bba06ce72bc28168d : SUCCESS in 24m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/70b48dbe8ba04352a0d8d2ce70231f10 : SUCCESS in 6m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c80b0b15042a4712ba76c85ed03c67fe : SUCCESS in 7m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f28db352a5a4ec699154e57aba8cdb3 : SUCCESS in 3m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c0bdb7c7fb124367bc2ed2804f434783 : SUCCESS in 18m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3580f5fbcb8d4928bc38a1c9ac239832 : SUCCESS in 1h 00m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac01444a1c5b48e893edc5cf52a5bf8a : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":6},{"id":"567761e09c3a7dae710d34f04102908905a4b5da","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-11 09:07:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"399a4dc7f54182b1097020d7152b8262406d50b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 10:11:18.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d9938d20fdf4f5d93efacee5953f222\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee999c2398d941ee92827ded836dc4ee : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/513890b50d0548529b0533ca94f2dda2 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cff7c29775494c78a157546bbf4de128 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b0a7c8ea68d42719f011e003514fffc : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3510adddb8ba46a49b39bb3464a7f646 : SUCCESS in 3m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f1cc54ddc5ff4743af407fdf60e416a4 : SUCCESS in 7m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e88b16f007c748d1bdf883626ad24fa5 : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/faab38cbe41a415595a17d8dd7c313df : FAILURE in 3m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0c0dd45bde4a4308b988b20ebabdcf7c : FAILURE in 4m 46s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9d54ac3f75cf4f979b6f0a35eecc5d0a : SUCCESS in 6m 43s (non-voting)\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2c3477ddb63044a895199d61476595af : SUCCESS in 3m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4165cb88dea342eab94cbe4f9c6740e5 : SUCCESS in 2m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fc4ae1669a6744f2a762023677256df6 : SUCCESS in 7m 46s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/87f157b1a1ec46d085beb4e9b4c8cc5e : SUCCESS in 3m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/094f960ecdb343f596ba1726b6834026 : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ede4ffcc72ba480bbf877124605f6bab : SUCCESS in 7m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5def4affeabf4675abe90df7fc0d8bd0 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5208c415134d466b8ce325bb13a7a470 : SUCCESS in 8m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2815fed6357d40fda5025ed6dabb162b : SUCCESS in 9m 28s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0a6f368303c740599d81a7098d9e580a : FAILURE in 5m 19s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ec4f9f6527a4419495d43b9320e8e96c : SUCCESS in 3m 24s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4c385a0bfd694c88870582b20b6739bf : SUCCESS in 11m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/45bf4ff03a65420f8e4579d15f8d1eef : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/748809d112564859af159d25a2afad28 : FAILURE in 6m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/48f6a10983a941b9a0167335927bc6e6 : FAILURE in 4m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f9417884544d4657b5bd70248230a732 : SUCCESS in 18m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b821f7d62d542a9a17f740f5c88eacc : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9d2e4f9bf41d4233b626c30cff381dd9 : SUCCESS in 8m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7ebe35fffe8d4e12bf8d7ee6add9b438 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e232994c05da496da56b3657293d8787 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6f3bfc4a48a4edcac295dc1dd64ec78 : SUCCESS in 28m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b49575bc24ec4c9e8513101c03e0314b : SUCCESS in 6m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1c40f554aaea44ccbd0f4ef3a849d74e : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b975b84539ac4ec08b733bc02dbc199b : SUCCESS in 5m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3b17c6d8e01a4cea847fefe7abc77ebb : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2ff505bb71042fb86ef63c2a5d0ae19 : SUCCESS in 35m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e9d522dc6c94fb094aa25ea8d79def1 : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":6},{"id":"85743ed38ee4e2ba8836c1804e558133269670aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 10:12:00.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.d/project.yaml\n* tox.ini\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"803613f093798e0902071341b3a96ae338ba51a8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 10:12:29.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"800428b3e03871f2e465e1ec503a420fc359859a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 10:12:49.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"48f2270ee4e97b62a8d17a3c9a15d2da0854281e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:12:49.000000000","message":"Patch Set 7:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/requirements on branch stable/2025.1.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/requirements/.zuul.d/project.yaml@stable/2025.1\", line 42, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/requirements/.zuul.d/project.yaml@stable/2025.1\", line 43, column 9","accounts_in_message":[],"_revision_number":7},{"id":"2af2fc7b28fa91b0f60590715a669fc39266f7c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-02 10:13:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c3a45210e0218793d63fd3ec4edaa25dbbd5538d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-02 11:16:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"8811ff3a58cca7c11dca6e0e558ea39565d0d265","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-02 11:54:09.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f53e199a37b75874fe520c00f715b5972d4057c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 12:10:29.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af117446adfa4a3ebc7c4469644ecb8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a8f0dfbc014a6fbc9913817f1f0545 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5562d84be454cf1b1e59f4314d51729 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9a175bb822d4b97bf950767d08754fd : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82eb60885ea845a693a2c6bbf644b5dd : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17c318d10ee845c8acc19670970a7286 : SUCCESS in 2m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9281ca0ff00e4947b5cc67523bc2180d : SUCCESS in 16m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fdbe58002dc94796992fbc341b896529 : SUCCESS in 4m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ea04ead870534c739dbc9079d85a8922 : SUCCESS in 7m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d17eeba54f694c40a3b9eb5d5270496c : SUCCESS in 9m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a63234997d424378bea88feb5bb447b6 : SUCCESS in 4m 44s (non-voting)\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/90dd3e8c278b4801843a8f08e1ab1e12 : SUCCESS in 3m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/02699e83716e43609afe7631003fe77d : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/66c55e4b84a3418993e47a0780caa19b : SUCCESS in 11m 33s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2336bf59a4a9478da44cf3cec76ed65a : SUCCESS in 7m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bda9009ce7824377968b330dd3470bd3 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/645bfa82aff2457e9ec1e58768c5d2dd : SUCCESS in 4m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/56e8cb48ac6d4ebdb5f5309db96b55e9 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74b39a19aebc45f18be2800f9b15f82a : SUCCESS in 6m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/774791a20ed34d038e2c1a8464f34d69 : SUCCESS in 8m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1a49feeaae064c5ba333ff08c2db38af : FAILURE in 6m 41s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7776401aa0e44d169f65529da8a32696 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1ffc7185a40c4a0a85d1386ed13bde64 : SUCCESS in 6m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eca33ed691c14732ba392b576e897503 : SUCCESS in 3m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b646a66f9308420192aef0e615a8db53 : FAILURE in 3m 48s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/129f8d5ab0ec4202b8cb255accc08d54 : FAILURE in 2m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/515f262739684470af7960230081c6c7 : SUCCESS in 24m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/95237036326548349d3d07cb681d68b4 : SUCCESS in 4m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b3113e069a514ebca1423336959109d9 : SUCCESS in 8m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/078debbbefb04d91a62298d9ff76e1c6 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9422ee00f5a4affa13db0323b6aedb4 : SUCCESS in 2m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bc632231c834e979336823ad39552e1 : SUCCESS in 21m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1c62718c9c89459b9fdeb11e93da9b75 : SUCCESS in 10m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e5233602e8a64694af3b0fa667c55000 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f81ed5b5fb14491bde2477ec93eda39 : SUCCESS in 6m 17s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/69333530fd5a47629a091390b6b67d8f : SUCCESS in 19m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bbbc431a9624ed99e15bbbeabb6e781 : SUCCESS in 35m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f8b4f892da94948a022809175b7b8d5 : SUCCESS in 1h 50m 37s","accounts_in_message":[],"_revision_number":9},{"id":"29f70d36b2a057b42cdecee37f5082f6fff6d662","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-02 12:27:20.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"66f82377fb9305ab0597bf10f330a90264fc6c43","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-08 15:52:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d31ee95cb040ac8d083a8177784767d6ccbfae82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-08 15:56:41.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"a984082ba4df05593bf329583435cbe7139764db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-08 15:56:48.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"58afdbcce4cf5d85189468ac59895d17a55688ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:56:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43003a6302649538958538d7f10c03a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e4431e271744a9a2f1c674534759c7 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcbf4c4b2a3543d8b7de75911940130f : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/674982ef3950432d9c18ccbf9b47c105 : SUCCESS in 2m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc6b531d32054d4f93f47d40f127cfaa : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31eb09f5882648909ab90d77ee1ede2e : SUCCESS in 5m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb74a99169b74c3381c67c9d20087c4e : SUCCESS in 8m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5fd10caf0d5d42228ff896641e8eb8ff : SUCCESS in 5m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9a2d7f1ad6564a67a6acaae08410156c : SUCCESS in 19m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a08ad9085d7548d58f15a62a1def99a6 : SUCCESS in 9m 24s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c79e938763234b60b5e7347e9f9770eb : SUCCESS in 4m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b3ae94dcc1654ac3a65e484396b69367 : SUCCESS in 4m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/225c927a937b4d9faaeb5f2c0dfabad7 : SUCCESS in 4m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a2fd097afa14425c8f8a4d6914797c03 : SUCCESS in 8m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b76adb32c62e4a439cf600c6fc862271 : SUCCESS in 6m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bffc7447f3e14c8fb8ab46c400e4c18f : SUCCESS in 4m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/51c8b228556e4bbb810a4ad829f0b4f5 : SUCCESS in 9m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5c6a9f4c66294a5aa38058ba275d310f : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dbc21e005a794fff80b21002722be874 : SUCCESS in 7m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e502a7f091c549e09164966a0c9968f8 : SUCCESS in 8m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/710a39c0419e4082ae729a65a6eca499 : FAILURE in 3m 59s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eb2d4ef61372490abc3ad1a308c4e6e3 : SUCCESS in 1m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3ed61869324746b39765d961b585d15b : SUCCESS in 9m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/549840d2d76c4d4f96d86fd1c6aee39d : SUCCESS in 4m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cc70a348bd7c4c97a339f0fdcb9dd74f : FAILURE in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/620e768dbeac4e579420459f7a7dfda4 : FAILURE in 4m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/042600e2845f4fd49c3102b029994215 : SUCCESS in 16m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/865eef10b17643fa9f80d6a78db144d8 : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1ec1db9a2fa347cbb392cb559e71ef4b : SUCCESS in 11m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2e14d5aa72d342278f8d9ccc7873953e : SUCCESS in 7m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a620bbc49ebf42bd9f19854e29bcbc1f : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbc22656057743568c9d074b28b74397 : SUCCESS in 23m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/badea638e5e74a9ebaf1a2d52d22fcdb : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4c624269abb3401492243afc9702c47b : SUCCESS in 3m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e193bcaaa98c4b75a6775d50e844bff6 : SUCCESS in 8m 07s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/727b26c0acf24b86937a6b37edd8b140 : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ef4002e0ecb4ec2bbfba0747d861947 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e6adb20cb0e41d6902eb48df6da5a05 : SUCCESS in 1h 53m 09s","accounts_in_message":[],"_revision_number":10},{"id":"4dc8ba898a0c27d7e0e8d4bac906299960ce54fe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-08 17:58:00.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"326d38aa3e7e9be516df21fe37d68861124b03e7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-08 19:24:40.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"7160ce9b7df8b7de7c0a857a089642d23fcc83d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-09 08:10:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d2597123deeaf45c62f0ab64f3727fc12fef510d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-09 08:14:07.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d18e1fbf1a4ba68a3287e77250f690f4a7a3e831","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-09 08:14:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d090102bfa4986b3c25e2597a58c51fe8f186ed5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-09 08:23:40.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"fedf61878122490768e9a36e94f4b1a47a26aaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:09:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19006308aa60424c856fa3731c8c82e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f494a3d1bb284ad4bdc07fc043eabcae : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/508f8d3231c2487c80dfc16bd80cd2db : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3d4b9f7bdaa4d62a2c792471206964f : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01b8ced7000c4cb199367cc60c4338ca : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc6ec41fa0cd497c9f457c3b2b667fd8 : SUCCESS in 3m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7bfd398545174c61a2bc61bf3318347f : SUCCESS in 17m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ec3adbd2686b4d61b5b8bc788bfe925b : SUCCESS in 11m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3d4c7eb11abf446291a3178ac739c76e : SUCCESS in 18m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f1c3c6377856446ab41b8d4a5ca58356 : SUCCESS in 16m 01s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/21787a4d6d5d4f889bc8abed5b7de168 : SUCCESS in 6m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6b75f3da5b7b465aa2dc5ee3d50c42d6 : SUCCESS in 6m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6a015c7dfd1c4c65963c21d7978aa7e3 : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/89720dfb413a428e86f3fefe5d27f7d4 : SUCCESS in 12m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c6fae9000c3e4af8bd4b27daf98a564e : SUCCESS in 6m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9ff11705145a41bea11f3510abb7f984 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/762db61941424b2caf8cd8bd47fbe36f : SUCCESS in 9m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9d3c99251539423895cdfb4349ad45b9 : SUCCESS in 5m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/996ac97734104fc3b5e3b259a1bec464 : SUCCESS in 8m 39s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3120674fc23f48b69855ae3e87e49737 : SUCCESS in 9m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4eb2c31f64614b2487cd2e9259ac8a90 : SUCCESS in 7m 50s (non-voting)\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0eacb552a0654e81aa323dd3b51cb5d4 : SUCCESS in 3m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a6903201ccfa4256bc4af1443cb92fc4 : SUCCESS in 9m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/abf252877e994b4992b448caf9f67896 : SUCCESS in 5m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e33cd2554be74656a5ad30d1ba269cdc : FAILURE in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e8009c75be3a40729937a30b158f1a09 : FAILURE in 5m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/389f3335375d4ef3b514a45350113df8 : SUCCESS in 23m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c29bcdcdb2744077bd4b572ffa89d827 : SUCCESS in 9m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/015761c91a984ff79022dd68473cf6ec : SUCCESS in 14m 36s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d2d14af4586d4004beabe022b1ea0fe7 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3709a654bc544c91b385942b9e468390 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df22ac19570e40068d81c75068089401 : SUCCESS in 27m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0fbae0af72474dbf9cafd8d191614610 : SUCCESS in 6m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d8d12a050fda49b595038be82b1355d6 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d7a90319b464d7e8a80ec67d04d5fa8 : SUCCESS in 4m 28s (non-voting)\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9631a71970a2465b825c34b39b5ba440 : SUCCESS in 15m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c8e23ed0fb0433d996678100e96d2ce : SUCCESS in 1h 09m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/691002c38b5f4460a89b988f002ab1d6 : SUCCESS in 1h 44m 21s","accounts_in_message":[],"_revision_number":11},{"id":"b98f567ed2de921cc277c74c430bf648d0b1ed03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:10:06.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"7bdf448a0de58fa0cba1dc054de19c347b47f845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:53:25.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f7b850eadb4a07b66c636b03d67f0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd20cf9a06b8481e97a3e8f34d49b8c9 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/686cb3761be34e799c55de96acb001bc : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0cd286239084efaaa13f778ae1b60a5 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3233a98f05cf42c38aa93a79c5ea2d8a : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f76f2ac0b88449e8df6ee16a5c80669 : SUCCESS in 4m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e00519767b304fbf972c5e4a5773bca2 : SUCCESS in 16m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c0716b02c504a3694cc2c531f43467f : SUCCESS in 9m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ed9f66f2ffa94f34ad727c848c79b3c3 : SUCCESS in 16m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/63510d20856140e98464c92a3dbe8b45 : SUCCESS in 21m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/48c58dc4974248d98d9bdcdaae8c5a0c : SUCCESS in 12m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6e33666fc68748debf48685cf7b33517 : SUCCESS in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/86eebe43978c44708fd92688103745e0 : SUCCESS in 5m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d3e55b1907f94843a344e897af003f5f : SUCCESS in 8m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d8af9154f87d4942ab8d72e80249418b : SUCCESS in 5m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/045ce67ac27043f4918a25bcf860e5c4 : SUCCESS in 7m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/82ef696439604012aad9b065998e0390 : SUCCESS in 8m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/855cb142aefa4ab0bbc80ffb695725bb : SUCCESS in 4m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/48d7bccf21674f2a9ea7b12bbe935b66 : SUCCESS in 10m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d19154232dde42d9b844d77aa504565c : SUCCESS in 20m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dfa1de1539b04dbca580d45a10dcb107 : SUCCESS in 9m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/341d3a490add4b8abe9dbde3250ced7e : SUCCESS in 17m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0797e957d7374ead94500a8acf77db5a : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2bb86004de64ec3836f1d24c87200cd : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7766d7d7ad2541cab4155338998ffbf5 : SUCCESS in 24m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5a5f5ce7356d455e9e0df0841acc4687 : SUCCESS in 6m 52s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/df6c42c86ee344158adbe53435352f6c : SUCCESS in 5m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2aa5ae9d3c9e4848a7a81a0134f062b5 : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a76799e2c8449ca9933ea757d4147d1 : SUCCESS in 52m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bbfa21b55394211bab7a416c43e6e38 : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":11},{"id":"7dfdc7e5fd77264453cbcfddc20ff99e6b814cc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:53:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"e2cc265d7017555637f878dd94340549ef5af80a","revisions":{"ca92db4198572485fd9773dd9a0dde726f91907b":{"kind":"REWORK","_number":1,"created":"2026-02-10 11:10:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/27/976227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/1"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:10:54.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0","message":"[stable-only] Bump pbr version to 7.0.0\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1.\n\npbr version 7.0.0 dropped hard dependency of pkg_resources module.\n\nThis is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca92db4198572485fd9773dd9a0dde726f91907b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca92db4198572485fd9773dd9a0dde726f91907b"}]},"branch":"refs/heads/stable/2025.1"},"ca424b2a9db9c9a057b8080fa27a5dbf451af239":{"kind":"REWORK","_number":2,"created":"2026-02-16 12:02:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/27/976227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/2"}}},"commit":{"parents":[{"commit":"65790a52f92c0bdba49f0e137f2e1038ace27cb8","subject":"update constraint for python-ironicclient to new release 5.10.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65790a52f92c0bdba49f0e137f2e1038ace27cb8"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-16 11:58:56.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, we need to pin setuptools to 81.0.0 for some tox jobs in\norder to avoid other issues regarding missing pkg_resources.\n(Mostly the issue is with xstatic-* packages that still uses\npkg_resources in their pinned stable versions).\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca424b2a9db9c9a057b8080fa27a5dbf451af239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca424b2a9db9c9a057b8080fa27a5dbf451af239"}]},"branch":"refs/heads/stable/2025.1"},"7c8e35bd2dafd82ae57dd70ebb218a9e873083bb":{"kind":"REWORK","_number":3,"created":"2026-02-17 18:32:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/976227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/3"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-17 18:32:15.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, we need to pin setuptools to 81.0.0 for some tox jobs in\norder to avoid other issues regarding missing pkg_resources.\n(Mostly the issue is with xstatic-* packages that still uses\npkg_resources in their pinned stable versions).\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c8e35bd2dafd82ae57dd70ebb218a9e873083bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c8e35bd2dafd82ae57dd70ebb218a9e873083bb"}]},"branch":"refs/heads/stable/2025.1"},"fdd0dfcacf181a41a569600924037cfb65ca6ebb":{"kind":"REWORK","_number":4,"created":"2026-03-05 15:17:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/27/976227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/4"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-05 15:17:13.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdd0dfcacf181a41a569600924037cfb65ca6ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdd0dfcacf181a41a569600924037cfb65ca6ebb"}]},"branch":"refs/heads/stable/2025.1"},"8d26aac1e59adc4b9ba6f708b9c08f669b393ea9":{"kind":"REWORK","_number":5,"created":"2026-03-05 15:47:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/27/976227/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/5"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-05 15:41:25.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath\n- the following 3 cross jobs needs to be set as non-voting temporarily:\n  * cross-aodh-py312\n  * cross-cinder-py312\n  * cross-keystone-py312\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d26aac1e59adc4b9ba6f708b9c08f669b393ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d26aac1e59adc4b9ba6f708b9c08f669b393ea9"}]},"branch":"refs/heads/stable/2025.1"},"b08879a5a4e2063f6688d53b60bdca8ca9978f14":{"kind":"REWORK","_number":6,"created":"2026-03-05 16:01:32.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/27/976227/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/6"}}},"commit":{"parents":[{"commit":"7b62011174e030d696d7ca42eb11ad82f2b19a3d","subject":"[stable-only] remove pkg_resouces from reqiurements-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b62011174e030d696d7ca42eb11ad82f2b19a3d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-05 16:00:03.000000000","tz":60},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n- the following 3 cross jobs needs to be set as non-voting temporarily:\n  * cross-aodh-py312\n  * cross-cinder-py312\n  * cross-keystone-py312\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b08879a5a4e2063f6688d53b60bdca8ca9978f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b08879a5a4e2063f6688d53b60bdca8ca9978f14"}]},"branch":"refs/heads/stable/2025.1"},"85dec8b10009a68b41ab7ac410e07a6e4204e146":{"kind":"REWORK","_number":7,"created":"2026-04-02 10:12:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/976227/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/7"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 10:12:00.000000000","tz":0},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n- the following 3 cross jobs needs to be set as non-voting temporarily:\n  * cross-aodh-py312\n  * cross-cinder-py312\n  * cross-keystone-py312\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85dec8b10009a68b41ab7ac410e07a6e4204e146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85dec8b10009a68b41ab7ac410e07a6e4204e146"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"ac3349b1f3c258dfb9f2c003f133027c6dae5d86":{"kind":"REWORK","_number":8,"created":"2026-04-02 10:12:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/976227/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/8"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 10:12:28.000000000","tz":0},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n- the following 3 cross jobs needs to be set as non-voting temporarily:\n  * cross-aodh-py312\n  * cross-cinder-py312\n  * cross-keystone-py312\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac3349b1f3c258dfb9f2c003f133027c6dae5d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac3349b1f3c258dfb9f2c003f133027c6dae5d86"}]},"branch":"refs/heads/stable/2025.1"},"327fb56bb6b1a7ccb050cb2ab6c86821af84b43b":{"kind":"REWORK","_number":9,"created":"2026-04-02 10:12:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/976227/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/9"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-02 10:12:48.000000000","tz":0},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n- the following 3 cross jobs needs to be set as non-voting temporarily:\n  * cross-aodh-py312\n  * cross-cinder-py312\n  * cross-keystone-py312\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/327fb56bb6b1a7ccb050cb2ab6c86821af84b43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/327fb56bb6b1a7ccb050cb2ab6c86821af84b43b"}]},"branch":"refs/heads/stable/2025.1"},"f1e48c4764e5f5b2e082d0a4f5ef29b8c270f4b4":{"kind":"REWORK","_number":10,"created":"2026-04-08 15:56:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/976227/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/10"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-08 15:56:41.000000000","tz":0},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-py39-check-uc refactor was missed in gate fix\n  patch [1] so this patch does that\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1e48c4764e5f5b2e082d0a4f5ef29b8c270f4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1e48c4764e5f5b2e082d0a4f5ef29b8c270f4b4"}]},"branch":"refs/heads/stable/2025.1"},"e2cc265d7017555637f878dd94340549ef5af80a":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-04-09 08:14:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/976227/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/976227/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/976227/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/976227/11"}}},"commit":{"parents":[{"commit":"dc1a4fe109658aa0d8b25575526e758bf768adc3","subject":"[stable-2025.1-only] Add pip.sh to py39 allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc1a4fe109658aa0d8b25575526e758bf768adc3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-10 11:05:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-09 08:14:07.000000000","tz":0},"subject":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools","message":"[stable-only] Bump pbr version to 7.0.0 and pin setuptools\n\nIn general we do not allow to bump upper constraints (especially not\nMAJOR versions), but the latest setuptools release (82.0.0) dropped\npkg_resources that causes issues when generating docs on stable/2025.1,\nas the pinned pbr version still uses pkg_resurces, but pbr version\n7.0.0 dropped hard dependency of pkg_resources module.\n(This is only issue on stable/2025.1, as on newer branches pbr upper\nconstraint is \u003e7.0.0.)\n\nFurthermore, to fix the gate a couple of changes needed to be added:\n- requirements-tox-pyXXX-check-uc jobs need pinned setuptools otherwise\n  they fail with missing pkg_resources in xstatic-objectpath, hence\n  this patch caps setuptools via virtualenv (which bundles setuptools)\n  to all tox targets as a safe workaround\n\n[1] I4cde27f048889dd557c0473138bd8a75c1e1cc58\n\nChange-Id: I7420baa4585dc3f9a97574a5e521bfd5f6c8c7d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2cc265d7017555637f878dd94340549ef5af80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2cc265d7017555637f878dd94340549ef5af80a"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
